)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~688093","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I9bb305d866f3d65653b95a25a5b9f5ecde5af0b0","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"centos-private-epel","hashtags":[],"change_id":"I9bb305d866f3d65653b95a25a5b9f5ecde5af0b0","subject":"Add Centos GPG key custom url","status":"MERGED","created":"2019-10-11 09:53:43.000000000","updated":"2019-11-01 08:14:50.000000000","submitted":"2019-10-18 23:24:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688093-1571441083355-fe29816b","meta_rev_id":"b15d3b6404d3ec28a5cc51183474b6d66d9bf45b","_number":688093,"virtual_id_number":688093,"owner":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 23:24:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-10-17 21:08:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":2,"date":"2019-10-18 21:54:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":1,"date":"2019-10-18 21:54:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-17 21:08:50.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-18 21:54:39.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-18 23:24:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e035b94425c832d5515f77134723721a4a7eaa6f","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-11 09:53:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b1f1e82e8e14f8d6ee8affe35c86e33ddc172e6","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-11 10:39:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"90086b5ff6ef54cc2569d72e02d03afce6a04f7d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-11 12:18:58.000000000","message":"Patch Set 2: Code-Review-1\n\nCan you add this variable commented out inside defaults with appropriate comment or/and add a release note?\nIt\u0027s really very useful in terms of having variables documentated","accounts_in_message":[],"_revision_number":2},{"id":"b9c9e86b2ffcfa031fb9e2acb6d2d5480f3e9022","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 12:29:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b2b58d85633142829ed18affc55f760c : SUCCESS in 6m 11s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/878ae580524e4a0a9f6354eabc79cb30 : SUCCESS in 1h 23m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f394dd46c6734eceaaa81bd1b10c1072 : SUCCESS in 1h 37m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/693686eaac31453984e5cf3083d8ebc6 : SUCCESS in 1h 08m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd40f6745c9048278f0982a14443d9c3 : SUCCESS in 1h 05m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e334ed1827604ee196fac86c5f32d9aa : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"7bd98bb9efb9b683da03acf80fa3d74ede0cfa7a","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-11 12:36:55.000000000","message":"Patch Set 2:\n\n\u003e Can you add this variable commented out inside defaults with\n \u003e appropriate comment or/and add a release note?\n \u003e It\u0027s really very useful in terms of having variables documentated\n\nI never write a release note, is there a doc ?","accounts_in_message":[],"_revision_number":2},{"id":"2949db7f9a87e8aa010216c936b6876976c71198","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-11 12:44:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1722f814fa8e509ea70a201ee4639118a5145178","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-11 12:45:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f95bee33510d97ca79735fd42ccbad625a70f62b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-11 12:46:50.000000000","message":"Patch Set 3:\n\nThere\u0027s a doc here https://docs.openstack.org/reno/latest/user/usage.html","accounts_in_message":[],"_revision_number":3},{"id":"206b40e80e1083e3c9c34aea2ffaa5fd71b24909","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-11 13:16:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ee88b08149afc84f4cee320047e14149fbc5799a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 14:15:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a0c8dc430f634d10bd336deb47b87075 : SUCCESS in 6m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/46b7f3f2d2cb47899326b7654d48e415 : SUCCESS in 1h 11m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5dbc05deafc34c75b93b35c69bd2fdf1 : SUCCESS in 1h 25m 18s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b1414a6904d9442cb7245bd2d19f1cba : SUCCESS in 1h 18m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe4571d61ec545d59b93624992a8bb26 : SUCCESS in 1h 01m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0114e48235a42e388999444b55e74c4 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":3},{"id":"a2b4d28ac88149d661dba9c1903049017613edf2","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-11 16:10:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6022dcb897a020e3568258df21095cd1da8aa88b","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-11 16:23:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"64a020a17f86ebb5d431131cb5df5d5dbeb7da17","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-11 16:25:23.000000000","message":"Patch Set 5:\n\nI removed the commented lines in defaults because I put a release note. Add lines in defaults add complexity to this file mostly in the other osa repo I have to fix.","accounts_in_message":[],"_revision_number":5},{"id":"210c8cd1b5f5394a2810664758874238d4b4c7c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 18:12:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/63b47645909148f6b55754b31c288f99 : SUCCESS in 10m 39s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e75d7f1342ab41e89650057975cc8fbb : SUCCESS in 1h 31m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3b2e5a634c4f40e78040dc538dba15e6 : SUCCESS in 1h 42m 55s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/43588b6470a645b2a79f956c2d9cb43a : SUCCESS in 1h 26m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52ff8ea2a979447d9602e200db610c25 : SUCCESS in 1h 16m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f2000827e0d489790bedf0915f8a9ae : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcb1d069654f44fe99ad67f0630d96f9 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":5},{"id":"615254ae0c4b4d6f2b27a20401d9c8d09f10cc2e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-11 20:25:54.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nSorry for the confusion, please also add the var commented to the defaults/main.yml as Dmitry suggested, the role documentation is autogenerated from the defaults file.","accounts_in_message":[],"_revision_number":5},{"id":"2bb163e59ad16d15da886f9af9eb3d04e6918df5","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-14 12:21:20.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a6af4b2c41a6e17d6d6cce3375bad37973a3ecb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 13:53:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7add61323dcf49e9841b5ba566c13200 : SUCCESS in 7m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/9d236fd22fc3480abe833fb7503c8e9d : SUCCESS in 1h 11m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/84abe8002d7347e691ea5f65a9448d63 : SUCCESS in 1h 30m 22s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e36aa4db9bb94fe6ae677042a0a6a811 : SUCCESS in 1h 20m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3a5e2ccea864f928fc005c3a5e814da : SUCCESS in 1h 21m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/992bf17410b8462c8629e7ae12d29b6e : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6774abb330d4ace8e60746ebc4f4b28 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":6},{"id":"d0e5aef6593f2888499630acbb583dc4b08c7329","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-14 14:02:58.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"66a876443e4316765fc5faa0798692307018dc07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 16:12:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a5858ff99282452d97bb6b286925cf8c : SUCCESS in 7m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3277e26a75d2404ba1ac17608c96c533 : SUCCESS in 1h 43m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d4c05fb5cd914751b78422a7aba29572 : SUCCESS in 1h 49m 02s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d254b2f3d0604e1b89903c80b93e6028 : SUCCESS in 1h 18m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88948899122840b591a96aac1425674b : SUCCESS in 1h 14m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9763b67d804d4c7c8b08854723ed6bc9 : SUCCESS in 2m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90957881834f4e63a243c2b6a648a3ad : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":7},{"id":"eb04aba33f330e1432daefc51c9c5e4e6b41d33d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-15 07:53:04.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a52377abfc525095d130fb30a38aeae2cba83de3","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-17 09:07:45.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3f43eb15b815727b1ddce849908129a4c900438e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-17 10:43:23.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)\n\nAs we\u0027re using another mirror in CI [1], we can\u0027t just drop centos_epel_mirror\n\n[1] https://opendev.org/openstack/openstack-ansible/src/branch/master/tests/roles/bootstrap-host/tasks/prepare_aio_config.yml#L85","accounts_in_message":[],"_revision_number":8},{"id":"babe0d26cd58bb24765a91787936e8c0ccb91737","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-17 10:44:17.000000000","message":"Patch Set 8:\n\nAnd we have to fix effect of https://review.opendev.org/688164","accounts_in_message":[],"_revision_number":8},{"id":"33ecdfea32861b25f17b63c9ed1b6df7abcf4b5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 10:57:37.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/94a77f53c5924f5bb254e3a33011542d : SUCCESS in 5m 45s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/7ea51d0ec42b43018cc6ee01c41bfd30 : SUCCESS in 1h 23m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/779819eb2baa4affacb6eb90e83f66cd : SUCCESS in 1h 47m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0d3b619ccbfb4661ba1a14b0cb89c30b : FAILURE in 57m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/919ec4ba86c941948bad27fd8d0d475d : SUCCESS in 1h 19m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/337632e244d04d1cac0b671d835492f8 : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42d20188c97245589bffc51dadb4a6a7 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":8},{"id":"70af99f8e94500bcb9f2d1564f2e6457013a6552","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-17 11:44:46.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4968c6992139090394928f3fe199035b351f245c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-17 11:55:26.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)\n\njust one small nit","accounts_in_message":[],"_revision_number":9},{"id":"df36dd35d4dca9cc4cf6fe2b03383dc255ae1e49","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-17 12:30:12.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"26fa02d91b68fc09d9d892d3e9827038b11825b1","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-17 12:31:03.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"2bd1358cb0ca0cc03a792786b97b3bbaa6816c73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 13:55:45.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2edb9c54ef4f422f9affbcbd01a936d4 : SUCCESS in 5m 16s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/75458aa7bca44a028441c97b7204f5e6 : SUCCESS in 1h 17m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e9bd2bdc78d84e8ab4b9938f39fd6774 : FAILURE in 1h 19m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b3f1cc1336874a269b837fed397806d8 : FAILURE in 47m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f9cb159b3fa4bd58f05c47075e2ccd5 : SUCCESS in 1h 08m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/262250f5dc29492d941645b10b951318 : SUCCESS in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a2a9a8b9406442baa9bc4d28fe338e4 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":10},{"id":"eb48b30d7bcfab730dd0c28d1082d21ae6e8509e","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-17 14:24:48.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"bf74692bfe068872fe6c652ef6461dfd591b8e56","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-17 15:26:03.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"724033b26156d5c450b7513870814553f94c9c17","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-17 15:28:59.000000000","message":"Patch Set 11: Code-Review-1\n\n(4 comments)\n\nAs you suggested to use another naming:)","accounts_in_message":[],"_revision_number":11},{"id":"45d4883dafa15076f4ed2a5c8fa4412fb9f57d08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 16:49:35.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b6d7d0e455f84622ab2580a07196ec1b : SUCCESS in 6m 02s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/45ed9aa47e6d4095be9dc340432608db : SUCCESS in 1h 11m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/57c44d218e3a4cbe9d9544428a20366c : FAILURE in 1h 17m 02s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7cd24413b87046bba35793ac53a5a5e3 : FAILURE in 58m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86dcd4fcb92943a3bdc1cfc265629bfb : SUCCESS in 1h 05m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bcee00d87f8413384a480018acb9a87 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/177e02313c36439a9f3b119d64dd1d9b : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":11},{"id":"60f5ec5c5c43d5c4ff02bb55ce41f5f97f4b40ac","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-17 18:57:37.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f698f5fa6205f81a5105e070d29c4bf8c61eeaa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 20:36:48.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/812d6c8eb4a24c939f7f214f0a1010e6 : SUCCESS in 6m 17s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/c5b4f277964640009336c44da1e89a21 : SUCCESS in 1h 12m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/61f4d0e0d0b3413f9659932247ccd4a7 : SUCCESS in 1h 34m 05s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0933df173f2f484490fb9d5a43b61d52 : FAILURE in 47m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/118a64e824eb462ba01dfe813eb0ea9f : SUCCESS in 1h 10m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c01eeaafd8cf40608b937f3f8e179317 : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3c9a75811164eb18537b7bec68d81d5 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":12},{"id":"5c35c2105f44fe891aab8ed4c262879f47e53d79","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-17 21:08:02.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"337e6ed8f3c23dc7506a1556438490630e187d80","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-17 21:08:50.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"76f2e7797e4f79d9fa333d976713f92f6b3f031b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 22:42:48.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/eca4d119e72b44f6a1ee33232314ad19 : SUCCESS in 5m 02s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/71426eba7b1f4c93be285480e9a7e2c9 : SUCCESS in 1h 16m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3665fcdac3d046b69e70cbdcc5fd8be8 : SUCCESS in 1h 32m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3fe677b0533841b687c2cd1d9c86b077 : FAILURE in 51m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0184b02ace04806b86121aa50fd89f9 : SUCCESS in 1h 15m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74c45620235a4494a6d23485cd427296 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc1d4b4da69c4786899f7e287595dd0c : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":12},{"id":"bb52ef85b2e4541050f49a7deff08b2ea9cb172a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-18 18:35:44.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"0f2eac5815255eee63d8e033867dddb8e1843598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 20:12:12.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b51d7bd74b7a466db5d129e4737fbd7b : SUCCESS in 5m 03s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/863ed069115e4da49760616867fbb2ba : SUCCESS in 1h 15m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/29c4f430393841fab281521d6c731658 : SUCCESS in 1h 31m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3d4a8d3180434ff5bfb7e34e8b66cc3b : SUCCESS in 1h 17m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/045636e537bd45d08d83c447558fd9f7 : SUCCESS in 1h 01m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cc6c523adf244f3a846fa962f6e3348 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03249576aae14f84bc60f56eceedd036 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":12},{"id":"89a8fe032d1c45d04430383e4c95baef88193f1c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-18 21:54:39.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"310c04f8526df44aae1d602989268341e44c0ac7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 21:54:50.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"dbf350d6130181d317e1d8b27d237a721ac3502f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 23:24:43.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bf2f3381191a431986ba52afe726ea41 : SUCCESS in 5m 41s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a4def382d607450180a967b7b8f476c0 : SUCCESS in 1h 09m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bd70baa609f941ff8cb4732aac4069c7 : SUCCESS in 1h 27m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ebf6e68d7918464da99544178c6cc04d : SUCCESS in 1h 10m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1c520a7de95457db237f008e50f9934 : SUCCESS in 1h 08m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77cf542b040a4c8f83f5d4b7825f90c6 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e13f4a79e56641799bde4b48379829c6 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":12},{"id":"1dc345b8db3c4cf7e048363f87183e84329a78ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 23:24:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"d3d16f1d62de90ebd30cb95ae625274407b7fc8e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 23:25:50.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dc8f2e03fd74a8bbe474e11b5ca3c81 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d9a6cb5f2f54661b19246a74c9bbf7e : SUCCESS in 52s","accounts_in_message":[],"_revision_number":12},{"id":"d5b5a033da0c2d044b97b68963d7ddc37c28af22","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-10-31 23:14:54.000000000","message":"Patch Set 12: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 979a5cb6c3c331dce9f9ce7c459ca46b64e06814","accounts_in_message":[],"_revision_number":12},{"id":"c8ec476f3cdad6a16e84880ca3d20291c1996aa9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-01 08:14:50.000000000","message":"Patch Set 12: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit f0d137e1659cb513f3d51c5177b8df734fce82f1","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"aecff065f1c07ff321e62d8bd9718a462019b53d","revisions":{"f912799922017439865830af2b390d36b3d45e6c":{"kind":"REWORK","_number":1,"created":"2019-10-11 09:53:43.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/93/688093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/93/688093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/1"}}},"commit":{"parents":[{"commit":"3c02d26e25345c52e6d0ffc7894f7f96c5d8357f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3c02d26e25345c52e6d0ffc7894f7f96c5d8357f"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 09:51:09.000000000","tz":120},"committer":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 09:53:41.000000000","tz":120},"subject":"Add Centos GPG key custom url","message":"Add Centos GPG key custom url\n\nWhen using a custom repo with centos_epel_mirror, you maybe need to\nchange the gpg key url because offline env\n\nYou can use this variable: centos_epel_gpg_key\n\nChange-Id: I9bb305d866f3d65653b95a25a5b9f5ecde5af0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/f912799922017439865830af2b390d36b3d45e6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/f912799922017439865830af2b390d36b3d45e6c"}]},"branch":"refs/heads/master"},"c0303773b28dcdecdb56cea36859f8ece47c3ffe":{"kind":"REWORK","_number":2,"created":"2019-10-11 10:39:29.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/93/688093/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/93/688093/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/2"}}},"commit":{"parents":[{"commit":"3c02d26e25345c52e6d0ffc7894f7f96c5d8357f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3c02d26e25345c52e6d0ffc7894f7f96c5d8357f"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 09:51:09.000000000","tz":120},"committer":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 10:39:06.000000000","tz":120},"subject":"Add Centos GPG key custom url","message":"Add Centos GPG key custom url\n\nWhen using a custom repo with centos_epel_mirror, you maybe need to\nchange the gpg key url because offline env\n\nYou can use this variable: centos_epel_gpg_key\n\nChange-Id: I9bb305d866f3d65653b95a25a5b9f5ecde5af0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c0303773b28dcdecdb56cea36859f8ece47c3ffe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c0303773b28dcdecdb56cea36859f8ece47c3ffe"}]},"branch":"refs/heads/master"},"116ffa5db25385ea5c7c35af74c51bde42dcc969":{"kind":"REWORK","_number":3,"created":"2019-10-11 12:44:45.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/93/688093/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/93/688093/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/3"}}},"commit":{"parents":[{"commit":"3c02d26e25345c52e6d0ffc7894f7f96c5d8357f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3c02d26e25345c52e6d0ffc7894f7f96c5d8357f"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 09:51:09.000000000","tz":120},"committer":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 12:43:47.000000000","tz":120},"subject":"Add Centos GPG key custom url","message":"Add Centos GPG key custom url\n\nWhen using a custom repo with centos_epel_mirror, you maybe need to\nchange the gpg key url because offline env\n\nYou can use this variable: centos_epel_gpg_key\n\nChange-Id: I9bb305d866f3d65653b95a25a5b9f5ecde5af0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/116ffa5db25385ea5c7c35af74c51bde42dcc969"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/116ffa5db25385ea5c7c35af74c51bde42dcc969"}]},"branch":"refs/heads/master"},"22ed7d39511b7fbf57449ac44b1fc74e846745e2":{"kind":"REWORK","_number":4,"created":"2019-10-11 16:10:56.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/93/688093/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/93/688093/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/4"}}},"commit":{"parents":[{"commit":"3c02d26e25345c52e6d0ffc7894f7f96c5d8357f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3c02d26e25345c52e6d0ffc7894f7f96c5d8357f"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 09:51:09.000000000","tz":120},"committer":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 16:10:08.000000000","tz":120},"subject":"Add Centos GPG key custom url","message":"Add Centos GPG key custom url\n\nWhen using a custom repo with centos_epel_mirror, you maybe need to\nchange the gpg key url because offline env\n\nYou can use this variable: centos_epel_gpg_key\n\nChange-Id: I9bb305d866f3d65653b95a25a5b9f5ecde5af0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/22ed7d39511b7fbf57449ac44b1fc74e846745e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/22ed7d39511b7fbf57449ac44b1fc74e846745e2"}]},"branch":"refs/heads/master"},"5e10dac81c17bdd8692dcb8db500ace4120336f8":{"kind":"REWORK","_number":5,"created":"2019-10-11 16:23:08.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/93/688093/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/93/688093/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/5"}}},"commit":{"parents":[{"commit":"3c02d26e25345c52e6d0ffc7894f7f96c5d8357f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3c02d26e25345c52e6d0ffc7894f7f96c5d8357f"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 09:51:09.000000000","tz":120},"committer":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 16:22:20.000000000","tz":120},"subject":"Add Centos GPG key custom url","message":"Add Centos GPG key custom url\n\nWhen using a custom repo with centos_epel_mirror, you maybe need to\nchange the gpg key url because offline env\n\nYou can use this variable: centos_epel_gpg_key\n\nChange-Id: I9bb305d866f3d65653b95a25a5b9f5ecde5af0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5e10dac81c17bdd8692dcb8db500ace4120336f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5e10dac81c17bdd8692dcb8db500ace4120336f8"}]},"branch":"refs/heads/master"},"8f655fb44301fce70be8687056fe7cba8d4b6465":{"kind":"REWORK","_number":6,"created":"2019-10-14 12:21:20.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/93/688093/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/93/688093/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/6"}}},"commit":{"parents":[{"commit":"3c02d26e25345c52e6d0ffc7894f7f96c5d8357f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3c02d26e25345c52e6d0ffc7894f7f96c5d8357f"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 09:51:09.000000000","tz":120},"committer":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-14 12:20:14.000000000","tz":120},"subject":"Add Centos GPG key custom url","message":"Add Centos GPG key custom url\n\nWhen using a custom repo with centos_epel_mirror, you maybe need to\nchange the gpg key url because offline env\n\nYou can use this variable: centos_epel_gpg_key\n\nChange-Id: I9bb305d866f3d65653b95a25a5b9f5ecde5af0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/8f655fb44301fce70be8687056fe7cba8d4b6465"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/8f655fb44301fce70be8687056fe7cba8d4b6465"}]},"branch":"refs/heads/master"},"8dba0944c3b8cc492a6ff68f91fc13d1b3dc4bbd":{"kind":"REWORK","_number":7,"created":"2019-10-14 14:02:58.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/93/688093/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/93/688093/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/7"}}},"commit":{"parents":[{"commit":"3c02d26e25345c52e6d0ffc7894f7f96c5d8357f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3c02d26e25345c52e6d0ffc7894f7f96c5d8357f"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 09:51:09.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-10-14 14:02:48.000000000","tz":0},"subject":"Add Centos GPG key custom url","message":"Add Centos GPG key custom url\n\nWhen using a custom repo with centos_epel_mirror, you maybe need to\nchange the gpg key url because offline env\n\nYou can use this variable: centos_epel_gpg_key\n\nChange-Id: I9bb305d866f3d65653b95a25a5b9f5ecde5af0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/8dba0944c3b8cc492a6ff68f91fc13d1b3dc4bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/8dba0944c3b8cc492a6ff68f91fc13d1b3dc4bbd"}]},"branch":"refs/heads/master"},"9ffcda254d3d2101246faef05a55af8bbe118795":{"kind":"REWORK","_number":8,"created":"2019-10-17 09:07:45.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/93/688093/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/93/688093/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/8"}}},"commit":{"parents":[{"commit":"3c02d26e25345c52e6d0ffc7894f7f96c5d8357f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3c02d26e25345c52e6d0ffc7894f7f96c5d8357f"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 09:51:09.000000000","tz":120},"committer":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-17 09:07:33.000000000","tz":120},"subject":"Add Centos GPG key custom url","message":"Add Centos GPG key custom url\n\nWhen using a custom repo with repo_centos_epel_mirror, you maybe need to\nchange the gpg key url because offline env\n\nYou can use this variable: repo_centos_epel_gpg_key\n\nChange-Id: I9bb305d866f3d65653b95a25a5b9f5ecde5af0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/9ffcda254d3d2101246faef05a55af8bbe118795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/9ffcda254d3d2101246faef05a55af8bbe118795"}]},"branch":"refs/heads/master"},"d1cbfaf48de1bb4e80f6f02e35dca9ca274efc48":{"kind":"REWORK","_number":9,"created":"2019-10-17 11:44:46.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/93/688093/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/93/688093/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/9"}}},"commit":{"parents":[{"commit":"3c02d26e25345c52e6d0ffc7894f7f96c5d8357f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3c02d26e25345c52e6d0ffc7894f7f96c5d8357f"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 09:51:09.000000000","tz":120},"committer":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-17 11:44:07.000000000","tz":120},"subject":"Add Centos GPG key custom url","message":"Add Centos GPG key custom url\n\nWhen using a custom repo with repo_centos_epel_mirror, you maybe need to\nchange the gpg key url because offline env\n\nYou can use this variable: repo_centos_epel_key\n\nChange-Id: I9bb305d866f3d65653b95a25a5b9f5ecde5af0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d1cbfaf48de1bb4e80f6f02e35dca9ca274efc48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d1cbfaf48de1bb4e80f6f02e35dca9ca274efc48"}]},"branch":"refs/heads/master"},"e4c0bafa75384625bf6bf7200f0f942b2678ff7d":{"kind":"REWORK","_number":10,"created":"2019-10-17 12:31:03.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/93/688093/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/93/688093/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/10"}}},"commit":{"parents":[{"commit":"3c02d26e25345c52e6d0ffc7894f7f96c5d8357f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3c02d26e25345c52e6d0ffc7894f7f96c5d8357f"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 09:51:09.000000000","tz":120},"committer":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-17 12:30:38.000000000","tz":120},"subject":"Add Centos GPG key custom url","message":"Add Centos GPG key custom url\n\nWhen using a custom repo with repo_centos_epel_mirror, you maybe need to\nchange the gpg key url because offline env\n\nYou can use this variable: repo_centos_epel_key\n\nChange-Id: I9bb305d866f3d65653b95a25a5b9f5ecde5af0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/e4c0bafa75384625bf6bf7200f0f942b2678ff7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/e4c0bafa75384625bf6bf7200f0f942b2678ff7d"}]},"branch":"refs/heads/master"},"d59032dd7f13fd69222d4c77b2e798b96957e2e7":{"kind":"REWORK","_number":11,"created":"2019-10-17 15:26:03.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/93/688093/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/93/688093/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/11"}}},"commit":{"parents":[{"commit":"3c02d26e25345c52e6d0ffc7894f7f96c5d8357f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3c02d26e25345c52e6d0ffc7894f7f96c5d8357f"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 09:51:09.000000000","tz":120},"committer":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-17 15:24:51.000000000","tz":120},"subject":"Add Centos GPG key custom url","message":"Add Centos GPG key custom url\n\nWhen using a custom repo with repo_centos_epel_mirror, you maybe need to\nchange the gpg key url because offline env\n\nYou can use this variable: repo_centos_epel_gpg_key\n\nChange-Id: I9bb305d866f3d65653b95a25a5b9f5ecde5af0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d59032dd7f13fd69222d4c77b2e798b96957e2e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d59032dd7f13fd69222d4c77b2e798b96957e2e7"}]},"branch":"refs/heads/master"},"aecff065f1c07ff321e62d8bd9718a462019b53d":{"kind":"REWORK","_number":12,"created":"2019-10-17 18:57:37.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/93/688093/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/93/688093/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/93/688093/12"}}},"commit":{"parents":[{"commit":"3c02d26e25345c52e6d0ffc7894f7f96c5d8357f","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3c02d26e25345c52e6d0ffc7894f7f96c5d8357f"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-11 09:51:09.000000000","tz":120},"committer":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-10-17 18:56:01.000000000","tz":120},"subject":"Add Centos GPG key custom url","message":"Add Centos GPG key custom url\n\nWhen using a custom repo with repo_centos_epel_mirror, you maybe need to\nchange the gpg key url because offline env\n\nYou can use this variable: repo_centos_epel_key\n\nChange-Id: I9bb305d866f3d65653b95a25a5b9f5ecde5af0b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/aecff065f1c07ff321e62d8bd9718a462019b53d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/aecff065f1c07ff321e62d8bd9718a462019b53d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
