)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~690324","triplet_id":"openstack%2Fopenstack-ansible-repo_server~stable%2Ftrain~I9546a2ff92276af0cdd0d246b010b539037a2823","project":"openstack/openstack-ansible-repo_server","branch":"stable/train","topic":"create-train","hashtags":[],"change_id":"I9546a2ff92276af0cdd0d246b010b539037a2823","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","status":"MERGED","created":"2019-10-22 19:04:17.000000000","updated":"2019-10-24 10:43:23.000000000","submitted":"2019-10-24 10:40:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690324-1571913656647-cd6aae0a","meta_rev_id":"7401f25cee4c3518ab8b070ebd51672239183c0b","_number":690324,"virtual_id_number":690324,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-24 10:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-10-24 10:27:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-10-24 10:27:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 10:27:18.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-24 10:40:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb23f5088e288aac39445fbdf3a03b76e71a4b07","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-10-22 19:04:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcb5432150cc79107632e0dd79995e4f8197c14a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 19:26:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/01d4a987f81846a4b253c55cd41a95b9 : SUCCESS in 9m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a559ffc48e14e2fb9a8bc972e80f564 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0ea865209ff4791a053370f28968c9a : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"7a44bcdbbf5aa9217cffaf48be9f2bed6a992995","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-24 10:27:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"624c76e432f424ba5b5ea00957d03aca2250fb85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 10:27:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32cc5dcd04624c5e35b2c01c50685a83667e3292","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 10:40:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/37350fc7a12e4058bb4d3540af7213f4 : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9cf7564af004fb89f905a680b4b6b83 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e4bfedfa79f40879d9c7f0f66ff8034 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"0287fce3993c255059d54293a93bf3db008ec756","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 10:40:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9061876661f33f6465537eff62ff9af01a374a6e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 10:43:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/252223bfd560470fa401d16063893a94 : SUCCESS in 2m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1e4796bc5224ac98646fc82bb5f2441 : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b9bb40450e11bd81d643ccb9ea1a4f39d6bd34c","revisions":{"6b9bb40450e11bd81d643ccb9ea1a4f39d6bd34c":{"kind":"REWORK","_number":1,"created":"2019-10-22 19:04:17.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/24/690324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/24/690324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/24/690324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/24/690324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/24/690324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/24/690324/1"}}},"commit":{"parents":[{"commit":"1e9d40cff2481c04bd7216c038a10177944efc02","subject":"Update .gitreview for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/1e9d40cff2481c04bd7216c038a10177944efc02"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-22 19:04:14.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-22 19:04:14.000000000","tz":0},"subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","message":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/train branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I9546a2ff92276af0cdd0d246b010b539037a2823\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/6b9bb40450e11bd81d643ccb9ea1a4f39d6bd34c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/6b9bb40450e11bd81d643ccb9ea1a4f39d6bd34c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
