)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~693508","triplet_id":"openstack%2Fopenstack-ansible-repo_server~stable%2Fstein~I8dcb3c97e9593877a4a420bd32b50ae29d9d311c","project":"openstack/openstack-ansible-repo_server","branch":"stable/stein","topic":"Offline_repo-stable/stein","hashtags":[],"change_id":"I8dcb3c97e9593877a4a420bd32b50ae29d9d311c","subject":"Add possibility to overwrite nginx public repo","status":"ABANDONED","created":"2019-11-08 09:47:33.000000000","updated":"2019-11-11 23:02:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fac82c78d5ad2c213290f308886feed490d25eda","_number":693508,"virtual_id_number":693508,"owner":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-11-08 11:35:35.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-08 11:13:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-11-08 11:47:49.000000000","_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":-1,"date":"2019-11-08 20:01:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-08 11:13:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-08 11:35:35.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-11-08 20:01:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"c957a08012d269668277c873d5faf0aa1596d017","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-11-08 09:47:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ca312a0246480db187d87e426f6b143ecc2b2cdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 11:13:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/723c7e1e315e416f83fec25128c85fa3 : SUCCESS in 4m 18s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/9f211c3772c94993ad1b3976aaf30bb2 : SUCCESS in 1h 21m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1d9cc98a2e2a4422bbdcf7f02681edcc : SUCCESS in 1h 16m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d795baa02484efd9ca127873c64a8d5 : SUCCESS in 1h 00m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c1d3d1d0a334906b1928529e20fad7b : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":1},{"id":"07bfbdf32e908b1788ade992b1ccf012a5f5b1c5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-08 11:35:35.000000000","message":"Patch Set 1:\n\nthis brings new functionality, so I\u0027m not sure how appropriate this backport is.","accounts_in_message":[],"_revision_number":1},{"id":"787d0eb4951251321e65d3519e2a2fa89387c4c6","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-11-08 11:47:49.000000000","message":"Patch Set 1:\n\n\u003e this brings new functionality, so I\u0027m not sure how appropriate this\n \u003e backport is.\n\nI need to install a stein osa using local mirrors, i need to change these urls. For me not a new functionnality but a new install option","accounts_in_message":[],"_revision_number":1},{"id":"795338ca068b93054364286e034037803331e0ac","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-08 20:01:26.000000000","message":"Patch Set 1: Code-Review-1\n\nI agree with Dmitry, the defaults/main.yml files on OSA roles are the equivalent of our API and these are pretty much never touched on stable branches. I also think that your requirement to deploy from local mirrors is new functionality in the context of a deployment project like OSA. The features you need are present in the Train release candidate branch.","accounts_in_message":[],"_revision_number":1},{"id":"dcf5b96dac92016252032bf28ced981ae3f5bbfa","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-11-11 23:02:13.000000000","message":"Abandoned\n\nconsidered as a new feature can\u0027t be backported to stein","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"59f46c5e67c98f5beaf43194f794a350d7e0cc24","revisions":{"59f46c5e67c98f5beaf43194f794a350d7e0cc24":{"kind":"REWORK","_number":1,"created":"2019-11-08 09:47:33.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/08/693508/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/08/693508/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/08/693508/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/08/693508/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/08/693508/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/08/693508/1"}}},"commit":{"parents":[{"commit":"979a5cb6c3c331dce9f9ce7c459ca46b64e06814","subject":"Add Centos GPG key custom url","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/979a5cb6c3c331dce9f9ce7c459ca46b64e06814"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@alterway.fr","date":"2019-11-04 19:17:59.000000000","tz":60},"committer":{"name":"kourosh vivan","email":"kourosh.vivan@alterway.fr","date":"2019-11-08 09:47:33.000000000","tz":0},"subject":"Add possibility to overwrite nginx public repo","message":"Add possibility to overwrite nginx public repo\n\nFor Centos, we use by default public repo for nginx. You can change this\nbehaviour with these role-wide variables:\n- repo_centos_nginx_mirror\n- repo_centos_nginx_key\n\nOr with these osa-wide variables:\n- centos_nginx_mirror\n- centos_nginx_key\n\nChange-Id: I8dcb3c97e9593877a4a420bd32b50ae29d9d311c\n(cherry picked from commit bd8a24716c519badc985aceb08d00b113ab5f857)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/59f46c5e67c98f5beaf43194f794a350d7e0cc24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/59f46c5e67c98f5beaf43194f794a350d7e0cc24"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
