)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~698290","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~Id6e964d2428afcf19a4167c18aad147be7079ee7","project":"openstack/openstack-ansible-repo_server","branch":"master","hashtags":[],"change_id":"Id6e964d2428afcf19a4167c18aad147be7079ee7","subject":"epel: use vendored gpg keys","status":"ABANDONED","created":"2019-12-10 16:26:04.000000000","updated":"2020-07-30 20:08:13.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b717fe6bbb78f41ea4c209f6ee7c4ec85f0cf524","_number":698290,"virtual_id_number":698290,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-10 22:02:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},"all":[{"value":-1,"date":"2019-12-13 10:20:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},{"value":0,"date":"2019-12-10 18:14:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-10 18:14:34.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-12-10 22:02:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-13 10:20:11.000000000","updated_by":{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},"reviewer":{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},"state":"REVIEWER"}],"messages":[{"id":"f2268683800c14bf9b18ea115b6743675cdfdfa3","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-12-10 16:26:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17dffc10d26e2e12be315d5ee53e9a627a8b0d19","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-12-10 18:09:40.000000000","message":"Patch Set 1: Code-Review-1\n\nDoesn\u0027t epel key exist in database? Can we add it by id \"352C64E5\" and define centos_epel_key in [1]?\n\n[1] https://opendev.org/openstack/openstack-ansible/src/branch/master/tests/roles/bootstrap-host/templates/user_variables.aio.yml.j2","accounts_in_message":[],"_revision_number":1},{"id":"a20b1da0c4b360079bdc570bf2700d9a0e51f7de","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-12-10 18:12:35.000000000","message":"Removed Code-Review-1 by Dmitriy Rabotyagov (noonedeadpunk) \u003cnoonedeadpunk@ya.ru\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"c7e6062166d9ae61d745567681cd282e8ea3fb75","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-12-10 18:14:34.000000000","message":"Patch Set 1:\n\nYeah, it work only for absoleting keys... Anyway I\u0027d prefer doing this for integrated repo for CI, as centos_epel_key was introduced not so far ago so that deployers could override keys and repo location.","accounts_in_message":[],"_revision_number":1},{"id":"27eeb913caa072f3b6dcd63a7f535f0370e6b020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 22:02:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a0e0d04110704c988fb965febfe07a3f : SUCCESS in 10m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b44269d537414c458bb230b2afc93b3c : SUCCESS in 1h 11m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5584001618a14ce682131fbf7befcfbe : FAILURE in 19m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/957a9073b6554b7b9249caefc355ca1a : FAILURE in 26m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9139c19b315540d4951e79fd63f07cd3 : SUCCESS in 1h 12m 19s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/59bb5c472d2c465fbad472f3e29248d1 : SUCCESS in 2h 29m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9016f61ac7d841aea0d58ce9763c1ca1 : SUCCESS in 1h 45m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d0c4b1f2c6b4dceaa1d324f537a9123 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":1},{"id":"60441b35d50e5ae3536b8cae0ce6c2b12b47cfdf","author":{"_account_id":25591,"name":"Chenjun Shen","email":"chenjun.shen@gmail.com","username":"cshen"},"date":"2019-12-13 10:20:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nSee my comment in the code.","accounts_in_message":[],"_revision_number":1},{"id":"b717fe6bbb78f41ea4c209f6ee7c4ec85f0cf524","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-07-30 20:08:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7bdf5c3867d8f920100ee70b4b2413691596bfa9","revisions":{"7bdf5c3867d8f920100ee70b4b2413691596bfa9":{"kind":"REWORK","_number":1,"created":"2019-12-10 16:26:04.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/90/698290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/90/698290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/698290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/698290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/698290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/698290/1"}}},"commit":{"parents":[{"commit":"71ebeb123b346b459f0f02cbac733473ce30a521","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/71ebeb123b346b459f0f02cbac733473ce30a521"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-12-10 16:25:29.000000000","tz":-300},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-12-10 16:26:03.000000000","tz":-300},"subject":"epel: use vendored gpg keys","message":"epel: use vendored gpg keys\n\nIn order to avoid remote network issues, we vendor the GPG key\nand install it locally.\n\nChange-Id: Id6e964d2428afcf19a4167c18aad147be7079ee7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/7bdf5c3867d8f920100ee70b4b2413691596bfa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/7bdf5c3867d8f920100ee70b4b2413691596bfa9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
