)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~709532","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~Iab8d28c32e534e62759dc2bc72bd6368fbea471d","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"osa/el8","hashtags":[],"change_id":"Iab8d28c32e534e62759dc2bc72bd6368fbea471d","subject":"Add Centos-8 support","status":"MERGED","created":"2020-02-24 13:55:56.000000000","updated":"2020-05-27 16:45:48.000000000","submitted":"2020-05-27 16:43:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709532-1590597829278-649d007f","meta_rev_id":"8bd95c3ea9b10a933f51b001d7df0602c63ccb7e","_number":709532,"virtual_id_number":709532,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-27 16:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"all":[{"value":2,"date":"2020-05-27 15:02:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":2,"date":"2020-05-27 12:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"all":[{"value":1,"date":"2020-05-27 15:02:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 12:24:34.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-05-27 15:02:12.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"},{"updated":"2020-05-27 16:43:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"095c222ef4f16af41d41d8a98faede024feacb40","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-24 13:55:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0773e3e8621e47be9a7be8edfe6ea5ea1c20996","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-24 14:26:13.000000000","message":"Topic set to osa/el8","accounts_in_message":[],"_revision_number":1},{"id":"fee74f564153c310facd0a0956958a574e3980e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 15:50:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b98e8f1d970742b3aaa0653ca00d8d57 : SUCCESS in 8m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/faf2dda7ff54411c96d4b406f8dc80f9 : SUCCESS in 1h 04m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/55c9eb360a0247f5a563e1bec4b5928b : FAILURE in 20m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c0c91c13577641e493607b18aac448c1 : SUCCESS in 1h 28m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecc7cd653f8441dc803bc2e6a4bc5104 : SUCCESS in 1h 29m 29s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b782be43c2324cd2b01e26ac2963449d : FAILURE in 1h 17m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3238f53494be43239286ca07f0cd1e4f : FAILURE in 1h 04m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ccfc72684674af380b1785c4b3947c8 : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":1},{"id":"4bba93ec278b0be5773510beac97bb6353f7fd07","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-24 18:36:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c2faad71360afa3c28a994a9e5be31987efc5eec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 20:11:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/833f3a2500714864b744dd7da89fb568 : SUCCESS in 8m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bf54cd895b504b28a71830bf5afcc958 : SUCCESS in 1h 08m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/28f5d99606f543949fa9cd17517d35d0 : FAILURE in 19m 58s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d592e51351504aec9215a348e8d5be5f : SUCCESS in 1h 27m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a49beed6eec3403a84386c77454ef036 : SUCCESS in 1h 13m 45s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/707eb7e1c1fd4a99abec3690de7b2de5 : FAILURE in 1h 21m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7fab36f7f8114283ab13739292a26b6c : FAILURE in 1h 00m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17fdfb9e199342e69417c39705c9601a : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":2},{"id":"d00aa908f71733db5f3603707da6d1bcc4129fbd","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-04-30 14:12:44.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ee3aedb02f569b002f6921f90f6521194380d4cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 16:17:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5091f8ae359a4c4da4e173f15b3a689f : SUCCESS in 8m 37s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0e357ef06f9b4d438fcfa15971dcbe23 : SUCCESS in 1h 06m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/01ff294461164d038672a4bde1cba631 : FAILURE in 23m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c2085e24316c4e82aa48e19aff3cd375 : SUCCESS in 1h 27m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f45b5865e4614817acb96d1bcafb5360 : SUCCESS in 1h 30m 43s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/158b0a639dd24c2b97ae6f8a85407dde : FAILURE in 27m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09865f6b93834a55a93aaaad05eebf1a : SUCCESS in 1h 57m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3736ff7f79db46d5ae64b3bd76c0819e : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"27018aca8d831971c8d9c66c5a1d942e6a1ed33a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-14 19:23:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3b0253cb02af3381ffe9d9d01edbee15f55d353d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-14 20:02:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"105804fed7fc3e014e753bc42516a6ac37b81f6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 22:17:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/91b5dbca13aa41e3809daa8ca18ef876 : FAILURE in 6m 42s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f8459eca53174d049f35f0da7a8dfaf0 : SUCCESS in 58m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2754b9c37296408eb27075dd4a71eac6 : FAILURE in 18m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd44ab8551c3465ab1396d4b8218cc94 : SUCCESS in 1h 10m 57s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/259d8a5f63f24996abde883cf7e4897d : SUCCESS in 2h 12m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6a11d0a5cb6406d90fed379d577e684 : SUCCESS in 1h 48m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e8268df8c974724b4314c8621510f82 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":5},{"id":"7f758f7956acbb3d7656b0669faa7fe1e496c251","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-15 15:00:13.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"b65ad409f3bc7b0bda2222078358639855987c34","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-15 17:20:11.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"dbb335e8e11bc6d0a2bd8a01e4dceff1b65113b5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-15 20:58:14.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"875fd49fe6b899622955cf3cd86ae68cf04f87b7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-15 20:59:15.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"3acffbd47fff46b83cb2bb23850b4a5e743cfa76","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-15 21:18:46.000000000","message":"Patch Set 9: Workflow-1\n\nwaiting for integrated repo results","accounts_in_message":[],"_revision_number":9},{"id":"6e8d886a0703a3560a321a7f74dc721cf2db50bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 23:34:46.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3a7a5d0f7c344a40aef551e3d307ccb8 : SUCCESS in 10m 17s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0e31a66ad5724dc4aeb1c851a67f24b4 : SUCCESS in 1h 08m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5135e0aca95f4bf09cdffb1f685a8798 : FAILURE in 21m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/056d24d631cf44e396e77b2b6cf5bf2c : SUCCESS in 1h 27m 29s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/26b8f9bad2ce4416bccc53f57d56bcbd : SUCCESS in 2h 27m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90bd8fce90c44d018f24eb823320b84e : SUCCESS in 1h 56m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f26f74dd25c4c21acff12a68cab0962 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":9},{"id":"a5bbed92392ba231f080438089c4572c32c86098","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-25 17:59:11.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"5127a4316a9c50d6c0a17bc960abe8cc7cad1ddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 20:08:30.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/420b9346f2ab40d0be514c01c2589d57 : SUCCESS in 11m 22s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a0c3c8a1f7744ccfb8853f051957bc9e : SUCCESS in 58m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ade88307c2284b289ca293fd0849d3af : FAILURE in 15m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6d0579c34694db28c8d4c694a9ea14f : SUCCESS in 1h 07m 59s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/57b0299b07d94e4eb74f8e1b06b297a3 : SUCCESS in 2h 07m 27s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/304f54409d754edfb49d192f06da0143 : SUCCESS in 1h 37m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05266e8e4dda44b8bf26392d11441282 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":9},{"id":"72b45b28f643873763f42efa18d310bbc60ee0ce","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-26 12:00:20.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"1dfd8734a61a6b41e92dd1cb0b6ed7ff1ace99ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 14:58:51.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bbab47a1e2da4dd38f398dcc9979f4cb : SUCCESS in 7m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/22da2b32cd0945e6a2266c9b608cd8ef : SUCCESS in 1h 09m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/61bd816d6eda451288af64498b148b8a : FAILURE in 25m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4d510b04c844fafa33e2821bb23becd : SUCCESS in 1h 21m 39s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/561fd4364bd540adb4734741c93e87c8 : SUCCESS in 2h 51m 09s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9dd5879aa6cf467c8f4fb44c8c2dd5c4 : SUCCESS in 1h 50m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e9048c042c4467888d07787a563eba7 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":9},{"id":"af53db6e6a6752711a64513bda66bd0f5d42ed92","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-26 18:41:11.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"cfb7d7562d76ecb63efd7c253b0e0d76e17b15a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 21:30:23.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/678241f39ef74015912ec93e87097e5a : SUCCESS in 9m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6759cca7b28347efa03e83f71220b78e : SUCCESS in 1h 23m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8659ac760a8a42cd8d823ecebeb1ade3 : SUCCESS in 1h 40m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/013b9f625f264daaabd45d5c1c3c74d2 : SUCCESS in 1h 25m 57s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/20b915da859441b5bd0732ee91f980d0 : SUCCESS in 2h 41m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5caee8cc8f82468999940c8b314d0da0 : SUCCESS in 1h 59m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/899750bcb9db4a64aeb7cef4a96fd0a3 : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":10},{"id":"d318b1d8ffd4033844140c50567d5ec27b29ee87","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-27 08:30:58.000000000","message":"Patch Set 10: Workflow-1","accounts_in_message":[],"_revision_number":10},{"id":"92c33a3b48e81d4bb55c771e10c64f27b7462c99","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-27 10:20:46.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"26ce3ef11d1812f8c968a885d34512279a484eb3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-27 12:23:58.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"d1cfd3c4f5750d517392d4c0f56d45cf2ca765db","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-27 12:24:34.000000000","message":"Patch Set 11: Code-Review+2\n\nthis one ready too - https://zuul.opendev.org/t/openstack/build/c7f0f4a6190f478ba2ca7bafd694e0c0/log/job-output.txt#10403 passed repo server","accounts_in_message":[],"_revision_number":11},{"id":"9349216aee7821b223f31037a1044c31d2139e4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 13:28:14.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6c0e93ae849e473aa514370a1256e0fa : SUCCESS in 6m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c55d090bbab145ea9b8afb3e894af726 : SUCCESS in 1h 03m 02s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/16297c14b4d74c66a2e80a4487ab8b03 : SUCCESS in 1h 31m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/234de4e797d04d799f3bd905c40ac4f7 : SUCCESS in 1h 20m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f479c657ce1441ed82c2d884fe082d8f : SUCCESS in 2h 59m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b965fe85f1504d3692debae57fd0bd3f : SUCCESS in 1h 53m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48896515f62349c0a5e16cf713d1d729 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":11},{"id":"fa32a63dc658f2b8e372888afc0ba76d47fd28b1","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2020-05-27 15:02:12.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"671c574da7775ad0460692e0a5c42458e2f3b99a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 15:02:23.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"d7ce7fff4b1be2ed9ffd970f4bc18028e16f1ca3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 16:43:49.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2b64bdf03be94acbb5cb8c95ca07a4fc : SUCCESS in 9m 46s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a2d5d9ee3100431686bc9ea4af9035be : SUCCESS in 1h 16m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6a0993e8fbe2430e9ea5842f585baebc : SUCCESS in 1h 36m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5386a6d4aa1a40c0ad0de9daa93c37c6 : SUCCESS in 1h 19m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/718ea7721db745f98bc1578d1de2402b : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":11},{"id":"93831eeab932cd361e194702fbb8bca5d2c0c022","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 16:43:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"8bd95c3ea9b10a933f51b001d7df0602c63ccb7e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 16:45:48.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f5dc98b911049d0b6fab1d805c49e11 : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"eea73c1011d22ef1714a388e80b67be9de905048","revisions":{"4cd058967486627ed5bbf69936e6bc9f492bd3b3":{"kind":"REWORK","_number":1,"created":"2020-02-24 13:55:56.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/32/709532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/32/709532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/1"}}},"commit":{"parents":[{"commit":"48daf9453d27a96e02d597afdcec65bb1b495d49","subject":"tox: Keeping going with docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/48daf9453d27a96e02d597afdcec65bb1b495d49"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-24 13:49:15.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-24 13:49:15.000000000","tz":0},"subject":"WIP Centos-8 support","message":"WIP Centos-8 support\n\nThis disables installation of lsyncd as there is no package for centos-8\n\nChange-Id: Iab8d28c32e534e62759dc2bc72bd6368fbea471d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/4cd058967486627ed5bbf69936e6bc9f492bd3b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/4cd058967486627ed5bbf69936e6bc9f492bd3b3"}]},"branch":"refs/heads/master"},"f79748a3446b4ed5fe2a59bf081677a4c660820f":{"kind":"REWORK","_number":2,"created":"2020-02-24 18:36:58.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/32/709532/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/32/709532/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/2"}}},"commit":{"parents":[{"commit":"48daf9453d27a96e02d597afdcec65bb1b495d49","subject":"tox: Keeping going with docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/48daf9453d27a96e02d597afdcec65bb1b495d49"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-24 13:49:15.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-24 18:36:51.000000000","tz":0},"subject":"WIP Centos-8 support","message":"WIP Centos-8 support\n\nThis disables installation of lsyncd as there is no package for centos-8\n\nChange-Id: Iab8d28c32e534e62759dc2bc72bd6368fbea471d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/f79748a3446b4ed5fe2a59bf081677a4c660820f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/f79748a3446b4ed5fe2a59bf081677a4c660820f"}]},"branch":"refs/heads/master"},"29cc0748e66a25f884bfdd0e679a1adbb80d0487":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-30 14:12:44.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/32/709532/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/32/709532/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/3"}}},"commit":{"parents":[{"commit":"1205049670ba56a8babe273b4def7647ddcb9577","subject":"Cleanup py27 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/1205049670ba56a8babe273b4def7647ddcb9577"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-24 13:49:15.000000000","tz":0},"committer":{"name":"Marc Gariépy (mgariepy)","email":"gariepy.marc@gmail.com","date":"2020-04-30 14:12:44.000000000","tz":0},"subject":"WIP Centos-8 support","message":"WIP Centos-8 support\n\nThis disables installation of lsyncd as there is no package for centos-8\n\nChange-Id: Iab8d28c32e534e62759dc2bc72bd6368fbea471d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/29cc0748e66a25f884bfdd0e679a1adbb80d0487"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/29cc0748e66a25f884bfdd0e679a1adbb80d0487"}]},"branch":"refs/heads/master"},"da0e2beb35445ab8f4209c8a619bf26cbc5199d0":{"kind":"REWORK","_number":4,"created":"2020-05-14 19:23:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/709532/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/32/709532/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/4"}}},"commit":{"parents":[{"commit":"810892848361666716022dc1aa2d7c03cbe54dbd","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/810892848361666716022dc1aa2d7c03cbe54dbd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-24 13:49:15.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-14 19:23:15.000000000","tz":180},"subject":"WIP Centos-8 support","message":"WIP Centos-8 support\n\nThis disables installation of lsyncd as there is no package for centos-8\n\nChange-Id: Iab8d28c32e534e62759dc2bc72bd6368fbea471d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/da0e2beb35445ab8f4209c8a619bf26cbc5199d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/da0e2beb35445ab8f4209c8a619bf26cbc5199d0"}]},"branch":"refs/heads/master"},"b7bc7b2522e26d0b36c50f02dec01be1ab69ae9a":{"kind":"REWORK","_number":5,"created":"2020-05-14 20:02:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/709532/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/32/709532/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/5"}}},"commit":{"parents":[{"commit":"810892848361666716022dc1aa2d7c03cbe54dbd","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/810892848361666716022dc1aa2d7c03cbe54dbd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-24 13:49:15.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-14 20:02:33.000000000","tz":180},"subject":"WIP Centos-8 support","message":"WIP Centos-8 support\n\nThis disables installation of lsyncd as there is no package for centos-8\n\nChange-Id: Iab8d28c32e534e62759dc2bc72bd6368fbea471d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b7bc7b2522e26d0b36c50f02dec01be1ab69ae9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b7bc7b2522e26d0b36c50f02dec01be1ab69ae9a"}]},"branch":"refs/heads/master"},"18fd2e81261b293b04145e61ffc96a25dbfa2175":{"kind":"REWORK","_number":6,"created":"2020-05-15 15:00:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/709532/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/32/709532/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/6"}}},"commit":{"parents":[{"commit":"810892848361666716022dc1aa2d7c03cbe54dbd","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/810892848361666716022dc1aa2d7c03cbe54dbd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-24 13:49:15.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-05-15 15:00:10.000000000","tz":0},"subject":"WIP Centos-8 support","message":"WIP Centos-8 support\n\nThis disables installation of lsyncd as there is no package for centos-8\n\nChange-Id: Iab8d28c32e534e62759dc2bc72bd6368fbea471d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/18fd2e81261b293b04145e61ffc96a25dbfa2175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/18fd2e81261b293b04145e61ffc96a25dbfa2175"}]},"branch":"refs/heads/master"},"3b88d490e2e26a8cbcecc50b6d4bf20b500fd03b":{"kind":"REWORK","_number":7,"created":"2020-05-15 17:20:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/709532/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/32/709532/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/7"}}},"commit":{"parents":[{"commit":"810892848361666716022dc1aa2d7c03cbe54dbd","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/810892848361666716022dc1aa2d7c03cbe54dbd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-24 13:49:15.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-15 17:20:01.000000000","tz":180},"subject":"WIP Centos-8 support","message":"WIP Centos-8 support\n\nThis disables installation of lsyncd as there is no package for centos-8\n\nChange-Id: Iab8d28c32e534e62759dc2bc72bd6368fbea471d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3b88d490e2e26a8cbcecc50b6d4bf20b500fd03b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3b88d490e2e26a8cbcecc50b6d4bf20b500fd03b"}]},"branch":"refs/heads/master"},"df424d51a774bbfff30d415961e558d81dfa88e3":{"kind":"REWORK","_number":8,"created":"2020-05-15 20:58:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/709532/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/32/709532/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/8"}}},"commit":{"parents":[{"commit":"810892848361666716022dc1aa2d7c03cbe54dbd","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/810892848361666716022dc1aa2d7c03cbe54dbd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-24 13:49:15.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-05-15 20:58:09.000000000","tz":0},"subject":"WIP Centos-8 support","message":"WIP Centos-8 support\n\nThis disables installation of lsyncd as there is no package for centos-8\n\nChange-Id: Iab8d28c32e534e62759dc2bc72bd6368fbea471d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/df424d51a774bbfff30d415961e558d81dfa88e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/df424d51a774bbfff30d415961e558d81dfa88e3"}]},"branch":"refs/heads/master"},"bbab400570a090df8a02b3ad8c90dc1944a9a77e":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2020-05-15 20:59:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/709532/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/32/709532/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/9"}}},"commit":{"parents":[{"commit":"810892848361666716022dc1aa2d7c03cbe54dbd","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/810892848361666716022dc1aa2d7c03cbe54dbd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-24 13:49:15.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-05-15 20:59:07.000000000","tz":0},"subject":"Add Centos-8 support","message":"Add Centos-8 support\n\nThis builds lsyncd from source files for CentOS 8 as there is no package\nfor it.\n\nChange-Id: Iab8d28c32e534e62759dc2bc72bd6368fbea471d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/bbab400570a090df8a02b3ad8c90dc1944a9a77e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/bbab400570a090df8a02b3ad8c90dc1944a9a77e"}]},"branch":"refs/heads/master"},"01f2f4c9080285ac755d7ee8b55f01ac5b3c4d38":{"kind":"REWORK","_number":10,"created":"2020-05-26 18:41:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/709532/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/32/709532/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/10"}}},"commit":{"parents":[{"commit":"810892848361666716022dc1aa2d7c03cbe54dbd","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/810892848361666716022dc1aa2d7c03cbe54dbd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-24 13:49:15.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-26 18:41:00.000000000","tz":180},"subject":"Add Centos-8 support","message":"Add Centos-8 support\n\nThis builds lsyncd from source files for CentOS 8 as there is no package\nfor it.\n\nChange-Id: Iab8d28c32e534e62759dc2bc72bd6368fbea471d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/01f2f4c9080285ac755d7ee8b55f01ac5b3c4d38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/01f2f4c9080285ac755d7ee8b55f01ac5b3c4d38"}]},"branch":"refs/heads/master"},"eea73c1011d22ef1714a388e80b67be9de905048":{"kind":"REWORK","_number":11,"created":"2020-05-27 10:20:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/709532/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/32/709532/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/32/709532/11"}}},"commit":{"parents":[{"commit":"810892848361666716022dc1aa2d7c03cbe54dbd","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/810892848361666716022dc1aa2d7c03cbe54dbd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-24 13:49:15.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-05-27 10:20:41.000000000","tz":0},"subject":"Add Centos-8 support","message":"Add Centos-8 support\n\nThis builds lsyncd from source files for CentOS 8 as there is no package\nfor it.\n\nChange-Id: Iab8d28c32e534e62759dc2bc72bd6368fbea471d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/eea73c1011d22ef1714a388e80b67be9de905048"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/eea73c1011d22ef1714a388e80b67be9de905048"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
