)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~714841","triplet_id":"openstack%2Fopenstack-ansible-repo_server~stable%2Ftrain~Iaa2e52b26ba89802e06665ebe43fdf18e515abd7","project":"openstack/openstack-ansible-repo_server","branch":"stable/train","hashtags":[],"change_id":"Iaa2e52b26ba89802e06665ebe43fdf18e515abd7","subject":"Exclude repo_build_global_links_path from sync","status":"MERGED","created":"2020-03-25 06:37:37.000000000","updated":"2020-04-06 15:45:44.000000000","submitted":"2020-04-06 15:42:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714841-1586187752686-87f092a5","meta_rev_id":"1e74ac17c5cf4fd9e8667f83799ffef937187bc2","_number":714841,"virtual_id_number":714841,"owner":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-06 15:42:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-04-06 13:31:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-03 11:32:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2020-04-06 13:31:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 11:32:02.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-04-06 13:31:00.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-04-06 15:42:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96c659358699f3d0a06f0ea6a53e7aca40754e13","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2020-03-25 06:37:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"60ec062a87339df6e5cfe4b5ff94396ccf1a13c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 08:26:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5160c398653e4c81a0e2989633f9c4dc : SUCCESS in 8m 43s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4fdb322687c04887bc6896a27cc1c933 : SUCCESS in 1h 06m 11s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/25c6f03cf1994b7d98d3abf60e492721 : FAILURE in 1h 17m 59s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6079663b1a854401bbbb87b938d7ee08 : RETRY_LIMIT in 7m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/dc16210b0b654665909f3224cd8577d9 : SUCCESS in 1h 26m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c26e571ee914c968112750a24f77266 : SUCCESS in 1h 12m 55s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d0d2326bce224d52bac1121dcd2366f9 : FAILURE in 1h 21m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da65ec6182b941cfaf33fa8da76ec31f : SUCCESS in 1h 43m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09cdd7acc49248b68a149fecdd463313 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6ed9b871cba2de39413ab8b50f05befb7a89cd3e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-02 20:03:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"59aac8fbce362dc7cf380722ed1e1a886a8983bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 22:42:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/55619574eeeb45d5b6c4eaed306e355e : SUCCESS in 7m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5ddd77d3d34a4980b5b83d3f4bfb2719 : SUCCESS in 1h 25m 54s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/987368abfb354b09a9ca3d079f517b62 : SUCCESS in 1h 29m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3138c3274f394b1197796d61616bca13 : SUCCESS in 1h 37m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/71415a3b71da4b1785d3fd9e8a8d2ad0 : SUCCESS in 1h 20m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5611a62a804142d693685369eb945a91 : SUCCESS in 1h 21m 20s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/969e061b66584de89cd5901e1aa9301b : SUCCESS in 2h 26m 47s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41be9ecbcca64ee6bf4a0ccc34b111e6 : SUCCESS in 1h 40m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51d3327810ce4ac2b224f9dfe4c261a9 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"78fad58d381f8a03ecc59835401cd9417573712b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-03 11:32:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7195b0fc9c2f7270e849e4562d4972b1c1121832","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-06 13:31:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87cbee7673512ebe3f64bced259343dbe30de04a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 13:31:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eb1c375c5d795a87ea8d59fb72d85bbbbb9efdd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 15:42:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2a26e9e04a5bf3df5067ee78c7ab009a8d928e77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 15:42:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/df2e753104654bde82107a94476f04c2 : SUCCESS in 11m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3b9411bd97b547eabe892ee9957b8dc4 : SUCCESS in 1h 15m 47s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/cda25e38a33f42d18efafb78f8b20553 : SUCCESS in 1h 19m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a172cbe1f16740219afdb85f6bb3ed82 : SUCCESS in 1h 42m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/19a90f7e732945fca3a1e03f103e4e7b : SUCCESS in 1h 22m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6124379866334cf3bc83cae03b3bbbe1 : SUCCESS in 1h 07m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6105da0c7d3b407a84df280842c3b97b : SUCCESS in 1h 55m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07cc31e5236a4908bf929e23158d9ba3 : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1e74ac17c5cf4fd9e8667f83799ffef937187bc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 15:45:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f41ccd9c41f34bba9be69fef3ca3e905 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7a86cbe69610c1533a14b5cdc3812e0f4bdbf8f","revisions":{"a7a86cbe69610c1533a14b5cdc3812e0f4bdbf8f":{"kind":"REWORK","_number":1,"created":"2020-03-25 06:37:37.000000000","uploader":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"ref":"refs/changes/41/714841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/41/714841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/41/714841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/41/714841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/41/714841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/41/714841/1"}}},"commit":{"parents":[{"commit":"9385ce77c5350da72ee34e11efd8e0d289d14781","subject":"lsyncd: use relative paths in exclude statement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/9385ce77c5350da72ee34e11efd8e0d289d14781"}]}],"author":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2020-02-02 20:53:56.000000000","tz":60},"committer":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2020-03-25 06:37:37.000000000","tz":0},"subject":"Exclude repo_build_global_links_path from sync","message":"Exclude repo_build_global_links_path from sync\n\nThese files are only used by the repo_build during wheel requirements\nand should only be needed and used on the repo_container they\u0027re on.\n\nWhen you\u0027re transitioning from one distribution release to another,\ne.g. xenial -\u003e bionic, syncing the links directory between these\nrepo_containers can break the wheel building in weird ways.\n\nDepends-on: I3bd6d3d987e32ee11c5f1fcb5c1b4b0fc797e7f9\nChange-Id: Iaa2e52b26ba89802e06665ebe43fdf18e515abd7\n(cherry picked from commit 860224c64e69a9ebba4450c26730de1de74532ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/a7a86cbe69610c1533a14b5cdc3812e0f4bdbf8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/a7a86cbe69610c1533a14b5cdc3812e0f4bdbf8f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
