)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~725024","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I42785ee153955eb6e72bef01c26791ae8eec2087","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"git_daemon","hashtags":[],"change_id":"I42785ee153955eb6e72bef01c26791ae8eec2087","subject":"Remove apt-cacher-ng cleanup tasks","status":"MERGED","created":"2020-05-02 16:00:25.000000000","updated":"2020-05-04 11:06:18.000000000","submitted":"2020-05-04 11:04:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725024-1588590287587-e143bcec","meta_rev_id":"db66be46492468d9c1126b19396022a281ff7c6e","_number":725024,"virtual_id_number":725024,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-04 11:04:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-05-04 09:33:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":2,"date":"2020-05-02 22:18:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-05-04 09:33:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-02 22:18:24.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-05-04 09:33:19.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-05-04 11:04:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a53c9e5b5c648928036c39354a917c54e1348f42","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-02 16:00:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18655e5133e6a845b0249dc371d12e7f99441315","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-02 16:08:52.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"617075523001c641d36c7d0067677b79877a6343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 18:01:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9c352240d0f24d0dbf5e0365a2b6d2c4 : SUCCESS in 7m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5fc069a4e97e45089bf23dad13032e54 : SUCCESS in 1h 03m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5acbdbe7e56b4119869fadd24a446eaf : SUCCESS in 1h 45m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f141e976d29c42f49f1e8abb4a629612 : FAILURE in 22m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/798efb3496f44c588bf07d84685f27f0 : SUCCESS in 1h 04m 15s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/02fd807e6d4145ab9f040f9339443b3c : FAILURE in 17m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e886e82c83446508c62c91297e0110f : SUCCESS in 1h 50m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d563df54baca4bf881c2c4c360f8dca0 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a2dcc5d292110359762f8f4505ed670511734e51","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-02 18:49:04.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"721355000961476553f89fd075b24922909133b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 20:57:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0f8c8f5499db452297810163e6e19bbe : SUCCESS in 5m 21s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/02c04076315a4f0384d707e533619d73 : SUCCESS in 1h 07m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0e59c994b7a446f18da0e13577a5baa7 : SUCCESS in 1h 23m 23s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/aee8d0722b874a73bf53b21117e8d741 : SUCCESS in 1h 23m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae4651f03faa4d63a873ae1d6a83fbac : SUCCESS in 1h 12m 53s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b2cd75937e5d4535b327c3a189797bc1 : FAILURE in 19m 22s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc5044e6895f40ffbbebe943890720ac : SUCCESS in 2h 02m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5412df61ea24ca9811108ac774dfa60 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":3},{"id":"11b8dccc33e4fab241c73e353cfdc4468fafd853","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-02 22:18:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1f80dee24732d1b7737dbe8f36e4cf7e67aa0716","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-05-04 09:33:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4b191d3ba989f48cc1e1d58e13c79110c6e2286e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 09:33:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"432c00bc3ea1ba581209cf8e2f5ed3a31abc541b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 11:04:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/aa7a55a64e314ac3bee745a6003f35fe : SUCCESS in 6m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/77f0e9dbf0194f1db554d914521affba : SUCCESS in 1h 04m 58s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bec792c8c2f54a8b98dedf7285080ecd : SUCCESS in 1h 25m 43s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/bf38c3ae552d4f858c71be67b7926d76 : SUCCESS in 1h 25m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a93eecc83f2442b38d2e54feb89462f5 : SUCCESS in 1h 09m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1591c40078524dc6a1de84e95719c297 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":3},{"id":"18f6e7c61758c6e1828f4cd0b5abc2c6f38c8ad6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 11:04:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"47c43724ff18428e59c6d9a32a5449253598ba82","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 11:06:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c56ad9bfe61847f78a36e907c6f63cfb : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cf53ebacbdaed44aecd6cba2510d36636994b51b","revisions":{"6dd83afc0cd9e5fa474e2caec4939d8c61e13fbb":{"kind":"REWORK","_number":1,"created":"2020-05-02 16:00:25.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/24/725024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/24/725024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/24/725024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/24/725024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/24/725024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/24/725024/1"}}},"commit":{"parents":[{"commit":"748d86411be67092bea71939802259d885f1dcde","subject":"Remove git deamon functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/748d86411be67092bea71939802259d885f1dcde"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-02 15:57:05.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-02 15:57:05.000000000","tz":60},"subject":"Remove apt-cacher-ng cleanup tasks","message":"Remove apt-cacher-ng cleanup tasks\n\nThe apt-cache was removed a long time ago and these tasks should\nalready have been removed in the Train release. Remove them now.\nThere is no change for existing deployments as these tasks are no\nlonger required as the apt-cache component of is already retired.\n\nChange-Id: I42785ee153955eb6e72bef01c26791ae8eec2087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/6dd83afc0cd9e5fa474e2caec4939d8c61e13fbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/6dd83afc0cd9e5fa474e2caec4939d8c61e13fbb"}]},"branch":"refs/heads/master"},"a269138580ed7ed3a7eb05335c08ba1d29d298d9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-02 16:08:52.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/24/725024/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/24/725024/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/24/725024/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/24/725024/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/24/725024/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/24/725024/2"}}},"commit":{"parents":[{"commit":"748d86411be67092bea71939802259d885f1dcde","subject":"Remove git deamon functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/748d86411be67092bea71939802259d885f1dcde"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-02 15:57:05.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-02 16:08:50.000000000","tz":0},"subject":"Remove apt-cacher-ng cleanup tasks","message":"Remove apt-cacher-ng cleanup tasks\n\nThe apt-cache was removed a long time ago and these tasks should\nalready have been removed in the Train release. Remove them now.\nThere is no change for existing deployments as these tasks are no\nlonger required as the apt-cache component of th repo server is already\nretired.\n\nChange-Id: I42785ee153955eb6e72bef01c26791ae8eec2087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/a269138580ed7ed3a7eb05335c08ba1d29d298d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/a269138580ed7ed3a7eb05335c08ba1d29d298d9"}]},"branch":"refs/heads/master"},"cf53ebacbdaed44aecd6cba2510d36636994b51b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-05-02 18:49:04.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/24/725024/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/24/725024/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/24/725024/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/24/725024/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/24/725024/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/24/725024/3"}}},"commit":{"parents":[{"commit":"748d86411be67092bea71939802259d885f1dcde","subject":"Remove git deamon functionality","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/748d86411be67092bea71939802259d885f1dcde"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-02 15:57:05.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-02 18:48:59.000000000","tz":0},"subject":"Remove apt-cacher-ng cleanup tasks","message":"Remove apt-cacher-ng cleanup tasks\n\nThe apt-cache was removed a long time ago and these tasks should\nalready have been removed in the Train release. Remove them now.\nThere is no change for existing deployments as these tasks are no\nlonger required as the apt-cache component of the repo server is already\nretired.\n\nChange-Id: I42785ee153955eb6e72bef01c26791ae8eec2087\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/cf53ebacbdaed44aecd6cba2510d36636994b51b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/cf53ebacbdaed44aecd6cba2510d36636994b51b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
