)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~735236","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I73c3c12fe937cda95455dcd5cd708d03a316b392","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"interpreter","hashtags":[],"change_id":"I73c3c12fe937cda95455dcd5cd708d03a316b392","subject":"remove ansible_python_interpreter","status":"MERGED","created":"2020-06-11 19:59:25.000000000","updated":"2020-06-17 13:11:27.000000000","submitted":"2020-06-16 11:39:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735236-1592307585504-035ca625","meta_rev_id":"16fa5238ba0faa461a0797f152fc796f6f75ee95","_number":735236,"virtual_id_number":735236,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-15 15:29:17.000000000","_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-16 11:39:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2020-06-16 09:53:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-15 19:08:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2020-06-16 09:53:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-15 19:08:38.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-06-16 09:53:35.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-06-16 11:39:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a74089cdbde0120b4dc26b79c7c1b24cfccd9668","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-11 19:59:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9bd32d7879a1227e5d5f1671e08d3a7ba6191279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 22:02:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/96c6557ddc564df6b443fde0d865ca3f : SUCCESS in 9m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5aa952ca504b48ec9e4f6e12cc685a65 : SUCCESS in 1h 11m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4c06b1f5cf3e45b19c1609b21df9a02d : SUCCESS in 1h 39m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bd1626d25334d28b163036115d0fed0 : SUCCESS in 1h 16m 02s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/16024476f0ae48cca2e84979a4db474e : FAILURE in 1h 42m 19s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63b4e85b70e04a38a5af47ead577c7bd : FAILURE in 1h 10m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21cdcb4b39324ea0b619568706a531d8 : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e5e057f2eba613c9d1317bac1d10b7b845b3fbb5","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-15 15:29:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8124f066e6f768a336053fd9ad813b54c44f332d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 17:38:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/eb83714834714566bb936c6ea68c0b3b : SUCCESS in 10m 00s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0403da7a10b44002a5f02400104e5cf5 : SUCCESS in 1h 12m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/85ede989a6f54843bde8e259c90e0d86 : SUCCESS in 1h 37m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7defae3d0ac245c7b24308f9192f2cea : SUCCESS in 1h 17m 23s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/757e4fc0b1974d1ead88e1129e2f2e60 : FAILURE in 1h 30m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/321ece1949304380894a2e5039fc2bc5 : SUCCESS in 1h 57m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f2d568f4db84604b7e65e79732dcd96 : SUCCESS in 23m 14s","accounts_in_message":[],"_revision_number":1},{"id":"caddab5956d5bc6b57c223b97cbebe5bb2f75f9b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-15 19:08:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9e607255cfa8329b52debf8b60abe52806b44abc","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-06-16 09:53:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b6d094f50cd0181117d3e11958cdb7fcfc810e89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 09:53:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"92af6e17b1104d22e19fd0d1ac0223013696ad03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 11:39:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bc34225f9e6e478b9368dfa14ce4a851 : SUCCESS in 10m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5644b332910048e5b3ad98378ecc74a0 : SUCCESS in 1h 15m 13s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/908016762b3f405fb6afdfe60a16dbea : SUCCESS in 1h 40m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b59d39eca98b437d9683b94d550a1468 : SUCCESS in 1h 28m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb7d6750fe954c3a9ca4e6dfaf4be15a : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a9c6246d54d025750f5318ff3a9d95ea51e721d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 11:39:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"008fe060f759773101b638bd5ed626a4e8303c3d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 11:44:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9391daba7184f29b752e9137c6a8d09 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"536218103c6961163192c3fb445eb4c63589e430","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 13:11:27.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 5edfb37f72e8bc2f30e0c757812d3a23e6cfd07e","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ead7a35ea913d014fc4655e0e0a05faf5a8d8e84","revisions":{"ead7a35ea913d014fc4655e0e0a05faf5a8d8e84":{"kind":"REWORK","_number":1,"created":"2020-06-11 19:59:25.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/36/735236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/36/735236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/735236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/735236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/735236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/735236/1"}}},"commit":{"parents":[{"commit":"3640957e1b83b1bec62b592221e2e538cf4cf0cf","subject":"Merge \"Restart rsyncd if the configuration file is changed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3640957e1b83b1bec62b592221e2e538cf4cf0cf"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-06-11 18:01:21.000000000","tz":-240},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-06-11 19:59:24.000000000","tz":-240},"subject":"remove ansible_python_interpreter","message":"remove ansible_python_interpreter\n\nansible_python_interpreter is now set to auto in the tests repo.\nit doesn\u0027t need to be overwritten in the test anymore.\n\nDepends-On: https://review.opendev.org/734676/\nChange-Id: I73c3c12fe937cda95455dcd5cd708d03a316b392\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/ead7a35ea913d014fc4655e0e0a05faf5a8d8e84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/ead7a35ea913d014fc4655e0e0a05faf5a8d8e84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
