)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~747613","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~Ia8c1fad5728bd6adec510f8bc1d7bb754393de93","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"osa/el8","hashtags":[],"change_id":"Ia8c1fad5728bd6adec510f8bc1d7bb754393de93","subject":"Add Centos-8 support","status":"MERGED","created":"2020-08-24 04:15:09.000000000","updated":"2020-08-25 13:03:43.000000000","submitted":"2020-08-25 11:58:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747613-1598356698408-f63fd487","meta_rev_id":"0dd26c7a4bd35b7c61c4643e81a55bd50071bfd4","_number":747613,"virtual_id_number":747613,"owner":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-25 11:58:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":2,"date":"2020-08-25 07:28:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-25 10:16:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-25 10:16:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-25 07:28:23.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-08-25 10:16:07.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-08-25 11:58:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a64e5ed4a2a0b36ae72d36f1a3b72b6e37ac947","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-08-24 04:15:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba3db9209ff9f001d322178bab26daad57297874","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 06:19:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c25b3316aaeb4b02818cfae136369b5f : SUCCESS in 6m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b412bb720bfe4100bb1aae98cf9d96b0 : SUCCESS in 1h 00m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1e9ca51320b646f782b37668ef2e4aac : SUCCESS in 1h 15m 49s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/59202c8526a5441899f0fc0efedc494c : FAILURE in 24m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/862fcfbd14bf4bfe8a14a8a0f4eaa9ca : SUCCESS in 1h 07m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3affae15a33c4908bdbbdc7d57136e4a : FAILURE in 48m 20s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3c6497d3362343aeb49e629ec7afc9e9 : SUCCESS in 2h 00m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c0a76db472b4e64ba11f527b04a9507 : SUCCESS in 1h 39m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7391821120524cccbbe0f917817ac1a6 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"db56bd0f639c92dd2771dd23d14ecf3b6ae27ef7","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-08-24 13:00:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d739917c68697f6eb346600c049f17d9d99a07cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 15:38:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2bdb6ef8f52745cc83bf867dbef78005 : SUCCESS in 8m 15s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/30a8c22c7d7c4d34b805512cccdb014e : SUCCESS in 1h 12m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aaf03c5bec8b4ffeacb374260a31bad1 : SUCCESS in 1h 24m 51s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ede652aa43564ed5990bf56bae6ddd34 : FAILURE in 19m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17d97eb96113412c9b950ab3ca7c48a9 : SUCCESS in 1h 19m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f93351f199ce4dd99fc5c5f179b1198d : SUCCESS in 58m 37s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0dedcd4f78cc4e96846c330fc12d6582 : SUCCESS in 2h 21m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17e1a22162cc49759e1bdf316d3108bc : SUCCESS in 1h 44m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15d5e10f7cfa48e2a5e3a6622bc87ad2 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"355695bb6a43b61073a278196e313492b69da766","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-08-25 04:04:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6cafda253264a0cab71efb2b4ba3b5324f8a7ffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 06:31:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7d21391fc6f24651982dcf1f1b3c9ce6 : SUCCESS in 5m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c326dcbf5b73413480f54d67db8f72f3 : SUCCESS in 47m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/209d563dfc4249db99d5c0a6e985d49f : SUCCESS in 56m 37s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7ea35f6dc7a3419ca26b8b4770aa5e68 : SUCCESS in 55m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/039f59c3d77346e8a5e9e2cffac53a3b : SUCCESS in 51m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1b1a6bd25a54cfb81850422614ce118 : SUCCESS in 58m 46s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5e33d5400e804e57991680c1a15e6439 : SUCCESS in 2h 06m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/486a1a9822dd4bbfa274c25981996178 : SUCCESS in 1h 49m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ceb502a4a194a70947703b1861f77a8 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"f2182e94f9b6737aa959f363ca0754e8d73d84c6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-25 07:28:23.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c7d9d64a16b4cc422c84a99eb719f91e8cff312b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-25 10:16:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e7fefc13de02ea3d43f04a620e0a3c9a0b69800b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 10:16:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6590c570f824d74d27db5325b8e63e25c264897b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 11:58:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/760b9803333e456c9c05e9c7a0b0be82 : SUCCESS in 8m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cdb2c22555824a08a85e68709c7fc6a5 : SUCCESS in 1h 00m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6b428c4837e04b278426e02690c65f23 : SUCCESS in 1h 28m 33s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/06aba1b5389142e29d82410508edb35a : SUCCESS in 1h 13m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18a15726961f46eb85b427c4d9549c4a : SUCCESS in 1h 15m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e74b58f9919f43a7aaa1580482b291c9 : SUCCESS in 1h 04m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/336aa81f69334c55a4bff81a0b15ea9a : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e4e29a567c11a461c68dda4586d51a9d03f2c2c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 11:58:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"27a316dcbeac58a52d944db81dc76caa00edb9e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 12:00:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80ca6d712b5e43949cd16525f5839c9b : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2},{"id":"2d9074f1ca31d7c450d49d75d84aa11b27e80f65","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-25 12:02:44.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit af0a553c9fa32058904e52348bfe95eceb9959ca","accounts_in_message":[],"_revision_number":2},{"id":"52c7f44dcb0d8578fce1368f58aa1e59c1034c1e","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-08-25 13:03:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"03b9cc39260f35d1b7b433673874be4a29d4440b","revisions":{"d21adacca119a2fa065eaaadbaf5a0f6f5488531":{"kind":"REWORK","_number":1,"created":"2020-08-24 04:15:09.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/13/747613/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/13/747613/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/13/747613/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/13/747613/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/13/747613/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/13/747613/1"}}},"commit":{"parents":[{"commit":"d4a793fc7863f877744eb6b875f9455c6e1df2ff","subject":"Merge \"Do not install lsyncd when it\u0027s not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d4a793fc7863f877744eb6b875f9455c6e1df2ff"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-08-24 04:10:37.000000000","tz":-240},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-08-24 04:10:37.000000000","tz":-240},"subject":"Add Centos-8 support","message":"Add Centos-8 support\n\nReverting patch https://review.opendev.org/#/c/709532/ which is installing lsyncd using source\nbut now el8 epel repo providing distro rpm for lsyncd so we don\u0027t need to install from source.\nalso changing lsyncd.conf.lua path to /etc/lsyncd/ for centos8 support.\n\nChange-Id: Ia8c1fad5728bd6adec510f8bc1d7bb754393de93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d21adacca119a2fa065eaaadbaf5a0f6f5488531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d21adacca119a2fa065eaaadbaf5a0f6f5488531"}]},"branch":"refs/heads/master"},"03b9cc39260f35d1b7b433673874be4a29d4440b":{"kind":"REWORK","_number":2,"created":"2020-08-25 04:04:21.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/13/747613/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/13/747613/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/13/747613/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/13/747613/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/13/747613/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/13/747613/2"}}},"commit":{"parents":[{"commit":"d4a793fc7863f877744eb6b875f9455c6e1df2ff","subject":"Merge \"Do not install lsyncd when it\u0027s not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d4a793fc7863f877744eb6b875f9455c6e1df2ff"}]}],"author":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-08-24 04:10:37.000000000","tz":-240},"committer":{"name":"Satish Patel","email":"satish.txt@gmail.com","date":"2020-08-25 03:59:47.000000000","tz":-240},"subject":"Add Centos-8 support","message":"Add Centos-8 support\n\nReverting patch https://review.opendev.org/#/c/709532/ which is installing lsyncd using source\nbut now el8 epel repo providing distro rpm for lsyncd so we don\u0027t need to install from source.\nalso changing lsyncd.conf.lua path to /etc/lsyncd/ for centos8 support.\n\nadding rsync-daemon rpm because centos8 islocated daemon and client in seperate package.\n\nChange-Id: Ia8c1fad5728bd6adec510f8bc1d7bb754393de93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/03b9cc39260f35d1b7b433673874be4a29d4440b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/03b9cc39260f35d1b7b433673874be4a29d4440b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
