)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~837587","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I9008680a5f41287599d67f4ce70605b60bccabf3","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"osa-gluster","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2022-05-05 13:42:31.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2022-05-05 13:42:31.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-05-05 11:33:15.000000000","reason":"Dmitriy Rabotyagov replied on the change"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-05-05 13:42:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9008680a5f41287599d67f4ce70605b60bccabf3","subject":"Add facility to store repo contents on a remote mount","status":"MERGED","created":"2022-04-12 15:22:30.000000000","updated":"2022-05-05 13:43:45.000000000","submitted":"2022-05-05 13:42:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837587-osa-gluster","meta_rev_id":"4f5fdc7eba930fbd252e2e7492d5fa1aa2bcf08c","_number":837587,"virtual_id_number":837587,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-05 13:42:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-04-25 18:44:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":2,"date":"2022-05-05 11:33:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2022-05-05 11:33:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-12 15:26:31.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2022-04-12 20:44:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-14 20:07:43.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2022-04-25 18:44:35.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2022-04-26 07:54:43.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2022-05-05 11:33:15.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"6213bc6952ef53e9c0fe41d5c0c73fe6d09451f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-12 15:22:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8504f528d850927bfcc3ad2951408ef51e526a97","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-04-12 15:26:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7402ac5cd0971b160106035e20582ea4ca916b6c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-12 15:40:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d68a5c62fa929b99a909235fba40af19fb26a5ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-12 15:41:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"282e99a26750d8929b7107063a7efa6cd753d8fd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-12 15:45:42.000000000","message":"Topic set to osa-gluster","accounts_in_message":[],"_revision_number":2},{"id":"f36614adad937e1441ead19cf6d04ae32ee97a99","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-12 15:46:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4fe572eec6d5a2c9192680ac34cc307b12099339","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-12 17:14:41.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"1d11ce1d79d52cd4ba4df0048d4d64e62f08a704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 20:44:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f42caf31bc74c92a7437eea6ab7399f : SUCCESS in 13m 50s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d44eaeab795e48e0ae39639a65f21fea : SUCCESS in 1h 26m 14s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/95d81f3e241a4cbd9f3b8d7fe0f8e687 : SUCCESS in 1h 10m 35s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/6791040674f04360b3a8011a3dd940be : SUCCESS in 1h 20m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34e29e6dc71e444092b0da462129abc5 : SUCCESS in 1h 16m 43s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6726c43cce24c45893b6387cc5008c0 : SUCCESS in 2h 07m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aafb6f89676044b9b680db82b0b7ea79 : SUCCESS in 2h 13m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb17902c6c184ac89688598c1f7baeae : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":3},{"id":"2a98f033253e123c1164d834135f7ebcdf9c7ae5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-13 14:07:36.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"77e79b71fd68e4810d65ff657cae61afc6a3abbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 15:25:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55842d12a39e4bec8db2943e1b900dd3 : RETRY_LIMIT in 11m 18s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/8508a02b35a14d98b33e81fbe22316e2 : FAILURE in 33m 17s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b398f812995943b788a7e8fac8839723 : FAILURE in 26m 31s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/973a81ecb8a2496d8bd2cc5e8a6fb4cf : FAILURE in 32m 23s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/807f7d23a6d04cf9b903b012c9e48bd9 : RETRY_LIMIT in 7m 06s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e621f538e66446e5a0910402bb517a7a : FAILURE in 1h 14m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7679dbcfb3b54f55bc05dc03510c525b : FAILURE in 12m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f6a67239b2946c197252af4af72e594 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":4},{"id":"f8b730a9540c47d3515ec6fbc6c30255e2955745","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-13 17:15:02.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"5b75e75a3f076e5717f16bf7c4a11e475bdc8db5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 18:32:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36f0555241d5461f98f9d1c1af9d0915 : SUCCESS in 12m 46s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d74ea1dbd88049798845f731967cecba : FAILURE in 33m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/ec88ef3cf05246f0846f394679e1c0d3 : FAILURE in 26m 51s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/4681e6badee24780b035a4bd8ffd38b0 : FAILURE in 35m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db6642f2378842d5906952e3dfc8f462 : FAILURE in 24m 47s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f1acd2c7ceaf42babffaaecc893e45fe : FAILURE in 1h 13m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/929a8cf8c3324136813ff9f3880ad321 : FAILURE in 1h 04m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa027f922b60443db658b4e40cf4e469 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":5},{"id":"5e319c4a6761f557a7d096627f8b6fa43b689669","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-14 07:41:00.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"b6438ff3c4f4812a9752746611ae33fa3ed08281","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 10:01:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/486aea087e094b689dd6e261b33d5414 : SUCCESS in 14m 07s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2a5f43005374a06ba5f658828530115 : SUCCESS in 1h 24m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/59c3b32fba1e41d6b8175e92f6228daa : SUCCESS in 1h 35m 22s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/f8284653e8ac433b9f9dabc4c4fb0d77 : SUCCESS in 1h 24m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21aebdb9915d4b838db1008672ce82a4 : SUCCESS in 1h 13m 17s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/10bde38fbcf948408bb355fadcae89b0 : SUCCESS in 1h 58m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c003d33f92843018b788fe302d561c8 : SUCCESS in 2h 12m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdb2625a7a074c48b8207d2ef2455dfd : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":6},{"id":"5f6980167093d52e39ec13097666f6e0db6f2259","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-04-14 20:07:43.000000000","message":"Patch Set 6: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c205afdeb4041899d4ac720ea2b8cf4d06235c84","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-14 21:43:56.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6891376dbad5f2d6e812463511750a158ac800ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-19 11:10:42.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"1a15f4f72ffcb3ee317a5368c232758d8d862ec2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-19 11:10:51.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c9dffad4e9c261da561e38e33bd2188b9b9d6510","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 13:31:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61806f95c76647e9bcfc09820c4fa405 : SUCCESS in 14m 49s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d0bdd1b5170425792f0af5197658e96 : SUCCESS in 1h 27m 44s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4f263c616ce04a8e80e123d2074cebe3 : SUCCESS in 1h 05m 26s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/c664f0c05c0b4bae885d4c741f90ed5b : SUCCESS in 1h 28m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5f728847fad4ba19820352c07bcf44e : SUCCESS in 1h 09m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b47031fa555f4d4196253b92ab798332 : SUCCESS in 1h 58m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a45fbcae941b4cfd8466b087c8655428 : SUCCESS in 2h 11m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a741a811414d453cbf2af184cef0d6f8 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":7},{"id":"3acf491c15a06e20a49c242043412d04f05cce76","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-20 06:55:31.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"d952155afd5203f70c89ea7b7010b96feddd606c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-20 06:55:33.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"1f6b763449cdb27defac40e7f664c70440f9df76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 08:27:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39f0bc25732b49738fd1fa68bf779abe : SUCCESS in 13m 23s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a452ed185514324b644355b71ad42b2 : SUCCESS in 1h 19m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/279f95c7cff1446fae3d26d8ce1d744f : SUCCESS in 1h 07m 02s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/63dd6b8d99d64e5c9feae7627f8fefde : SUCCESS in 1h 26m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60108fb3fe7f41a59a20bbcea367d96a : SUCCESS in 1h 12m 34s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/502d61ff8ec44fb3ac7eea2844923fbb : FAILURE in 17m 10s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64ef19c785bc4ffe9b6d3344a107fd56 : FAILURE in 12m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aea7b004f9f46448fded430454791e8 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":8},{"id":"49d495265054feb8cf6056ad181594dc4fa9b568","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-25 15:19:24.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9e507640ee342fb4ac158d26a583e8871c05a32f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 17:28:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/795a1a91cfa648a9a72fe8ffe39c014f : SUCCESS in 12m 56s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0087aec7e386460b85afed016ec7c952 : SUCCESS in 1h 15m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/96c11343a7e14cffb652e4e57fea4bd0 : SUCCESS in 1h 07m 01s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/ab3199a14587413d92acad6fd63acbca : SUCCESS in 1h 24m 09s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a3f491accb54d5c91ac4c535e14a46e : SUCCESS in 1h 07m 11s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a50114f9a5b04af3b2b16b5e031fbae7 : SUCCESS in 1h 52m 17s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb67f4ab2dbb464db7750e940fd62b37 : SUCCESS in 1h 58m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9db40d618e114e8db5ffef1aa26f490c : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":8},{"id":"18a9dceed39ccdef3ecd5d892795755746dd8190","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-04-25 18:44:35.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"4e0a79d800db01231ddf8a243e632d857a10fe4f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-05 11:33:15.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"8005d02980eef73a9902ac12cff476bca5874f47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 11:35:54.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"7278ec4641029a477b43926630ec734f56868d4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 13:42:31.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ca4e67ec6524447891808d129c7e853 : SUCCESS in 13m 51s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/be298fef12074dd2b50334281041a74a : SUCCESS in 1h 30m 24s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1d8d394c1e14405cae2f63213e96178b : SUCCESS in 1h 08m 30s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/afedfbdad98b497b805688681c9d08ae : SUCCESS in 1h 24m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa36766ee07047ab87995853d62d94fb : SUCCESS in 1h 15m 22s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b42035163a284b37a18afe226b41b1b4 : SUCCESS in 2h 02m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be4e8000b9454d089b1e23748031391e : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":8},{"id":"6823f5ee8827019c00d0d9082293cbbf457975ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 13:42:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"4f5fdc7eba930fbd252e2e7492d5fa1aa2bcf08c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 13:43:45.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bae05f440f0447c9ee53df951dec1c7 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"c966363bd44f0e9032b35406d75fa61d46aad3cb","revisions":{"10a0158f0abc135525ae909b7ad02a339a7aef25":{"kind":"REWORK","_number":1,"created":"2022-04-12 15:22:30.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/87/837587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/87/837587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/1"}}},"commit":{"parents":[{"commit":"5aefc76d479d88f7b46da4d6630669ccf2fd4d95","subject":"Merge \"Use ssh_keypairs role to generate keys for repo sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5aefc76d479d88f7b46da4d6630669ccf2fd4d95"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-12 15:21:48.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-12 15:21:48.000000000","tz":60},"subject":"Add facility to store repo contents on a remote mount","message":"Add facility to store repo contents on a remote mount\n\nChange-Id: I9008680a5f41287599d67f4ce70605b60bccabf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/10a0158f0abc135525ae909b7ad02a339a7aef25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/10a0158f0abc135525ae909b7ad02a339a7aef25"}]},"branch":"refs/heads/master"},"1659f63cb62c2f6a5b55073c004f75d390a675b5":{"kind":"REWORK","_number":2,"created":"2022-04-12 15:41:02.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/87/837587/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/87/837587/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/2"}}},"commit":{"parents":[{"commit":"5aefc76d479d88f7b46da4d6630669ccf2fd4d95","subject":"Merge \"Use ssh_keypairs role to generate keys for repo sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5aefc76d479d88f7b46da4d6630669ccf2fd4d95"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-12 15:21:48.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-12 15:40:43.000000000","tz":0},"subject":"Add facility to store repo contents on a remote mount","message":"Add facility to store repo contents on a remote mount\n\nChange-Id: I9008680a5f41287599d67f4ce70605b60bccabf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/1659f63cb62c2f6a5b55073c004f75d390a675b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/1659f63cb62c2f6a5b55073c004f75d390a675b5"}]},"branch":"refs/heads/master"},"de178f5ff7549930c10967e2a47360aae319b53c":{"kind":"REWORK","_number":3,"created":"2022-04-12 17:14:41.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/87/837587/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/87/837587/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/3"}}},"commit":{"parents":[{"commit":"5aefc76d479d88f7b46da4d6630669ccf2fd4d95","subject":"Merge \"Use ssh_keypairs role to generate keys for repo sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5aefc76d479d88f7b46da4d6630669ccf2fd4d95"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-12 15:21:48.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-12 17:14:41.000000000","tz":0},"subject":"Add facility to store repo contents on a remote mount","message":"Add facility to store repo contents on a remote mount\n\nChange-Id: I9008680a5f41287599d67f4ce70605b60bccabf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/de178f5ff7549930c10967e2a47360aae319b53c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/de178f5ff7549930c10967e2a47360aae319b53c"}]},"branch":"refs/heads/master"},"17845131f8c889fc523dc363951f0784c8ea62e5":{"kind":"REWORK","_number":4,"created":"2022-04-13 14:07:36.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/87/837587/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/87/837587/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/4"}}},"commit":{"parents":[{"commit":"5aefc76d479d88f7b46da4d6630669ccf2fd4d95","subject":"Merge \"Use ssh_keypairs role to generate keys for repo sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5aefc76d479d88f7b46da4d6630669ccf2fd4d95"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-12 15:21:48.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-13 14:07:30.000000000","tz":0},"subject":"Add facility to store repo contents on a remote mount","message":"Add facility to store repo contents on a remote mount\n\nChange-Id: I9008680a5f41287599d67f4ce70605b60bccabf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/17845131f8c889fc523dc363951f0784c8ea62e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/17845131f8c889fc523dc363951f0784c8ea62e5"}]},"branch":"refs/heads/master"},"2936fa0c331fc229d0d4983fc2f05ddb143f1402":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-04-13 17:15:02.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/87/837587/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/87/837587/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/5"}}},"commit":{"parents":[{"commit":"5aefc76d479d88f7b46da4d6630669ccf2fd4d95","subject":"Merge \"Use ssh_keypairs role to generate keys for repo sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5aefc76d479d88f7b46da4d6630669ccf2fd4d95"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-12 15:21:48.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-13 17:15:02.000000000","tz":0},"subject":"Add facility to store repo contents on a remote mount","message":"Add facility to store repo contents on a remote mount\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/837706\nChange-Id: I9008680a5f41287599d67f4ce70605b60bccabf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2936fa0c331fc229d0d4983fc2f05ddb143f1402"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2936fa0c331fc229d0d4983fc2f05ddb143f1402"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2202a33e1775de3a40673113b7a7023b167c3211":{"kind":"REWORK","_number":6,"created":"2022-04-14 07:41:00.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/87/837587/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/87/837587/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/6"}}},"commit":{"parents":[{"commit":"5aefc76d479d88f7b46da4d6630669ccf2fd4d95","subject":"Merge \"Use ssh_keypairs role to generate keys for repo sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5aefc76d479d88f7b46da4d6630669ccf2fd4d95"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-12 15:21:48.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 07:40:56.000000000","tz":0},"subject":"Add facility to store repo contents on a remote mount","message":"Add facility to store repo contents on a remote mount\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/837706\nChange-Id: I9008680a5f41287599d67f4ce70605b60bccabf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2202a33e1775de3a40673113b7a7023b167c3211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2202a33e1775de3a40673113b7a7023b167c3211"}]},"branch":"refs/heads/master"},"65af00ca45684b02d434a00273c60a9c87dc89a5":{"kind":"REWORK","_number":7,"created":"2022-04-19 11:10:42.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/87/837587/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/87/837587/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/7"}}},"commit":{"parents":[{"commit":"5aefc76d479d88f7b46da4d6630669ccf2fd4d95","subject":"Merge \"Use ssh_keypairs role to generate keys for repo sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5aefc76d479d88f7b46da4d6630669ccf2fd4d95"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-12 15:21:48.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-19 11:10:42.000000000","tz":0},"subject":"Add facility to store repo contents on a remote mount","message":"Add facility to store repo contents on a remote mount\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/837706\nChange-Id: I9008680a5f41287599d67f4ce70605b60bccabf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/65af00ca45684b02d434a00273c60a9c87dc89a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/65af00ca45684b02d434a00273c60a9c87dc89a5"}]},"branch":"refs/heads/master"},"c966363bd44f0e9032b35406d75fa61d46aad3cb":{"kind":"REWORK","_number":8,"created":"2022-04-20 06:55:33.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/87/837587/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/87/837587/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/87/837587/8"}}},"commit":{"parents":[{"commit":"5aefc76d479d88f7b46da4d6630669ccf2fd4d95","subject":"Merge \"Use ssh_keypairs role to generate keys for repo sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5aefc76d479d88f7b46da4d6630669ccf2fd4d95"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-12 15:21:48.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-20 06:55:06.000000000","tz":0},"subject":"Add facility to store repo contents on a remote mount","message":"Add facility to store repo contents on a remote mount\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/837706\nChange-Id: I9008680a5f41287599d67f4ce70605b60bccabf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
