)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~837859","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I3109bd891d4c6b3522f5f741d9970093b1e882c8","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"osa-gluster","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2022-05-27 17:32:45.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-05-27 17:32:45.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-05-27 17:32:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3109bd891d4c6b3522f5f741d9970093b1e882c8","subject":"Clean up legacy lsycnd, rsync and ssh key config","status":"MERGED","created":"2022-04-14 12:09:25.000000000","updated":"2022-05-27 17:35:10.000000000","submitted":"2022-05-27 17:32:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837859-osa-gluster","meta_rev_id":"5b3c429b18a4836d29c6fc185468306ad79b7af4","_number":837859,"virtual_id_number":837859,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-27 17:32:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-05-25 12:27:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2022-05-25 20:00:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2022-05-25 20:00:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-14 15:47:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-25 18:48:49.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"CC"},{"updated":"2022-05-17 20:38:16.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"CC"},{"updated":"2022-05-20 19:05:48.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2022-05-25 12:27:30.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2022-05-25 20:00:05.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"9a317922108a2f83b475f343b9a8c59a8a90592f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-14 12:09:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40794ad84ac1c9a47fbfcbc07871b12dd56fcafd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-14 13:29:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"feb95716f1e0435083d27ab10a5526de60652803","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-14 13:53:10.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"908606a49bd94d62229f50b73d3c2394a323fe0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-14 14:20:36.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"be329f3ca1106aff3d060e30126ccdbeb8d73b5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-14 14:21:10.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"1039f62d3f8e5e503b6d78c12133f5349976a15e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 15:47:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47126a2497b3471188bf09983a2cc871 : FAILURE in 12m 47s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a57a079e18e4611bd946f02883d1784 : FAILURE in 31m 33s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/13bdd31579064e13a1f6de568e02a495 : FAILURE in 26m 26s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b247754c3e2449a68864f398e53376d5 : FAILURE in 30m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f350ad29355b4600ad6765d346bb5f20 : FAILURE in 28m 46s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a8a6c82798b42d28fe6444e26b32cee : FAILURE in 1h 10m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff620878f5044d7888c1c2d130063032 : POST_FAILURE in 1h 16m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/994f13df8f1844bfbaf5e8984f812993 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":5},{"id":"b4db368091fea8c5135e6113068a4f442b88d610","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-19 11:11:04.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"dcb9187ab2eca55940dc3b21e66712e555432e48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 12:36:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7a1043ef7b64e3d990570d4846023d2 : FAILURE in 16m 05s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/53ba89498e1a4fcea864d7e535845f68 : FAILURE in 35m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/01b7f1993f874090be0cbce4ff979e71 : FAILURE in 29m 30s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/0b4cef1867c44fa6a6592141fe1fbd43 : FAILURE in 35m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fbee788c47948ed8888268a0b8de8dd : FAILURE in 28m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/44ce7809cf5447a38a391f7f714d6dbb : FAILURE in 1h 16m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68d9b7f48fe047e8b910c940c26c1b0e : POST_FAILURE in 1h 14m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/503277dc922b4c5f888a85e14e496b15 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":6},{"id":"c83ca00ba7a8e81b5ee9ad424aee7c04b136076f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-20 06:55:43.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"ea23341bf979f8e86c8b6105fc0e3f27efe1913d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 07:39:57.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/439b5a0ce18b4ac498efc82416f49895 : FAILURE in 14m 22s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f46c6f902eea4b80ae76942a42e02562 : FAILURE in 31m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/141c1bba91c741d9a91a72d0cdcad4e5 : FAILURE in 24m 41s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/19d47d22f2e14125afca680db69bedc2 : FAILURE in 35m 18s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de363845b48a4ab28d537c45c39b4645 : FAILURE in 28m 34s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d59973741ef54fcaacf1d4e85a20bf8c : FAILURE in 17m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/460be4805f4f49fba9f66e21d7b54108 : FAILURE in 15m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc32e2bb984b4c4eabffa0e81fd3442b : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":7},{"id":"e72dfa04d80b2af30877fd2f39c0dc919aa3fc45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-20 07:44:17.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"7056777e720f3fbbc9dd99451ac60bfd68170c93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 08:27:19.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07804487146241faa0e9b0b70db86779 : FAILURE in 14m 34s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec3bcbd5b105439cbb349258bb1647a7 : FAILURE in 29m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/70fbe34b30ac4b11b3f92d18cddc6e65 : FAILURE in 27m 32s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/77016e8ba17546339876a2cd9f6ac8af : FAILURE in 35m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06f20ef081004ad7bada093cd5f73862 : FAILURE in 26m 37s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/692bf162c89040368689b1a6aba6c528 : FAILURE in 17m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/682d43e1b1134f26aaacc59b9ee522c4 : FAILURE in 16m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/897b944329df443ca5689823925da715 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":8},{"id":"e77c3f5633eea8ad47fdaf951cf480d68fb8aa27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-20 08:46:29.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"532f4f098898e6f9940858952ef1e31247b9da51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 10:29:21.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e5fc34935dc4a9cab0e6b420e497230 : SUCCESS in 13m 51s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b42c2572abd4901ba8d01abd2f5b7a1 : SUCCESS in 1h 19m 37s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/2511c56c94cf4a04a9b2d10da154b01a : SUCCESS in 1h 00m 19s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e853a22ffffa41278635ae6855a3fa84 : SUCCESS in 1h 33m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea388d8b11324bd78e5842e9d32e80e3 : SUCCESS in 1h 07m 06s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7d3670b3c1814edf870d466e603604b6 : FAILURE in 17m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99e4516c0af44251bd070d468d73cf71 : FAILURE in 15m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5001a2eac58340a888f2bb8953a5b761 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":9},{"id":"9511cc58af389f964da067bf4e4aeec85387b075","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-04-25 18:48:49.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a272ac20fc98482ec1771b0c22da56b823f8391b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 20:53:10.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cac48aa056444dabbb2c50ff6da5f90 : SUCCESS in 14m 39s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/eeabd5cf1f0c47b29d3945110f586015 : SUCCESS in 1h 36m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/04e0e46015594418a79efc2cc8e3af86 : SUCCESS in 1h 08m 40s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e7aa14a5f6f94199a869c03be818a703 : SUCCESS in 1h 20m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6697ada33e6a42ccb4068f0cc078bf1e : SUCCESS in 1h 08m 22s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d73e0ea9b5c04b3e9d24ff9b38782cb3 : SUCCESS in 1h 59m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7450e998de254ffbb2974538b7cc436e : FAILURE in 1h 02m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6192859f27934c17aa6dff903ce6d425 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":9},{"id":"47e500b92101b38b5568485bd722665f12e034e2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-26 10:27:10.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a5c28eab2aab40900721441e3e559ff00ae0ebb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-26 10:35:06.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"c10bf0dc2c00f319cc9ab7e1e783f11cfacbfd95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 12:22:24.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09247ca316d64c06bfd78c2975130611 : SUCCESS in 12m 50s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d45c0ec17e64860ac87e02cda3f3b7f : SUCCESS in 1h 29m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/ae6b20fa83f042e7bd842155179b0b1f : SUCCESS in 58m 13s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/c134220514b44fa8afe8f2f5138366c9 : SUCCESS in 1h 24m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4d86d7d46f0444087ffee40fddbcb28 : SUCCESS in 1h 25m 54s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b27c7b72373454cbd3285d178d4d575 : SUCCESS in 1h 41m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdf90087319f4e498eab938256dd0383 : FAILURE in 57m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1e1ffbfc60b42b2a96d45af1086b95d : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":10},{"id":"5e0ca097384d770bb2cc719c1d8ab8a3fb86fe6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-26 16:33:20.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"841d35c49cdb86fafa708a90b24b7128ade96ad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 17:50:54.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a001250cc3454faa85988b5a674c4a64 : FAILURE in 15m 22s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a081f84f8d464b08852986ec628d1a12 : FAILURE in 37m 14s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e613ef4401a14a9c9d8b653f7ada50b6 : FAILURE in 30m 52s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/448c8262715e496fa0c8df91c70f4d9e : FAILURE in 33m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd7796340d784b08b9a094ac9aac142a : FAILURE in 29m 36s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8878e9a062e84f0b8efd539747901148 : FAILURE in 1h 10m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/520a5e023f1b47c1b4e6c86f88b0180c : FAILURE in 1h 01m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f5ce5ae022445fe8526e2cd39d6105e : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":11},{"id":"7d3721a5086009e2963f11d16eabc24921fbcdf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-27 07:47:17.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"7c8c68aa5c4c649faab1a2cbfe97dcfba9d41061","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-27 07:48:16.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"7d8faa26907514faa6ded325356c3f498e7f0fda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 09:09:08.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c132dfdf080469680a1f2437faf0b33 : FAILURE in 14m 20s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/851062ff995940f79abd0e552e17e432 : FAILURE in 35m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b3b509b647624cf1a4933f197afaa33e : FAILURE in 27m 31s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/3fe34c192d0542c1948563979b3c3f53 : FAILURE in 36m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/465eefd032224780a672dba2cfa99f83 : FAILURE in 32m 35s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c4c26f80617845648c8effad79567071 : FAILURE in 1h 15m 22s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ebde560b3dd447e90f72fabf034ab07 : FAILURE in 1h 03m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d900bc88ef104c8da00e658d5d9bee46 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":13},{"id":"a74989dcd05c501ea2285247e8ad9e7e36f8d1af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-27 11:38:25.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"879a9a009b69b2e7692138290452b5ad19c24c17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 13:08:44.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5b49d51b4d4402ca253cf70001f2435 : FAILURE in 11m 16s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/663d6895f6484f0d8598b4da0c976228 : FAILURE in 42m 33s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/56f68a36b7d44c53bd17d37fed82f551 : FAILURE in 31m 24s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/9a4ca7f457024eb9bb873b42009874b7 : FAILURE in 31m 14s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fae36f0de4a43c19cc1f37e730b06d2 : FAILURE in 27m 22s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/57ee61b0d51447e9be86d91c0cec3429 : FAILURE in 1h 20m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dc311039c44401f9c546198fd1b5912 : FAILURE in 56m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb06b088837840ce93c1dae069122e26 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":14},{"id":"b8f2abc12ba00fa49c63b0ccf752dec76a2b19b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-27 13:20:55.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"dad7e8b2f7f173607c7512f588fe2fcd95de338e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-27 13:32:46.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"fdd1b08951d1cfa9fa052cd4fed21d7d680a592e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:38:38.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7630d13e1781473ea208b74c0c352485 : FAILURE in 15m 45s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0fc526a3086843ca83b1c05c7ad62a1c : SUCCESS in 1h 33m 19s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1e8ab61f6e4d4018a13799b244e39f23 : SUCCESS in 1h 01m 37s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b98d9edee2e7401fbfb704545e70063f : SUCCESS in 1h 16m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f07ab7f869af494f84dbe809dff7e4bb : SUCCESS in 1h 23m 17s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/78a8cf650c994836923746fe82c63ced : SUCCESS in 1h 56m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6e5b82cd0d14ab18be58675f3a8eba2 : FAILURE in 1h 08m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3052378ad09c4191bc78fac7621f1801 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":16},{"id":"41845000b239122a0b391a437f4187816a4ab5de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-27 21:12:46.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"a3002b95c7049183e6c4ba1d8429ed143797d9cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 23:23:43.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fe6fd56fb224f178b29b1ac4e57967c : SUCCESS in 14m 04s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9f180d5a6c347e08b032baf9d610aac : SUCCESS in 1h 41m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/fc9e1e8a941847cf88cf2d615d808323 : SUCCESS in 1h 10m 39s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/2d4f1d0cfd7244678bf616787d85cb51 : SUCCESS in 1h 20m 41s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ab3041f82e24a07a68351f0f4ff9d27 : SUCCESS in 1h 32m 48s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0054fa1386b4befbdcf3545e78c07cb : SUCCESS in 2h 07m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/197403d79835467ea07a6f9d6c26b6f8 : FAILURE in 1h 09m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/943522f2c59b4d80baf8723fc9edab43 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":17},{"id":"f7cdda4d1996cb45458c25794260b782a2f2c73a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-29 18:55:42.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"2c0bdccf474bb56d5feb0e5b169d974247882dd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 21:26:38.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffce03258c884feb8a02618c67ca95ce : SUCCESS in 19m 58s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c7e3394ccfe4416789b62d00247c8a88 : SUCCESS in 1h 36m 46s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/56840d6258784371b033ab16d682247b : SUCCESS in 1h 16m 09s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/1050a0e5185643c98df601877e58033b : SUCCESS in 1h 25m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33fb075ed3504a738e89aa1cb96228b1 : SUCCESS in 1h 15m 38s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/02a75f2b9fe54510a96d858ab7a59610 : SUCCESS in 2h 21m 21s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/319caa33ff3a4a95a7f1e9c4fc8896b2 : FAILURE in 1h 00m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c4f6c5f456b489abb204a865dee8c95 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":18},{"id":"cafc894f5f567c56592f93c144b3397ef64031ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-04 09:51:48.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"fc959bbb19cc41f75a7eb40d6e0ae3e6b6ede8b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 11:58:43.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5008b1b47ed3492e981ebd137fba9f27 : SUCCESS in 14m 30s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b511603ed61e40daba7410b691be5857 : FAILURE in 56m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1c3ca91cffd8411289938f9ad37085a0 : FAILURE in 44m 55s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/982dddfea11e424482dd6b485d6172da : FAILURE in 1h 00m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2015e15ebf44b0ca5116c4f69c0c050 : FAILURE in 45m 48s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f756f6cf90264582bb4132f458e7e788 : SUCCESS in 1h 59m 14s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28bcd9e0c75546128b105846b421d2c6 : POST_FAILURE in 58m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bb48884074144d09d87840f578951bb : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":19},{"id":"af93572b8309d92d802e0a05f71fdf053c54fc2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-05 11:49:12.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"ecafb689f747d8093e08df405b85fec13eda624a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-05 11:49:31.000000000","message":"Patch Set 21: Patch Set 20 was rebased","accounts_in_message":[],"_revision_number":21},{"id":"79d73b26f132fc5666686117fc1fc3e51543fa4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 13:51:26.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdd049a6c363463fbdc5fdaee8672c47 : SUCCESS in 12m 22s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d62670582fbd490fbd024655f30be700 : FAILURE in 54m 04s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/f4306c47e8424db3a38d22c4f4eeda26 : FAILURE in 42m 26s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/63cd388bba114bcfb38d3cdceb4d3ec1 : FAILURE in 51m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/498945e69de94aa5b95b3ff043cdc962 : FAILURE in 46m 50s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/83b8b9a9ef514aa9aa6900e4682520bf : SUCCESS in 1h 53m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/540d9a25c09048d2990d5c3900b2f84c : POST_FAILURE in 57m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4572d298f69e4c829f952801797d5427 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":21},{"id":"9c59b8bdbfe0bb095ad2b2e8c3b176e3756ddfcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-09 06:46:44.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"72ed59fee99dd92ece003a6deb2f453f270b3632","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 09:01:59.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/838f5a86e49049ed932173334d5d267a : SUCCESS in 14m 18s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/24a8386371064600b8bc400f928f39ae : FAILURE in 53m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/fb06ec5ed79644e286ef5ae01157b49c : FAILURE in 26m 49s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/165b73adbda34ac78e106d0af233684c : FAILURE in 1h 02m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32dc91e39e6648d3977db3e19103dc46 : FAILURE in 23m 08s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/53f8f40e6cef45ccaa94d98397eb4bd0 : SUCCESS in 2h 08m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36c43c53489943f49586f83ce1d22895 : POST_FAILURE in 58m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbdb38d2df1741c38f241079cde80c70 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":22},{"id":"aef594e55bfc693d61b489826e68f5e1e711bb09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-12 16:39:51.000000000","message":"Uploaded patch set 23: New patch set was added with same tree, parent, and commit message as Patch Set 22.","accounts_in_message":[],"_revision_number":23},{"id":"7268a18dc0b8b96e35f98e89f9d6e6d1fd166ef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 18:38:30.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a63060f4e2dc44debdd1fe1851292bef : SUCCESS in 23m 42s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b75e21f44a7b45689d827358371b8e88 : FAILURE in 1h 01m 15s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5d63b64c926d40e3ade714f13829c20c : FAILURE in 28m 33s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/aab0eacaa9cd4eca886b8da2af8c2e0e : FAILURE in 53m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce2fc5bbe42b4dd1bc1141c6cb6d20aa : FAILURE in 30m 55s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5bfe963ca6d94c30b98818765145a389 : SUCCESS in 1h 51m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/decaa53f6ab44a52bfa0c85ec3921bae : POST_FAILURE in 1h 05m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a96c69aec1ba4f3897abe118ed7461ae : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":23},{"id":"a28d1df5bb5fce428e143970638135a6eece57f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-16 11:03:19.000000000","message":"Patch Set 24: Patch Set 23 was rebased","accounts_in_message":[],"_revision_number":24},{"id":"32f2083d048f88b6edbc19ffc10593333b2bfd2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 12:57:24.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73a61f34be664cc5945eeb639c13f780 : SUCCESS in 19m 01s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c950d71c09a0473d978e40efbc10e108 : FAILURE in 55m 22s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/161aff42cd8546a8b7d629cf53f57669 : FAILURE in 28m 20s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/70f5f70d52e54aceb7018300db7cc668 : FAILURE in 1h 02m 25s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ad1014c977f4c7db1009cb2633d5149 : FAILURE in 29m 53s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/05c25a45e7954caca73844ab7fb07ea8 : SUCCESS in 1h 45m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd7106a934bf45d5804161eae843aaab : POST_FAILURE in 1h 02m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44ea77d2a114090829efdaefdf606e7 : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":24},{"id":"a5fcd4bc613f333e0d62bdbd7194fa599facd6f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-16 15:30:15.000000000","message":"Patch Set 25: Patch Set 24 was rebased","accounts_in_message":[],"_revision_number":25},{"id":"1bc01f4031b594164d1df2351c9fdf161065f598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 17:51:03.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4f466ed1e8f48ad8cff0df10ac4ca0f : SUCCESS in 13m 43s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c704dce3f9f1442388aa7e66a1b1985c : FAILURE in 52m 20s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/f868f4520fd44c2a84d0adede7dbd21b : FAILURE in 40m 30s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d87b3b23eed04b1ca79e68cd2219af12 : FAILURE in 1h 01m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57f3b5cd2ddd4f71af547a473d153aca : FAILURE in 44m 08s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ff1decd3e08c439ca4ee4de982be3a53 : SUCCESS in 2h 12m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c71b2d8fff341c080863a22f5d58f6b : POST_FAILURE in 57m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d0c44b146204b3db3bbe3f3d0ca4c82 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":25},{"id":"6ec4ac4052f95deb7528d062fb38e47a11523743","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-17 11:06:14.000000000","message":"Patch Set 26: Patch Set 25 was rebased","accounts_in_message":[],"_revision_number":26},{"id":"8b46e57e32f9885f3dd2155d3b46f387f47358e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 13:29:18.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0023b074ae3d4c42ab7fc7a53c830e2a : SUCCESS in 17m 14s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/b4e572f14e5a4ec491e8168f7ed112d0 : SUCCESS in 1h 24m 01s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1bfcaf999d2b4043afd0c2ca0c8321ba : SUCCESS in 1h 03m 58s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/4e671f8dd45e498daa7e1ae48a853afc : SUCCESS in 1h 25m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/207a4cc9f95d4eff8e6cce6cac6b05c9 : SUCCESS in 1h 12m 46s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ddf8aa8886f422797d8fb1156065df1 : SUCCESS in 2h 14m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e835dde20944b3993c4b2c12672a16b : POST_FAILURE in 1h 04m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23c82822de4945cb8dd85c26a51755ae : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":26},{"id":"99a6987a8fad026160564b5f85b239fa2ab1f16c","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-05-17 20:38:16.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"7bf0d0aab89baab5ecd183a8c8f44deefe742589","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-18 11:41:14.000000000","message":"Patch Set 27: Published edit on patch set 26.","accounts_in_message":[],"_revision_number":27},{"id":"89c80a8d8aeb249857ae6b8f7fe7c8a91365ba4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-18 11:41:18.000000000","message":"Patch Set 28: Patch Set 27 was rebased","accounts_in_message":[],"_revision_number":28},{"id":"2ed902251444cc5904e1728047d80d12c9fa6b06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 13:59:27.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4c39e4c398f4cffbf139c2a3f14dca2 : SUCCESS in 14m 52s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/848b8acb69fb4b6696c02d437817a000 : FAILURE in 58m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1b0cee518fca4d47a084e5b9c631a59f : FAILURE in 49m 04s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/67307c2b39a043049546aed5dcbd5af4 : FAILURE in 57m 34s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afd4392d014e4262bc90ec500791fea2 : FAILURE in 47m 03s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/16e7cd79b1c942c89ebbaa0ab10f3477 : SUCCESS in 2h 03m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54d494a2bc79433daa10476c71ae44a0 : SUCCESS in 2h 07m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1de8dfccb2e545d2bc77ee67099d1e12 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":28},{"id":"0006ba074166dd0c8eef0ff4d8e8ff862fc885d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-19 16:41:32.000000000","message":"Patch Set 29: Patch Set 28 was rebased","accounts_in_message":[],"_revision_number":29},{"id":"972a156394f289ea13fc489c59be983c7140c3e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 18:04:47.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46a2565441784d3f8e2c41b287065b2a : SUCCESS in 16m 58s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f97da17c826b4ffeacbe3f318ae2dea8 : FAILURE in 36m 14s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d9c50c1c44d544b1aba12eae69995b5f : FAILURE in 30m 29s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/2b0efa2515964e099e8b597e15b373e5 : FAILURE in 39m 13s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4294334473e64dd69d63a28d6549c0fe : FAILURE in 36m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba7ead6dc44349a6aeac34869510ecb4 : FAILURE in 1h 16m 44s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/099ea70d682a4a2c9001339254d3da44 : FAILURE in 1h 06m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da58db0e589348b297c2911f2b472a42 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":29},{"id":"814b269ca0f8d52750dcf6b2ae10b1451a838ddd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-20 19:05:48.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"9cae5723ae40c89a5fe361c86bcb4e50c92ad0cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 20:57:18.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13713579eb844de9ab862be19bf78042 : SUCCESS in 15m 20s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e87c52d0205f4450b1c903f93e344111 : SUCCESS in 1h 21m 58s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4cb612d4805c49ec8977018fd61ebe60 : SUCCESS in 1h 03m 40s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/4319bcdf3f58470e886e84dd12ae2d6b : SUCCESS in 1h 22m 29s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/643c2517f37e494198dd02e8a94ccadf : SUCCESS in 1h 09m 26s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/96b0031aaf144134a5e8191e81cbb51e : SUCCESS in 1h 46m 00s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c952897f24e4fe0991bb3ef25948265 : SUCCESS in 1h 43m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17ac67ebb8234a57a73e8828d5811f96 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":29},{"id":"5b3b014a88408e8cf2e45f0fe0e36155fcb9757b","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-05-25 12:27:30.000000000","message":"Patch Set 29: Code-Review+2","accounts_in_message":[],"_revision_number":29},{"id":"7acd57ab6cbe673d190cc56d06ad5a50be802dd5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-25 18:36:08.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"35b4fd7c488852887ee4012ba920452ea4487d03","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-25 20:00:05.000000000","message":"Patch Set 29: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":29},{"id":"3e180a3b26de16aedce6a7d1404591d8a309aa06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 20:02:28.000000000","message":"Patch Set 29: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":29},{"id":"cbdfffc7e4f888d5d2e3177200171a4282be552b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 20:47:34.000000000","message":"Patch Set 29: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":29},{"id":"7bc3f4a6385f6a6b25d0c26dc87a2a743717cab8","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-05-27 13:59:18.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"9a5270f54743cd96670ff91db92c73989690f8ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 15:42:51.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42f94bb33ec849838eff31b458a22ccb : SUCCESS in 11m 38s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/704fdadb0bcc436fa5404cc77032fe3b : SUCCESS in 1h 24m 33s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/aab1d569a0474533941dba81033fb9e4 : SUCCESS in 1h 16m 05s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/27ddc7913a204f2ea6cc164899c9296e : SUCCESS in 1h 28m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1cc889d6e2444c2923105ca10f37c4c : SUCCESS in 1h 11m 23s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c7f3c66a13542f484897ef97efb2425 : FAILURE in 1h 20m 08s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffeb8970f8874300a5401bed156d9e02 : SUCCESS in 1h 28m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e3ff0df0e4448cb96ab8fcf8c7578e4 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":29},{"id":"205e45586a655fd49bbd17702eab88a90b44d97a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 15:46:00.000000000","message":"Patch Set 29: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":29},{"id":"06afc14cc304beac66d21df52b3f7b45ccda1657","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 17:32:44.000000000","message":"Patch Set 29: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aefc83ac0c424bd9b5e63ec0595cc58b : SUCCESS in 14m 42s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/674842af68ac4f18afea6c0cd2aff7f9 : SUCCESS in 1h 29m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/fce2d6546ae64d6b99a570fa557a803b : SUCCESS in 1h 36m 04s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/94dfa9cf69cd416f9247c1809886f229 : SUCCESS in 1h 32m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98c7f9cf1217420396ed3fd51317acd3 : SUCCESS in 1h 18m 33s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b2a5fb7d54a4bb586d21bc99adabd2a : SUCCESS in 1h 27m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2178b5f367bb49be92d88af61e7e50ba : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":29},{"id":"97bd70256ec5c0068015b64efe47719df678b714","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 17:32:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":29},{"id":"5b3c429b18a4836d29c6fc185468306ad79b7af4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 17:35:10.000000000","message":"Patch Set 29:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b7646cf83c419f87523924ffe1a612 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":29}],"current_revision_number":29,"current_revision":"45d07fcf0c6605c9da72147d243cbe8d7e1ce8a7","revisions":{"2b809100355a4b3431f0f459a3486bcd93a6ba0a":{"kind":"REWORK","_number":1,"created":"2022-04-14 12:09:25.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/1"}}},"commit":{"parents":[{"commit":"2202a33e1775de3a40673113b7a7023b167c3211","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2202a33e1775de3a40673113b7a7023b167c3211"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:53:22.000000000","tz":60},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2b809100355a4b3431f0f459a3486bcd93a6ba0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2b809100355a4b3431f0f459a3486bcd93a6ba0a"}]},"branch":"refs/heads/master"},"e8ed82fe29069fed20cfda8d91a333de106d05e7":{"kind":"REWORK","_number":2,"created":"2022-04-14 13:29:37.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/2"}}},"commit":{"parents":[{"commit":"2202a33e1775de3a40673113b7a7023b167c3211","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2202a33e1775de3a40673113b7a7023b167c3211"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 13:29:37.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/e8ed82fe29069fed20cfda8d91a333de106d05e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/e8ed82fe29069fed20cfda8d91a333de106d05e7"}]},"branch":"refs/heads/master"},"a4d359f687ea8c97f3a0a7223858f57b463c7367":{"kind":"REWORK","_number":3,"created":"2022-04-14 13:53:10.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/3"}}},"commit":{"parents":[{"commit":"2202a33e1775de3a40673113b7a7023b167c3211","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2202a33e1775de3a40673113b7a7023b167c3211"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 13:52:37.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/a4d359f687ea8c97f3a0a7223858f57b463c7367"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/a4d359f687ea8c97f3a0a7223858f57b463c7367"}]},"branch":"refs/heads/master"},"080990cad63fe9adbe9213294b11b63f07795228":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-04-14 14:20:36.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/4"}}},"commit":{"parents":[{"commit":"5aefc76d479d88f7b46da4d6630669ccf2fd4d95","subject":"Merge \"Use ssh_keypairs role to generate keys for repo sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5aefc76d479d88f7b46da4d6630669ccf2fd4d95"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 14:20:36.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/080990cad63fe9adbe9213294b11b63f07795228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/080990cad63fe9adbe9213294b11b63f07795228"}]},"branch":"refs/heads/master","description":"Rebase"},"428d860b4abb1984dd8079f6b3315d92d6dc1ea8":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-04-14 14:21:10.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/5"}}},"commit":{"parents":[{"commit":"2202a33e1775de3a40673113b7a7023b167c3211","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2202a33e1775de3a40673113b7a7023b167c3211"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 14:21:10.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/428d860b4abb1984dd8079f6b3315d92d6dc1ea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/428d860b4abb1984dd8079f6b3315d92d6dc1ea8"}]},"branch":"refs/heads/master","description":"Rebase"},"3aa5e9b7b32ad40191617ff2cd46b93beba8f303":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-04-19 11:11:04.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/6"}}},"commit":{"parents":[{"commit":"65af00ca45684b02d434a00273c60a9c87dc89a5","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/65af00ca45684b02d434a00273c60a9c87dc89a5"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-19 11:11:04.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3aa5e9b7b32ad40191617ff2cd46b93beba8f303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3aa5e9b7b32ad40191617ff2cd46b93beba8f303"}]},"branch":"refs/heads/master","description":"Rebase"},"f3ad3ce16457bcc7d94b339ad8cb205fc0349dad":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-04-20 06:55:43.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/7"}}},"commit":{"parents":[{"commit":"5aefc76d479d88f7b46da4d6630669ccf2fd4d95","subject":"Merge \"Use ssh_keypairs role to generate keys for repo sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5aefc76d479d88f7b46da4d6630669ccf2fd4d95"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-20 06:55:43.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/f3ad3ce16457bcc7d94b339ad8cb205fc0349dad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/f3ad3ce16457bcc7d94b339ad8cb205fc0349dad"}]},"branch":"refs/heads/master","description":"Rebase"},"bdb65a71208dd25ce2d3f14f33f3f9023b04e10a":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-04-20 07:44:17.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/8"}}},"commit":{"parents":[{"commit":"c966363bd44f0e9032b35406d75fa61d46aad3cb","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-20 07:44:17.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/bdb65a71208dd25ce2d3f14f33f3f9023b04e10a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/bdb65a71208dd25ce2d3f14f33f3f9023b04e10a"}]},"branch":"refs/heads/master","description":"Rebase"},"9925ed960ed895e8048a7f3113b3f02d5977c079":{"kind":"REWORK","_number":9,"created":"2022-04-20 08:46:29.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/9"}}},"commit":{"parents":[{"commit":"c966363bd44f0e9032b35406d75fa61d46aad3cb","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-20 08:46:29.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/9925ed960ed895e8048a7f3113b3f02d5977c079"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/9925ed960ed895e8048a7f3113b3f02d5977c079"}]},"branch":"refs/heads/master"},"ec61bee7afe992decc52827745f1d8c46eda6ee2":{"kind":"REWORK","_number":10,"created":"2022-04-26 10:35:06.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/10"}}},"commit":{"parents":[{"commit":"c966363bd44f0e9032b35406d75fa61d46aad3cb","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-26 10:35:06.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/ec61bee7afe992decc52827745f1d8c46eda6ee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/ec61bee7afe992decc52827745f1d8c46eda6ee2"}]},"branch":"refs/heads/master"},"e953c13aed853ad7618ce251edaf1924261ab28e":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-04-26 16:33:20.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/11"}}},"commit":{"parents":[{"commit":"0abdd74d7601f02da6a798f709ff3cb33b403a82","subject":"Add upgrade path from lsyncd to shared filesystem.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0abdd74d7601f02da6a798f709ff3cb33b403a82"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-26 16:33:07.000000000","tz":60},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/e953c13aed853ad7618ce251edaf1924261ab28e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/e953c13aed853ad7618ce251edaf1924261ab28e"}]},"branch":"refs/heads/master"},"f4e246497738b1b569ea5f016e1d30f7b37a4b79":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2022-04-27 07:47:17.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/12"}}},"commit":{"parents":[{"commit":"98d6df63c17918d11933c6c5cf30f682a7420fb3","subject":"Add upgrade path from lsyncd to shared filesystem.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/98d6df63c17918d11933c6c5cf30f682a7420fb3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-27 07:47:17.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/f4e246497738b1b569ea5f016e1d30f7b37a4b79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/f4e246497738b1b569ea5f016e1d30f7b37a4b79"}]},"branch":"refs/heads/master","description":"Rebase"},"299b993c291e7aef4e17b4b374fee27cbfbccb4e":{"kind":"REWORK","_number":13,"created":"2022-04-27 07:48:16.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/13"}}},"commit":{"parents":[{"commit":"98d6df63c17918d11933c6c5cf30f682a7420fb3","subject":"Add upgrade path from lsyncd to shared filesystem.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/98d6df63c17918d11933c6c5cf30f682a7420fb3"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-27 07:48:16.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/299b993c291e7aef4e17b4b374fee27cbfbccb4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/299b993c291e7aef4e17b4b374fee27cbfbccb4e"}]},"branch":"refs/heads/master"},"9a75bd975887bfbe5d5a4cea9ae66b2a31bd399f":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2022-04-27 11:38:25.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/14"}}},"commit":{"parents":[{"commit":"b47ceef7141aad26a791daae4f4afc776367df95","subject":"Add upgrade path from lsyncd to shared filesystem.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b47ceef7141aad26a791daae4f4afc776367df95"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-27 11:38:25.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/9a75bd975887bfbe5d5a4cea9ae66b2a31bd399f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/9a75bd975887bfbe5d5a4cea9ae66b2a31bd399f"}]},"branch":"refs/heads/master","description":"Rebase"},"9d53855552b89eeadaf14573a2eb404b4ad8f2f3":{"kind":"REWORK","_number":15,"created":"2022-04-27 13:20:55.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/15"}}},"commit":{"parents":[{"commit":"b47ceef7141aad26a791daae4f4afc776367df95","subject":"Add upgrade path from lsyncd to shared filesystem.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b47ceef7141aad26a791daae4f4afc776367df95"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-27 13:20:55.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/9d53855552b89eeadaf14573a2eb404b4ad8f2f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/9d53855552b89eeadaf14573a2eb404b4ad8f2f3"}]},"branch":"refs/heads/master"},"6f880f545719a0be6b55899d2ec4183e9531afa7":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2022-04-27 13:32:46.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/16"}}},"commit":{"parents":[{"commit":"1f5b5f1a1fe755a6d126c5128dc38e89dc8139cc","subject":"Add upgrade path from lsyncd to shared filesystem.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/1f5b5f1a1fe755a6d126c5128dc38e89dc8139cc"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-27 13:32:46.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/6f880f545719a0be6b55899d2ec4183e9531afa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/6f880f545719a0be6b55899d2ec4183e9531afa7"}]},"branch":"refs/heads/master","description":"Rebase"},"613aee98bca7961337e80f3097f6a0db1e5329f8":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2022-04-27 21:12:46.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/17"}}},"commit":{"parents":[{"commit":"7ceddbb7b2bbeef679e3eb87ef092f3b5fd20c17","subject":"Add upgrade path from lsyncd to shared filesystem.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/7ceddbb7b2bbeef679e3eb87ef092f3b5fd20c17"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-27 21:12:46.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/613aee98bca7961337e80f3097f6a0db1e5329f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/613aee98bca7961337e80f3097f6a0db1e5329f8"}]},"branch":"refs/heads/master","description":"Rebase"},"be8b95735147f69d8cbf770c76ca5549b39cc9b2":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2022-04-29 18:55:42.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/18"}}},"commit":{"parents":[{"commit":"deda7e0a6663ba37915ee154a843797dac6394a2","subject":"Add upgrade path from lsyncd to shared filesystem.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/deda7e0a6663ba37915ee154a843797dac6394a2"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-29 18:55:42.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/be8b95735147f69d8cbf770c76ca5549b39cc9b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/be8b95735147f69d8cbf770c76ca5549b39cc9b2"}]},"branch":"refs/heads/master","description":"Rebase"},"46e612b1f08d7f7051a830f00b3905430dab1344":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2022-05-04 09:51:48.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/19"}}},"commit":{"parents":[{"commit":"d65f9e492ac087cbf21306971451a75c89300545","subject":"Remove all code for lsync, rsync and ssh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d65f9e492ac087cbf21306971451a75c89300545"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-04 09:51:48.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/46e612b1f08d7f7051a830f00b3905430dab1344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/46e612b1f08d7f7051a830f00b3905430dab1344"}]},"branch":"refs/heads/master","description":"Rebase"},"6819e05933cdf4adf6744b27ac356e82c2dbf26a":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2022-05-05 11:49:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/59/837859/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/20"}}},"commit":{"parents":[{"commit":"5aefc76d479d88f7b46da4d6630669ccf2fd4d95","subject":"Merge \"Use ssh_keypairs role to generate keys for repo sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5aefc76d479d88f7b46da4d6630669ccf2fd4d95"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-05-05 11:49:12.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/6819e05933cdf4adf6744b27ac356e82c2dbf26a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/6819e05933cdf4adf6744b27ac356e82c2dbf26a"}]},"branch":"refs/heads/master","description":"Rebase"},"572957e56cb1bb11a49f22b8b42969dff387e30c":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2022-05-05 11:49:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/59/837859/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/21"}}},"commit":{"parents":[{"commit":"f7cfdc2774fb3821bbd24371a2d6668196aed8c6","subject":"Remove all code for lsync, rsync and ssh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/f7cfdc2774fb3821bbd24371a2d6668196aed8c6"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-05-05 11:49:31.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/572957e56cb1bb11a49f22b8b42969dff387e30c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/572957e56cb1bb11a49f22b8b42969dff387e30c"}]},"branch":"refs/heads/master","description":"Rebase"},"d04d99eab274242d5b4d7586365ac99572cc2967":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2022-05-09 06:46:44.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/22"}}},"commit":{"parents":[{"commit":"d25455d9ba8bea1de2d299ff6532568b57d0e98a","subject":"Remove all code for lsync, rsync and ssh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d25455d9ba8bea1de2d299ff6532568b57d0e98a"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-09 06:46:44.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d04d99eab274242d5b4d7586365ac99572cc2967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d04d99eab274242d5b4d7586365ac99572cc2967"}]},"branch":"refs/heads/master","description":"Rebase"},"5f421412645ae579f1f97b9232496c36629946fd":{"kind":"NO_CHANGE","_number":23,"created":"2022-05-12 16:39:51.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/23"}}},"commit":{"parents":[{"commit":"001b51003b55256bd00d65ad3d2167bdea76d650","subject":"Remove all code for lsync, rsync and ssh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/001b51003b55256bd00d65ad3d2167bdea76d650"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-12 16:37:38.000000000","tz":60},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5f421412645ae579f1f97b9232496c36629946fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5f421412645ae579f1f97b9232496c36629946fd"}]},"branch":"refs/heads/master"},"2eeae9bf3ccf45c985754207487fe185f165d98d":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2022-05-16 11:03:19.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/24"}}},"commit":{"parents":[{"commit":"4a698b11aa696f336dcf57d079cf696b582c3560","subject":"Remove all code for lsync, rsync and ssh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/4a698b11aa696f336dcf57d079cf696b582c3560"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-16 11:03:19.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2eeae9bf3ccf45c985754207487fe185f165d98d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2eeae9bf3ccf45c985754207487fe185f165d98d"}]},"branch":"refs/heads/master","description":"Rebase"},"e7a235fb657f6f8018ef85b89511f26271167b08":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2022-05-16 15:30:15.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/25"}}},"commit":{"parents":[{"commit":"74fa2710599cbe10bdd97294ac266f7cec349a1a","subject":"Remove all code for lsync, rsync and ssh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/74fa2710599cbe10bdd97294ac266f7cec349a1a"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-16 15:30:15.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/e7a235fb657f6f8018ef85b89511f26271167b08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/e7a235fb657f6f8018ef85b89511f26271167b08"}]},"branch":"refs/heads/master","description":"Rebase"},"de2562c2721a4753bf8d9a4315becc7141aac08f":{"kind":"NO_CHANGE","_number":26,"created":"2022-05-17 11:06:14.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/26"}}},"commit":{"parents":[{"commit":"7854797c90d93bc722d70c95e7fef500242b6915","subject":"Remove all code for lsync, rsync and ssh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/7854797c90d93bc722d70c95e7fef500242b6915"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-17 11:06:14.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/de2562c2721a4753bf8d9a4315becc7141aac08f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/de2562c2721a4753bf8d9a4315becc7141aac08f"}]},"branch":"refs/heads/master","description":"Rebase"},"3fafa20d3ddf05273a939ade7fff2bc22399e9f9":{"kind":"REWORK","_number":27,"created":"2022-05-18 11:41:14.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/27"}}},"commit":{"parents":[{"commit":"7854797c90d93bc722d70c95e7fef500242b6915","subject":"Remove all code for lsync, rsync and ssh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/7854797c90d93bc722d70c95e7fef500242b6915"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-18 11:41:10.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3fafa20d3ddf05273a939ade7fff2bc22399e9f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3fafa20d3ddf05273a939ade7fff2bc22399e9f9"}]},"branch":"refs/heads/master"},"d6ab0c2af73f072b5b67eeb97176a3f033803c22":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2022-05-18 11:41:18.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/59/837859/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/28"}}},"commit":{"parents":[{"commit":"3d50e2636df33a541f16eda16c106cfeaac17428","subject":"Remove all code for lsync, rsync and ssh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3d50e2636df33a541f16eda16c106cfeaac17428"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-18 11:41:18.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d6ab0c2af73f072b5b67eeb97176a3f033803c22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d6ab0c2af73f072b5b67eeb97176a3f033803c22"}]},"branch":"refs/heads/master","description":"Rebase"},"45d07fcf0c6605c9da72147d243cbe8d7e1ce8a7":{"kind":"NO_CHANGE","_number":29,"created":"2022-05-19 16:41:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/59/837859/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/59/837859/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/59/837859/29"}}},"commit":{"parents":[{"commit":"03b55edaaeaf9e9e44371578bb7a179a063f26d9","subject":"Remove all code for lsync, rsync and ssh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/03b55edaaeaf9e9e44371578bb7a179a063f26d9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-14 11:51:31.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-05-19 16:41:32.000000000","tz":0},"subject":"Clean up legacy lsycnd, rsync and ssh key config","message":"Clean up legacy lsycnd, rsync and ssh key config\n\nThese are no longer needed when a shared filesystem is used to\nsyncronise the repo server contents\n\nChange-Id: I3109bd891d4c6b3522f5f741d9970093b1e882c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/45d07fcf0c6605c9da72147d243cbe8d7e1ce8a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/45d07fcf0c6605c9da72147d243cbe8d7e1ce8a7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
