)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~839411","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I976a5ea5f6b6ebd65c22e89657763fef87cf4b23","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"osa-gluster","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2022-05-18 11:16:55.000000000","reason":"Change was submitted"},"13095":{"account":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"last_update":"2022-05-18 11:16:55.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-05-18 08:58:08.000000000","reason":"Dmitriy Rabotyagov replied on the change"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-05-18 11:16:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I976a5ea5f6b6ebd65c22e89657763fef87cf4b23","subject":"Add upgrade path from lsyncd to shared filesystem.","status":"MERGED","created":"2022-04-26 16:33:20.000000000","updated":"2022-05-18 11:18:57.000000000","submitted":"2022-05-18 11:16:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839411-osa-gluster","meta_rev_id":"8614c7df217dd0fcf72dff8c1b149bdf21655d18","_number":839411,"virtual_id_number":839411,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-18 11:16:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2022-05-18 09:18:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":2,"date":"2022-05-18 08:58:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2022-05-18 09:18:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"CC":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-26 18:56:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-27 19:50:29.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"CC"},{"updated":"2022-05-04 09:38:26.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"CC"},{"updated":"2022-05-05 11:37:49.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-05-18 09:18:58.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"3d74636827c8a91c57a10c9ab0d86170353bc54d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-26 16:33:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"038986aa3ad95b361c98a948495cf4cbea510ee9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 18:56:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94ccf809ccb042da9e029f50e64a2e06 : FAILURE in 13m 40s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/dca5b53865624ea3a341a2620770b0cc : FAILURE in 36m 46s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/6e1018458afe4ae98226b2c79a5aa235 : FAILURE in 30m 17s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/869172a7909443f38cdc5c6f544610d6 : FAILURE in 34m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1276d777000e4a5d81342bc621b39891 : FAILURE in 27m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/0512fc678c3a4fb3a7489f5802af8d7d : SUCCESS in 2h 12m 43s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2fba6adfc004a69a5f80c96c8bf22ef : SUCCESS in 2h 03m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32cdf4fd835846e3a0927e1eb173cc58 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9d4c92e7995896f1d67120f63b7bb527265be211","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-27 07:46:31.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"37133ad8fef0ad0f467710f5eccd609af9741747","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 09:50:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac86fd843af94335b8484aee15d15df3 : FAILURE in 15m 42s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fc854807a0b435e9613b6dfd5967595 : SUCCESS in 1h 21m 51s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/51d2dbfb14f54084956b248261b0a48a : SUCCESS in 1h 02m 53s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/5e7aa7d28a0240a594ddd296eee8e039 : SUCCESS in 1h 26m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86eb679e15b9456fa360562b26df1593 : SUCCESS in 1h 14m 49s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c7ccdc1c73b0496fb1182aa325e0fe58 : SUCCESS in 1h 56m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fba4d62619145568635d3d4dde382a4 : SUCCESS in 1h 59m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb6272ecbabc44718aef62059a65c897 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":2},{"id":"39c21412ae06f81ba80d6e7274272d59b431537c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-27 11:37:02.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c7409e8c91b5d30e308b3563442b15980f12cb72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 13:28:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ddf9eee81544fad8abf974b8691df4a : FAILURE in 13m 53s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/7871c8d07c4f4e1bade049d54b8108b5 : SUCCESS in 1h 32m 15s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5337fc0e86c2495cb7db77c675482c26 : SUCCESS in 56m 32s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e6c4327276c6463d95cef157e49201ee : SUCCESS in 1h 28m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ff3e4b0e6a64a7094914c4dc4ae0a73 : SUCCESS in 1h 11m 22s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/71ca72bcefd0467aa173dd76a2473e8b : SUCCESS in 1h 39m 16s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a4e70c76ea14362b109a3c5e5aa34cc : SUCCESS in 1h 34m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c26132f84f6c4e8e90b4b90073dc84b2 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":3},{"id":"15724f77393a1fe9b1ad51da399878dcd42faef2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-27 13:32:09.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"40c684f79311057633530bdce56dcc8801b90683","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 15:47:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65d859cdc2cb496db562687578e9bee8 : FAILURE in 14m 13s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/661b9188c128410bba5c1a620a473755 : SUCCESS in 1h 23m 52s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a1ffd31007844aea9b8587a256deeab5 : SUCCESS in 1h 05m 51s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b6db4974d3ff4712afdd3954d9f6d628 : SUCCESS in 1h 15m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d319f0703bfa437e983cf488d1b92f04 : SUCCESS in 1h 11m 55s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd63db4636d949d69da0790efe16a190 : SUCCESS in 1h 52m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb118642f92644c2b84eb64fc1eb396b : SUCCESS in 2h 04m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5278af1029a04843bed23ba678f1b182 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":4},{"id":"320e75f48bf7a71d525025e64321abf3c56758c0","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-04-27 19:50:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"446e5ef3b53a1ce0cc13b0873d85a3dd84cc05b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-27 21:12:40.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"df93440f8ecfc3b6d87a82bda5680b5bbc57864e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 23:15:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1db667735ac847e2a33f99deb0892e95 : SUCCESS in 14m 09s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d6c1784d0fde49ab80e7dc6c808d5854 : SUCCESS in 1h 21m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/fa88a8e9e03847f58da8bed921752802 : SUCCESS in 1h 09m 20s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b8395fd3fe4246d6b52519561d1b6f76 : SUCCESS in 1h 29m 04s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf1bc8047db049fd867bdeeb167bb556 : SUCCESS in 1h 17m 52s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/079231464a0e47ad8be09d8cf4db873a : SUCCESS in 1h 59m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4778f1f7d754b43a3715af836810f4e : SUCCESS in 1h 52m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edccc1933bd84966831f68e0dc3c5751 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":5},{"id":"27d7dbe2c3cc9cd36578618621bb934a1890a106","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-04-28 15:15:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5f429dcd85ca822a1b157540147461402e17c497","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-04-28 15:16:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a5ed41ee36192f4100b0b79339730aeea08a1d2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-29 18:52:37.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"11f13e7aa5928bac39e2cbb8b382bb8e086a01b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-04-29 18:55:02.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"ced1cea055f6433ccf779e078bd80d0679b36d12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 21:04:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0b8aae438bb42a39f58588d2ff4beb4 : SUCCESS in 13m 38s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/353f22a5fe344060b7179546ca4be53b : SUCCESS in 1h 21m 27s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/0179fcf2066441fbac17f508b125ebe7 : SUCCESS in 1h 11m 18s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/9176630d406d468e91027f75951ae60b : SUCCESS in 1h 21m 47s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4d070688202492287d08c89e88fc264 : SUCCESS in 1h 12m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7501c7d4d1b145f4b079b39d59f86b7f : SUCCESS in 1h 58m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b20f96a2c6314eb1a184a2c9ed338660 : SUCCESS in 2h 00m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44cb52f6b064f16a7fad8ab874ddeb2 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":7},{"id":"be22ee5e45c2981e2d9b1683bb72b438158a6eb0","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-05-04 09:38:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"622b98e1a7d4cd1dc099f8801c9690caca5329b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-04 09:50:10.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"4f8d0f4ae46bda6c4b99105298c2927eb3280446","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-04 09:50:48.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"745470fd575d587462fa6bf2488cbf06151b4b0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 11:57:25.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d29ffc5cb01e4dc8ad907b9931b249df : SUCCESS in 14m 44s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2eaeda19452143149c5f86a6414d85be : SUCCESS in 1h 23m 33s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/3f22eea3da6e4a49bf8d79d0d7fd47a7 : SUCCESS in 1h 16m 17s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/45850dbfe8fa473d8f8a2ab0b4a52be1 : SUCCESS in 1h 27m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/573b5a6d6ac94d1c967e24ade6883015 : SUCCESS in 1h 20m 42s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/71c6512b325a4c518d414c342d2a3c83 : SUCCESS in 2h 02m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/263f852ee97142348c5d625abad8a8a8 : SUCCESS in 1h 57m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16d8508ceb8746cebaf314fb4c812dbb : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":8},{"id":"047c1ee63dd90e24ead6efadebe267fb3e5a8ad1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-05 11:37:49.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2acf04a042f495d3dd1b62b8217a1bf26d47ea10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-07 17:27:32.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"85aea3830fd180566eb935203c886dc067bc3a04","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-07 17:30:08.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"946e4bb76aac4bdb35ae84d4ff653ca6a1612798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 19:00:53.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7899aa129c864dc2933bd58fab3d3891 : SUCCESS in 11m 52s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/974fcf2c253e4944a74f7a1bbeb1c4db : SUCCESS in 1h 02m 57s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/0e4db127a7a3463fb1704c7b938d171b : SUCCESS in 50m 09s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/3cd3f9ed70aa4ea6996f8cba5f3e207b : SUCCESS in 1h 28m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd8597e0e5a34f14aa98d36b57ccf4c0 : SUCCESS in 53m 47s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5801d44af474228bb10634c99f7a5e5 : FAILURE in 55m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb5018b8dad043d8b54c2e8d743e2d0f : FAILURE in 44m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/633a8cf43eaf4ebeb8f2d12b57ea917c : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":9},{"id":"791201b50ffd930cedbece2a3aad97bc84462e8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-09 06:46:27.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"f135180e779191f399a013922ce1f74a9123fdc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 08:44:12.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ca8e8f7d0c8407b9d27b91e12d460ea : SUCCESS in 12m 12s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/ce01df4c11e449f58f2d08088978457b : SUCCESS in 1h 14m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/fda61ad9aae743978a8541d14013d198 : FAILURE in 29m 34s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/37b24e558b634417a4eed25306e020dc : SUCCESS in 1h 28m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b6d66a5ab4647668f33bf3784916017 : FAILURE in 27m 46s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/05310334f95043a79b2a337339185f50 : SUCCESS in 1h 54m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56ff43e44cb74022a8e1d0c196f6f0d0 : FAILURE in 1h 04m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8dc993a63a24c8ead3c24194e152f3d : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":10},{"id":"6a9bbf95b409bfc91dcc6482a94738772f25bd48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-12 16:39:51.000000000","message":"Uploaded patch set 11: New patch set was added with same tree, parent, and commit message as Patch Set 10.","accounts_in_message":[],"_revision_number":11},{"id":"c11611fb9007139a696938f508ea351d9af63416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 18:44:50.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f3d77ef28b64f36ad4004089b5935eb : SUCCESS in 17m 04s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ea9650b21f941b1b36f97a92963d2b0 : FAILURE in 1h 04m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/52c5f4c9ca4841aaa489da4094cde9e7 : FAILURE in 35m 47s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/6102b0c662f94f07a73b3999d80ca4d3 : FAILURE in 1h 07m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a56ae55ec1034039bf6a959d0e223998 : FAILURE in 28m 30s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7348d7d98e35418b9d9bc6ac16aae979 : SUCCESS in 2h 00m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0af68011bec4067af4c536485145977 : FAILURE in 1h 05m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3ae69ea22094316bedc6a4830439119 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":11},{"id":"94757bc787c87bcf07a9a8744a998c97f491724b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-16 09:31:10.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"0e9d3bd61f60a9dbe232a6240a80458a0ddc52c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 11:41:08.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7684c34b42ff44ab98a85fcc09db865d : SUCCESS in 15m 30s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f310b0f06e94952b7f20e9ab56b9370 : SUCCESS in 1h 14m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/c26014c717b640e1bcc32cd32d8d2660 : FAILURE in 28m 46s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/8e6c68c0834d456c983b5234dbb290bf : SUCCESS in 1h 28m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5758562d20824498b7e541056bcddca5 : FAILURE in 30m 21s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e4b9b64579f43b79b57b8047ba48ea9 : SUCCESS in 2h 03m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2a79ced870845f4b860c5f0200e4d92 : FAILURE in 1h 02m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ef56a7e8d7742aea7dd1ea5c39818cd : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":12},{"id":"1eaecbda0cda8c4780a3f607f80a28d3b3ee1faa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-16 15:29:59.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"28419ddf2d746be182345940c5985b1f374fd722","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 17:24:54.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/980fd383bb564a808bc7af9535a51818 : SUCCESS in 13m 16s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f42a4308ca54f609da2bec1958a2e33 : SUCCESS in 1h 26m 37s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7ccfc9a4243142a088a75b41b663544f : FAILURE in 46m 47s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b244670f220d40fbb646c2be06480567 : SUCCESS in 1h 27m 13s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de18aa49afd14f5b94a5e3fb9622ee20 : SUCCESS in 56m 33s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef67397a60184bfca1805f65a5903518 : SUCCESS in 1h 42m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5712ad5c3f7e4f17bf17ab440daeeb77 : SUCCESS in 1h 38m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d890df18d6674f59bf6d129c311915ad : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":13},{"id":"c9c9bbff7324403435e4f79a0263195278c37d5d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-16 20:12:51.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"49715f403921c59f2a471600d104afa17fad9ebc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 22:41:40.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4faa0ba917254b8daa5e2a1919a45688 : SUCCESS in 14m 01s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/8e4d203ab0ff41298b790ec59d5584e9 : SUCCESS in 1h 44m 52s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/35d525ccc8c14bdcacffb02d4a481e41 : SUCCESS in 1h 16m 35s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/0ca73fe142e64ce8bdda60f877394f0a : SUCCESS in 1h 36m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8321154796564e03bd104213ddab2f1f : SUCCESS in 1h 10m 41s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/890cf7c68d5e4b97b23732c26e92fe45 : SUCCESS in 2h 19m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f97d6e6b5e724182bbc6ce22e367f57e : SUCCESS in 1h 43m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/631fdca036394de295b0a98ae784087a : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":13},{"id":"8d05970749a4a5eeec55fabcc88f4e3565c200b9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-18 08:58:08.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"0e15a5665961202e88d9b8ff8df573df992e3cbd","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2022-05-18 09:18:58.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"97bf92cb38a1eaba4d91dc310a00272b33960cc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 09:21:20.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"e149ff62720848c0cbc1d4045919f67bf9610f28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 11:16:55.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/939946072b474ff5898024c4b540e6b9 : SUCCESS in 15m 23s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/18ca561600674ac4b82b1760766a8ae9 : SUCCESS in 1h 28m 26s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/da7fafdb0a544f3683f030515d3a9317 : SUCCESS in 58m 51s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/90fe1a3221f047d9b3d44d4c7810b430 : SUCCESS in 1h 13m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1da4a347160d483694ef41274e3c98ea : SUCCESS in 1h 22m 52s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a512c2855934a2ebc6e93c5cc1035ba : SUCCESS in 1h 53m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e66c4bfd7e5e4f93bc1a3fda0ecf4a5e : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":13},{"id":"a03f917cb2ff65d55b1ea4d0cb3f791353b743f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 11:16:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"8614c7df217dd0fcf72dff8c1b149bdf21655d18","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 11:18:57.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7acb15e3f5004da980c1db72857638b5 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"1f39aa02395c186a57d9b5aae738ce3680b90aa5","revisions":{"0abdd74d7601f02da6a798f709ff3cb33b403a82":{"kind":"REWORK","_number":1,"created":"2022-04-26 16:33:20.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/11/839411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/11/839411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/1"}}},"commit":{"parents":[{"commit":"c966363bd44f0e9032b35406d75fa61d46aad3cb","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-26 16:29:50.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-26 16:33:07.000000000","tz":60},"subject":"Add upgrade path from lsyncd to shared filesystem.","message":"Add upgrade path from lsyncd to shared filesystem.\n\nThis patch checks if the repo server content directory is a mountpoint,\nand if it is not, creates an archive of the repo server contents\nbefore mounting the shared filesystem. After the mount completes, the\narchive of repo server contents is copied to the shared filesystem.\n\nThis runs on every repo server in the deployment to include corner\ncases of repo servers running different OS versions or CPU architecures\nand will gather all of the content onto a single shared mount.\n\nChange-Id: I976a5ea5f6b6ebd65c22e89657763fef87cf4b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0abdd74d7601f02da6a798f709ff3cb33b403a82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0abdd74d7601f02da6a798f709ff3cb33b403a82"}]},"branch":"refs/heads/master"},"98d6df63c17918d11933c6c5cf30f682a7420fb3":{"kind":"REWORK","_number":2,"created":"2022-04-27 07:46:31.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/11/839411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/11/839411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/2"}}},"commit":{"parents":[{"commit":"c966363bd44f0e9032b35406d75fa61d46aad3cb","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-26 16:29:50.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-27 07:46:31.000000000","tz":0},"subject":"Add upgrade path from lsyncd to shared filesystem.","message":"Add upgrade path from lsyncd to shared filesystem.\n\nThis patch checks if the repo server content directory is a mountpoint,\nand if it is not, creates an archive of the repo server contents\nbefore mounting the shared filesystem. After the mount completes, the\narchive of repo server contents is copied to the shared filesystem.\n\nThis runs on every repo server in the deployment to include corner\ncases of repo servers running different OS versions or CPU architecures\nand will gather all of the content onto a single shared mount.\n\nChange-Id: I976a5ea5f6b6ebd65c22e89657763fef87cf4b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/98d6df63c17918d11933c6c5cf30f682a7420fb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/98d6df63c17918d11933c6c5cf30f682a7420fb3"}]},"branch":"refs/heads/master"},"b47ceef7141aad26a791daae4f4afc776367df95":{"kind":"REWORK","_number":3,"created":"2022-04-27 11:37:02.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/11/839411/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/11/839411/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/3"}}},"commit":{"parents":[{"commit":"c966363bd44f0e9032b35406d75fa61d46aad3cb","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-26 16:29:50.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-27 11:37:02.000000000","tz":0},"subject":"Add upgrade path from lsyncd to shared filesystem.","message":"Add upgrade path from lsyncd to shared filesystem.\n\nThis patch checks if the repo server content directory is a mountpoint,\nand if it is not, creates an archive of the repo server contents\nbefore mounting the shared filesystem. After the mount completes, the\narchive of repo server contents is copied to the shared filesystem.\n\nThis runs on every repo server in the deployment to include corner\ncases of repo servers running different OS versions or CPU architecures\nand will gather all of the content onto a single shared mount.\n\nChange-Id: I976a5ea5f6b6ebd65c22e89657763fef87cf4b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b47ceef7141aad26a791daae4f4afc776367df95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b47ceef7141aad26a791daae4f4afc776367df95"}]},"branch":"refs/heads/master"},"1f5b5f1a1fe755a6d126c5128dc38e89dc8139cc":{"kind":"REWORK","_number":4,"created":"2022-04-27 13:32:09.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/11/839411/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/11/839411/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/4"}}},"commit":{"parents":[{"commit":"c966363bd44f0e9032b35406d75fa61d46aad3cb","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-26 16:29:50.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-27 13:32:01.000000000","tz":0},"subject":"Add upgrade path from lsyncd to shared filesystem.","message":"Add upgrade path from lsyncd to shared filesystem.\n\nThis patch checks if the repo server content directory is a mountpoint,\nand if it is not, creates an archive of the repo server contents\nbefore mounting the shared filesystem. After the mount completes, the\narchive of repo server contents is copied to the shared filesystem.\n\nThis runs on every repo server in the deployment to include corner\ncases of repo servers running different OS versions or CPU architecures\nand will gather all of the content onto a single shared mount.\n\nChange-Id: I976a5ea5f6b6ebd65c22e89657763fef87cf4b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/1f5b5f1a1fe755a6d126c5128dc38e89dc8139cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/1f5b5f1a1fe755a6d126c5128dc38e89dc8139cc"}]},"branch":"refs/heads/master"},"7ceddbb7b2bbeef679e3eb87ef092f3b5fd20c17":{"kind":"REWORK","_number":5,"created":"2022-04-27 21:12:40.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/11/839411/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/11/839411/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/5"}}},"commit":{"parents":[{"commit":"c966363bd44f0e9032b35406d75fa61d46aad3cb","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-26 16:29:50.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-27 21:12:22.000000000","tz":0},"subject":"Add upgrade path from lsyncd to shared filesystem.","message":"Add upgrade path from lsyncd to shared filesystem.\n\nThis patch checks if the repo server content directory is a mountpoint,\nand if it is not, creates an archive of the repo server contents\nbefore mounting the shared filesystem. After the mount completes, the\narchive of repo server contents is copied to the shared filesystem.\n\nThis runs on every repo server in the deployment to include corner\ncases of repo servers running different OS versions or CPU architecures\nand will gather all of the content onto a single shared mount.\n\nChange-Id: I976a5ea5f6b6ebd65c22e89657763fef87cf4b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/7ceddbb7b2bbeef679e3eb87ef092f3b5fd20c17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/7ceddbb7b2bbeef679e3eb87ef092f3b5fd20c17"}]},"branch":"refs/heads/master"},"f5b8fe5a2d81791e39374c4c3517fa7b6aca84b5":{"kind":"REWORK","_number":6,"created":"2022-04-29 18:52:37.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/11/839411/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/11/839411/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/6"}}},"commit":{"parents":[{"commit":"c966363bd44f0e9032b35406d75fa61d46aad3cb","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-26 16:29:50.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-29 18:52:36.000000000","tz":0},"subject":"Add upgrade path from lsyncd to shared filesystem.","message":"Add upgrade path from lsyncd to shared filesystem.\n\nThis patch checks if the repo server content directory is a mountpoint,\nand if it is not, creates an archive of the repo server contents\nbefore mounting the shared filesystem. After the mount completes, the\narchive of repo server contents is copied to the shared filesystem.\n\nThis runs on every repo server in the deployment to include corner\ncases of repo servers running different OS versions or CPU architecures\nand will gather all of the content onto a single shared mount.\n\nChange-Id: I976a5ea5f6b6ebd65c22e89657763fef87cf4b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/f5b8fe5a2d81791e39374c4c3517fa7b6aca84b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/f5b8fe5a2d81791e39374c4c3517fa7b6aca84b5"}]},"branch":"refs/heads/master"},"deda7e0a6663ba37915ee154a843797dac6394a2":{"kind":"REWORK","_number":7,"created":"2022-04-29 18:55:02.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/11/839411/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/11/839411/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/7"}}},"commit":{"parents":[{"commit":"c966363bd44f0e9032b35406d75fa61d46aad3cb","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-26 16:29:50.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-29 18:54:59.000000000","tz":0},"subject":"Add upgrade path from lsyncd to shared filesystem.","message":"Add upgrade path from lsyncd to shared filesystem.\n\nThis patch checks if the repo server content directory is a mountpoint,\nand if it is not, creates an archive of the repo server contents\nbefore mounting the shared filesystem. After the mount completes, the\narchive of repo server contents is copied to the shared filesystem.\n\nThis runs on every repo server in the deployment to include corner\ncases of repo servers running different OS versions or CPU architecures\nand will gather all of the content onto a single shared mount.\n\nChange-Id: I976a5ea5f6b6ebd65c22e89657763fef87cf4b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/deda7e0a6663ba37915ee154a843797dac6394a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/deda7e0a6663ba37915ee154a843797dac6394a2"}]},"branch":"refs/heads/master"},"0a138d639224b242d20d78c665248bb200cac996":{"kind":"REWORK","_number":8,"created":"2022-05-04 09:50:10.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/11/839411/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/11/839411/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/8"}}},"commit":{"parents":[{"commit":"c966363bd44f0e9032b35406d75fa61d46aad3cb","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-26 16:29:50.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-04 09:50:10.000000000","tz":0},"subject":"Add upgrade path from lsyncd to shared filesystem.","message":"Add upgrade path from lsyncd to shared filesystem.\n\nThis patch checks if the repo server content directory is a mountpoint,\nand if it is not, creates an archive of the repo server contents\nbefore mounting the shared filesystem. After the mount completes, the\narchive of repo server contents is copied to the shared filesystem.\n\nThis runs on every repo server in the deployment to include corner\ncases of repo servers running different OS versions or CPU architecures\nand will gather all of the content onto a single shared mount.\n\nChange-Id: I976a5ea5f6b6ebd65c22e89657763fef87cf4b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0a138d639224b242d20d78c665248bb200cac996"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0a138d639224b242d20d78c665248bb200cac996"}]},"branch":"refs/heads/master"},"db38c78130dfa2541e11b0a4b047ae8c5ff4a5f1":{"kind":"REWORK","_number":9,"created":"2022-05-07 17:27:32.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/11/839411/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/11/839411/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/9"}}},"commit":{"parents":[{"commit":"c966363bd44f0e9032b35406d75fa61d46aad3cb","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-26 16:29:50.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-07 17:27:31.000000000","tz":0},"subject":"Add upgrade path from lsyncd to shared filesystem.","message":"Add upgrade path from lsyncd to shared filesystem.\n\nThis patch checks if the repo server content directory is a mountpoint,\nand if it is not, creates an archive of the repo server contents\nbefore mounting the shared filesystem. After the mount completes, the\narchive of repo server contents is copied to the shared filesystem.\n\nThis runs on every repo server in the deployment to include corner\ncases of repo servers running different OS versions or CPU architecures\nand will gather all of the content onto a single shared mount.\n\nChange-Id: I976a5ea5f6b6ebd65c22e89657763fef87cf4b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/db38c78130dfa2541e11b0a4b047ae8c5ff4a5f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/db38c78130dfa2541e11b0a4b047ae8c5ff4a5f1"}]},"branch":"refs/heads/master"},"8bf9b7be6c61ca27b9d831d0b6618c3245713912":{"kind":"REWORK","_number":10,"created":"2022-05-09 06:46:27.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/11/839411/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/11/839411/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/10"}}},"commit":{"parents":[{"commit":"c966363bd44f0e9032b35406d75fa61d46aad3cb","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-26 16:29:50.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-09 06:46:27.000000000","tz":0},"subject":"Add upgrade path from lsyncd to shared filesystem.","message":"Add upgrade path from lsyncd to shared filesystem.\n\nThis patch checks if the repo server content directory is a mountpoint,\nand if it is not, creates an archive of the repo server contents\nbefore mounting the shared filesystem. After the mount completes, the\narchive of repo server contents is copied to the shared filesystem.\n\nThis runs on every repo server in the deployment to include corner\ncases of repo servers running different OS versions or CPU architecures\nand will gather all of the content onto a single shared mount.\n\nChange-Id: I976a5ea5f6b6ebd65c22e89657763fef87cf4b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/8bf9b7be6c61ca27b9d831d0b6618c3245713912"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/8bf9b7be6c61ca27b9d831d0b6618c3245713912"}]},"branch":"refs/heads/master"},"02068d0aec4a07ec62ab2d7f9f4c3e6918de6d33":{"kind":"NO_CHANGE","_number":11,"created":"2022-05-12 16:39:51.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/11/839411/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/11/839411/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/11"}}},"commit":{"parents":[{"commit":"c966363bd44f0e9032b35406d75fa61d46aad3cb","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-26 16:29:50.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-12 16:37:15.000000000","tz":60},"subject":"Add upgrade path from lsyncd to shared filesystem.","message":"Add upgrade path from lsyncd to shared filesystem.\n\nThis patch checks if the repo server content directory is a mountpoint,\nand if it is not, creates an archive of the repo server contents\nbefore mounting the shared filesystem. After the mount completes, the\narchive of repo server contents is copied to the shared filesystem.\n\nThis runs on every repo server in the deployment to include corner\ncases of repo servers running different OS versions or CPU architecures\nand will gather all of the content onto a single shared mount.\n\nChange-Id: I976a5ea5f6b6ebd65c22e89657763fef87cf4b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/02068d0aec4a07ec62ab2d7f9f4c3e6918de6d33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/02068d0aec4a07ec62ab2d7f9f4c3e6918de6d33"}]},"branch":"refs/heads/master"},"0b37048f1b24000b09fdc1c763396594002b6e52":{"kind":"REWORK","_number":12,"created":"2022-05-16 09:31:10.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/11/839411/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/11/839411/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/12"}}},"commit":{"parents":[{"commit":"c966363bd44f0e9032b35406d75fa61d46aad3cb","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-26 16:29:50.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-16 07:54:37.000000000","tz":60},"subject":"Add upgrade path from lsyncd to shared filesystem.","message":"Add upgrade path from lsyncd to shared filesystem.\n\nThis patch checks if the repo server content directory is a mountpoint,\nand if it is not, creates an archive of the repo server contents\nbefore mounting the shared filesystem. After the mount completes, the\narchive of repo server contents is copied to the shared filesystem.\n\nThis runs on every repo server in the deployment to include corner\ncases of repo servers running different OS versions or CPU architecures\nand will gather all of the content onto a single shared mount.\n\nChange-Id: I976a5ea5f6b6ebd65c22e89657763fef87cf4b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0b37048f1b24000b09fdc1c763396594002b6e52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0b37048f1b24000b09fdc1c763396594002b6e52"}]},"branch":"refs/heads/master"},"1f39aa02395c186a57d9b5aae738ce3680b90aa5":{"kind":"REWORK","_number":13,"created":"2022-05-16 15:29:59.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/11/839411/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/11/839411/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/11/839411/13"}}},"commit":{"parents":[{"commit":"c966363bd44f0e9032b35406d75fa61d46aad3cb","subject":"Add facility to store repo contents on a remote mount","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c966363bd44f0e9032b35406d75fa61d46aad3cb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-04-26 16:29:50.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-16 15:29:59.000000000","tz":0},"subject":"Add upgrade path from lsyncd to shared filesystem.","message":"Add upgrade path from lsyncd to shared filesystem.\n\nThis patch checks if the repo server content directory is a mountpoint,\nand if it is not, creates an archive of the repo server contents\nbefore mounting the shared filesystem. After the mount completes, the\narchive of repo server contents is copied to the shared filesystem.\n\nThis runs on every repo server in the deployment to include corner\ncases of repo servers running different OS versions or CPU architecures\nand will gather all of the content onto a single shared mount.\n\nChange-Id: I976a5ea5f6b6ebd65c22e89657763fef87cf4b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/1f39aa02395c186a57d9b5aae738ce3680b90aa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/1f39aa02395c186a57d9b5aae738ce3680b90aa5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
