)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~841618","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I6e61e5e226c22d84d0a427359979dbb4338e890d","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"osa/centos-9","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-05-29 17:33:27.000000000","reason":"Dmitriy Rabotyagov replied on the change"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-05-29 20:36:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6e61e5e226c22d84d0a427359979dbb4338e890d","subject":"Use the same vars file for all versions of centos","status":"MERGED","created":"2022-05-12 16:39:51.000000000","updated":"2022-05-29 20:38:10.000000000","submitted":"2022-05-29 20:36:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841618-osa/centos-9","meta_rev_id":"c4f9abd5642f34eea9ec0a870b5036c673f6c315","_number":841618,"virtual_id_number":841618,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-29 20:36:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-05-25 12:29:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2022-05-25 19:18:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2022-05-25 20:00:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-12 18:41:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-20 19:05:52.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2022-05-25 12:29:46.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2022-05-25 19:18:42.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"236929561cda577f62b53077dd222169f954298b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-12 16:39:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b35f786555a679a3250f9e1d03d35cb8cd7af8dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 18:41:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be31c0923fbe49c0b95adeb9970e5e3a : SUCCESS in 14m 30s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/18bb17d0273b4a45b194612c1ab5dec8 : FAILURE in 56m 14s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/dcee242cce34449d8b5b1c48380fb286 : FAILURE in 27m 24s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/69dd5b5dafd54be289f3bc8454d541a4 : FAILURE in 1h 13m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b5c201099744421b5f15e177ac6489a : FAILURE in 29m 32s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9886653969643f3868f8ac766ee605d : SUCCESS in 1h 56m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b08c1c3fd54a4326bb4cf147f192af99 : POST_FAILURE in 56m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b042d305e5c465fa6cc4ecb6105753b : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d144980cb4d16bfb03af2e0d592ebf5df7055d04","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-13 07:09:16.000000000","message":"Topic set to osa/centos-9","accounts_in_message":[],"_revision_number":1},{"id":"d7368faa51950b46a3aa85b020fb0423aa38b2b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-16 11:03:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"6389788338593b874a8353d92cbc4c4b213b0109","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 12:53:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ce80112b6a84ccfa2d6dd755444ee69 : SUCCESS in 14m 53s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/80627b069a934402af6ae587277e477c : FAILURE in 1h 14m 41s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1da4d5d7645542008e96e1946a5cb245 : FAILURE in 28m 07s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/70bc06a612ac4a84b8be3d3aee83626e : FAILURE in 59m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d54f60c9c3be46bf89f96c77d54bfc35 : FAILURE in 32m 18s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c2eb4566c2e4d529bb4f39e4f692e73 : SUCCESS in 1h 45m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/849ec630583b44f3bf94e15debd7f8b5 : POST_FAILURE in 1h 00m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3ab74bd62e447ac9276b562dec7d05a : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":2},{"id":"736deef7bce66dde73a566130d08fd922690b62a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-16 15:30:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6a3febb22884246136e2750b0cc92877e7c188ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 17:45:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/685d0874c7b941e1860d5413dbea8cb4 : SUCCESS in 12m 40s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb372d26abb540fb91e8997c8e201a06 : FAILURE in 1h 01m 59s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/184a2d200e324d22a21e48bc2987ab6e : FAILURE in 46m 51s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/cb8a82eb7ec2455c827ab16bb22f61fc : FAILURE in 56m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f1c910829004fbdae775a19c967f27a : FAILURE in 51m 13s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c069c27ddf84121b4458cd3eeca544e : SUCCESS in 2h 06m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa794414cd4f42398a487ab179da6ea9 : POST_FAILURE in 1h 06m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a12a4d1fa7b4337aeb388684ed087ce : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":3},{"id":"4be3e3550e3bd3c183e7614a3b8a193c9be23795","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-17 11:06:21.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"64c0c1ff49f4c72c2bbd257718baf3cf585054cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 14:25:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc76f57bda694d65a902630e43046e98 : SUCCESS in 18m 14s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/af501c645010491481409804edab2869 : SUCCESS in 1h 17m 22s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/61cccaa74efc4532a728dfe12b97ea97 : SUCCESS in 1h 08m 17s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/ce6944f3b96d437a8f8633081a7c9dee : TIMED_OUT in 3h 11m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9ca96eb7dc44c8f936a0447af68324d : SUCCESS in 1h 06m 33s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/079426bfa0f14fde9731f963652619a8 : SUCCESS in 2h 08m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dcf39c2aaf44101b238b0d132ab6ca8 : POST_FAILURE in 1h 07m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c09ed41d6c34aacadc3df3eb0d19142 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":4},{"id":"fd597536d2553c18cb18ea4a2058287dc4ab22b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-18 11:41:25.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"2e1fefaed4c33211fba06f726478f3bb92324549","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 13:49:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aeb24ccd13ba439aa218a077ec69b262 : SUCCESS in 17m 24s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c9bc8c7a450482681676ac8756e1a71 : FAILURE in 54m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a07b65eee423448c8a30b3dfadc7ba02 : FAILURE in 41m 30s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/3931c46a3c4a4b5c962e62c62d82e0be : FAILURE in 59m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d405112de2ed454d95eea89d0d8de621 : FAILURE in 52m 45s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e1c104c2868a4241b89201ede6044350 : SUCCESS in 2h 01m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f86bade16bc94fc98accd50d009657a5 : SUCCESS in 2h 01m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3060e926038c46b692376d1ab43f9763 : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":5},{"id":"412312a0820fa7a25ee342c0d924388a417df9b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-19 16:41:55.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"cbaa326c887efbf385d45284b3838fa895cb7e7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 18:12:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0698055c7b5451da9e03bab6c54e77c : SUCCESS in 17m 12s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe849e6b6b984bcbbd5b944b15340d91 : FAILURE in 31m 41s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/679fc49dc938475db5050335c74d1011 : FAILURE in 30m 51s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/a47b0c58261d4de08cadcd14e7ad4185 : FAILURE in 36m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4474aff811174c9db13a400f6506df53 : FAILURE in 29m 35s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/aea7604a9991476da89a360a61967e2d : FAILURE in 1h 20m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2d26fc9cfdc405cb75c7c3205e99cf4 : FAILURE in 1h 03m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60be9812231244b6aac8ac90a254c358 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":6},{"id":"0ec50708b21d14b81dc02c591e03ca75dabf6146","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-20 19:05:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bdffca5c53b9810b31bafe9d89cc7134cf4bccc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 21:17:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4102e1bd165a4cb3b352094f3ee6bd8c : SUCCESS in 14m 14s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/495b5c7d51a44ae182357005a27360da : SUCCESS in 1h 19m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/cfc072592d8a4b7fa75de1aeb32b36bb : SUCCESS in 1h 03m 13s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e9d468d381c04aa098b34eb4aa5c5864 : SUCCESS in 1h 26m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c933e359dda47a3b7bbf99601ae6a36 : SUCCESS in 1h 10m 11s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/cf953b89a2cd4038b6f2b8755e5e6d9a : SUCCESS in 1h 56m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06e229c2c451491ebb19aac1463c14fd : SUCCESS in 1h 35m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f11146059fc94a3d9ac47adb36f844c8 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":6},{"id":"7ffb073712ada520b1e972e3b8faa15ee48c5747","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-05-25 12:29:46.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"793fbfa61318fb50f55e5ac56aa76d9a1086384e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-25 19:18:42.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"e07f95cd89222982ab6775aae1b413d3f24e1971","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-25 20:00:40.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"20e844154b415fa3202c878cf59f3f7f0e9d257e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 20:03:23.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"38b9583bce747f7efc5029ea85c4194a4cc88b16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 20:47:35.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"6c2b95179dd99900cba4bef6aa92b874ce0891e4","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-05-27 14:00:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"de5d321fdd7e99e198270527833c26609ddb329d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 17:20:53.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adf9b242d3c149a6bf7288fe47961a55 : SUCCESS in 14m 00s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f1abfe0b3cde4887a6e8c29f82d04c2e : SUCCESS in 1h 28m 50s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/595e608d4bda45d28fa3159e0208a1d6 : SUCCESS in 1h 08m 43s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/94b472a189c4431c841f377f2d525674 : TIMED_OUT in 3h 16m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3bf2ca856094575a4169ce89c6288b8 : SUCCESS in 1h 20m 30s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0838e8440f448c08c1b681edc89d80a : FAILURE in 1h 31m 47s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43a5f69700574a1f82da564f846a9543 : SUCCESS in 1h 44m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfe05c130e724069b98b8fc25073b138 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":6},{"id":"bd80e7bad1341120efa49d66454e9e36ba5a50e8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-27 17:53:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f8fb0292935b2535cb393da8928ef7583e3be838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 19:22:17.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4917c5ba4e274645995e3a75a08e1434 : SUCCESS in 15m 47s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/26af824ff9ee402a9059102669eabf11 : SUCCESS in 1h 24m 10s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1615b87fa22d4c348fb78b763c118c9f : SUCCESS in 1h 10m 33s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/0c4ec03005f94771ad426ace70e869a4 : FAILURE in 17m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b83ae0fc5d74424976b7032f633653c : SUCCESS in 1h 13m 19s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/75428a76de3f42cdaaa75f017d9e7da1 : FAILURE in 1h 24m 02s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb02731cbd1f48549d031cb7e597afb5 : FAILURE in 1h 24m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25f4b8bf13ff4df98e4cb44b276abed1 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":6},{"id":"f295080d8c80ce804249b32f57a91ad741ead248","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-27 21:31:19.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"59dbf5b2e08d19fd685681bd374f436d0cd4e546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 23:06:29.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dabdbd1d9d545138f734b5a5b82d0ff : SUCCESS in 16m 56s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4a72622e29d4ac395a33eaf881f0e1f : SUCCESS in 1h 15m 01s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/ad9660e3badc4c53960d70b361f3b0df : SUCCESS in 1h 16m 11s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/be93037453e04a52a8ba16cf46e69d10 : SUCCESS in 1h 28m 10s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/939e58044fd84c249d6f5abc260a8b50 : SUCCESS in 1h 15m 45s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/d86eeae1d5a54d66b5362eee4dc9dfb0 : FAILURE in 1h 22m 39s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a31b5036ef884bd7ac104d4ff79f9e8e : FAILURE in 1h 15m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81721da275a54eeaa1611e7e1dec4647 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":6},{"id":"9fff0464898736f9655b00f2783d22f35abb16d9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-28 05:22:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"208fdd997a0c3f9fbe6485b38bc92c16071318db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 07:00:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bfb47c7a30341d599a537d9c840368d : SUCCESS in 12m 11s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0116217db02348d9a14e5caad8a28594 : SUCCESS in 1h 20m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/208e5faca7d54d62b7b7b744e878ab9a : SUCCESS in 1h 05m 32s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e107204e660b489abdaabd8782bae90f : SUCCESS in 1h 29m 31s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b83dd618d51432e879020dae18c7329 : SUCCESS in 1h 12m 21s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc7c0d4dd3854c7eb8ff0200d531bbaa : FAILURE in 1h 23m 16s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57a48e5960254f88a719db464586d0fb : SUCCESS in 1h 18m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ad16588072e472082874251671eedb9 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":6},{"id":"de9f49a57d533c7e8b6ae632a57001f6a8a0409c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 07:02:37.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9f164e5c11352a0bf3dde0eadc4f0ca679dd3a2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 08:38:26.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a61068a960d94a81b36a106ec6b5d4d5 : SUCCESS in 11m 27s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e454d4b794c46d8b2abc45bf66ad760 : RETRY_LIMIT in 10m 03s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/fa7fc2bb251245f2875678acbbf047c0 : SUCCESS in 1h 13m 32s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/9eec6109a0be4c2c8e880f588104da1d : RETRY_LIMIT in 11m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abee4a020b3447188f5a4b867ea49657 : SUCCESS in 1h 08m 33s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c9b337c1a2c45478b0fbaeb46c60cf2 : SUCCESS in 1h 31m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea3dd254d6734369bfe49936d6e0548b : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":6},{"id":"884528a65d976a29afb99e7398aaf17595ff7e83","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-28 20:42:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9274c25785f28089fd7ce575753802fac0fa07e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 22:29:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42fcdd051c864f35bc8b82b3973a609d : SUCCESS in 13m 26s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/317c0df62e914e3cb6f3cf8e99004d82 : RETRY_LIMIT in 11m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/94e7da46a08646839d3f5a4e2738f350 : SUCCESS in 1h 09m 45s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d01dc1eb77be4cbd8deaff28610f8ee8 : RETRY_LIMIT in 10m 18s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/678a389b354a47b0b8b83d11963f6926 : SUCCESS in 1h 21m 28s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/045c1ed0261a42619b1feb0a977e0fa7 : FAILURE in 10m 58s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58566ccd75e24db885b4de9b95c26d01 : SUCCESS in 1h 43m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a4c050f620d4da2848029756293cede : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":6},{"id":"f78e14f6d14567b7379abf4043cff154febb036c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-29 17:33:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6a4486e1ee3d198edb6212694592eebced62ae7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 19:02:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0cf65019ce84703b6af905b5bd03039 : SUCCESS in 16m 56s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/041465dc1b3e4f4ca2a8367e0b44f1d3 : SUCCESS in 1h 14m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d89150c69d2e452fb21763df7a7441b3 : SUCCESS in 1h 12m 39s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/94d0aab1ee5548ea966c3b24c14998a6 : SUCCESS in 1h 25m 38s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/477fc7b2dbcf45f5a81242ec57c6b91f : SUCCESS in 1h 11m 55s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/e06afb79833e418b9fd940f99dc8605c : FAILURE in 1h 07m 00s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a7d9c17f1b7467b82820936e2875cc6 : SUCCESS in 1h 16m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b457fc2462143a7945f342675a8d200 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":6},{"id":"bfec687a8c3f24179e6a3ddca3243fa9e44804dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 19:04:26.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"259aa8c173ceaefd733b6fb51d2341fec471eaa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 20:36:49.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a02a5dace2f7468f86538a4db2d5a87e : SUCCESS in 10m 58s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/4cdf8c9940844c3cb53c764c63111bc8 : SUCCESS in 1h 20m 34s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d7c2eb7c4fa44f5a8ad793821c1a7780 : SUCCESS in 1h 05m 33s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/3aac7defe7ca4b2fae9daf2137b65331 : SUCCESS in 1h 28m 18s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51f67386a7de464ebd8ffec9a51e6829 : SUCCESS in 1h 09m 01s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/645854e7bdac4535a6ee0245ffb7ad4a : SUCCESS in 1h 27m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eed4c4549b84ccda92e259d408cdf61 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":6},{"id":"24ba17fda96caa87b7391ab8810ae529c4afea02","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 20:36:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"c4f9abd5642f34eea9ec0a870b5036c673f6c315","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 20:38:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740320000ce54359b0c32a0f493037fc : SUCCESS in 52s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"309c235a8a0d805157357a28e9b0a170651edc64","revisions":{"f54ad92807efc768743ba20a45f37e8a102d9391":{"kind":"REWORK","_number":1,"created":"2022-05-12 16:39:51.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/18/841618/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/18/841618/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/1"}}},"commit":{"parents":[{"commit":"5f421412645ae579f1f97b9232496c36629946fd","subject":"Clean up legacy lsycnd, rsync and ssh key config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5f421412645ae579f1f97b9232496c36629946fd"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-12 16:37:53.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-12 16:37:53.000000000","tz":60},"subject":"Use the same vars file for all versions of centos","message":"Use the same vars file for all versions of centos\n\nChange-Id: I6e61e5e226c22d84d0a427359979dbb4338e890d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/f54ad92807efc768743ba20a45f37e8a102d9391"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/f54ad92807efc768743ba20a45f37e8a102d9391"}]},"branch":"refs/heads/master"},"be363b5b166d188b41be63e33629c0672de9fb10":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-16 11:03:38.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/18/841618/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/18/841618/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/2"}}},"commit":{"parents":[{"commit":"2eeae9bf3ccf45c985754207487fe185f165d98d","subject":"Clean up legacy lsycnd, rsync and ssh key config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2eeae9bf3ccf45c985754207487fe185f165d98d"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-12 16:37:53.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-16 11:03:38.000000000","tz":0},"subject":"Use the same vars file for all versions of centos","message":"Use the same vars file for all versions of centos\n\nChange-Id: I6e61e5e226c22d84d0a427359979dbb4338e890d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/be363b5b166d188b41be63e33629c0672de9fb10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/be363b5b166d188b41be63e33629c0672de9fb10"}]},"branch":"refs/heads/master","description":"Rebase"},"70148fd5265f8f073b55f6f5411147b38dca533c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-05-16 15:30:25.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/18/841618/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/18/841618/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/3"}}},"commit":{"parents":[{"commit":"e7a235fb657f6f8018ef85b89511f26271167b08","subject":"Clean up legacy lsycnd, rsync and ssh key config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/e7a235fb657f6f8018ef85b89511f26271167b08"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-12 16:37:53.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-16 15:30:25.000000000","tz":0},"subject":"Use the same vars file for all versions of centos","message":"Use the same vars file for all versions of centos\n\nChange-Id: I6e61e5e226c22d84d0a427359979dbb4338e890d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/70148fd5265f8f073b55f6f5411147b38dca533c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/70148fd5265f8f073b55f6f5411147b38dca533c"}]},"branch":"refs/heads/master","description":"Rebase"},"9796d554b640380b342b0354d76fd176b353135e":{"kind":"NO_CHANGE","_number":4,"created":"2022-05-17 11:06:21.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/18/841618/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/18/841618/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/4"}}},"commit":{"parents":[{"commit":"de2562c2721a4753bf8d9a4315becc7141aac08f","subject":"Clean up legacy lsycnd, rsync and ssh key config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/de2562c2721a4753bf8d9a4315becc7141aac08f"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-12 16:37:53.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-17 11:06:21.000000000","tz":0},"subject":"Use the same vars file for all versions of centos","message":"Use the same vars file for all versions of centos\n\nChange-Id: I6e61e5e226c22d84d0a427359979dbb4338e890d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/9796d554b640380b342b0354d76fd176b353135e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/9796d554b640380b342b0354d76fd176b353135e"}]},"branch":"refs/heads/master","description":"Rebase"},"0a69283c92b4cda92b9876b2aa91b7c20f938709":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-05-18 11:41:25.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/18/841618/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/18/841618/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/5"}}},"commit":{"parents":[{"commit":"d6ab0c2af73f072b5b67eeb97176a3f033803c22","subject":"Clean up legacy lsycnd, rsync and ssh key config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d6ab0c2af73f072b5b67eeb97176a3f033803c22"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-12 16:37:53.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-18 11:41:25.000000000","tz":0},"subject":"Use the same vars file for all versions of centos","message":"Use the same vars file for all versions of centos\n\nChange-Id: I6e61e5e226c22d84d0a427359979dbb4338e890d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0a69283c92b4cda92b9876b2aa91b7c20f938709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/0a69283c92b4cda92b9876b2aa91b7c20f938709"}]},"branch":"refs/heads/master","description":"Rebase"},"309c235a8a0d805157357a28e9b0a170651edc64":{"kind":"NO_CHANGE","_number":6,"created":"2022-05-19 16:41:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/841618/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/18/841618/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/18/841618/6"}}},"commit":{"parents":[{"commit":"45d07fcf0c6605c9da72147d243cbe8d7e1ce8a7","subject":"Clean up legacy lsycnd, rsync and ssh key config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/45d07fcf0c6605c9da72147d243cbe8d7e1ce8a7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-12 16:37:53.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-05-19 16:41:55.000000000","tz":0},"subject":"Use the same vars file for all versions of centos","message":"Use the same vars file for all versions of centos\n\nChange-Id: I6e61e5e226c22d84d0a427359979dbb4338e890d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/309c235a8a0d805157357a28e9b0a170651edc64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/309c235a8a0d805157357a28e9b0a170651edc64"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
