)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~843931","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I941359b1b42aa4a874230a32b438dcefddfb2acb","project":"openstack/openstack-ansible-repo_server","branch":"master","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-06-04 00:04:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I941359b1b42aa4a874230a32b438dcefddfb2acb","subject":"Restart nginx after removing old repo content","status":"MERGED","created":"2022-05-30 15:13:37.000000000","updated":"2022-06-04 00:05:51.000000000","submitted":"2022-06-04 00:04:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843931","meta_rev_id":"c4b88a97e5cc7b0c81c4e22de94f1c82a1c6f6c7","_number":843931,"virtual_id_number":843931,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-04 00:04:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-06-03 12:22:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2022-06-03 06:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2022-06-03 12:22:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-30 16:52:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-30 21:56:39.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"CC"},{"updated":"2022-05-31 14:29:06.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2022-06-03 06:54:47.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2022-06-03 12:22:28.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"4ae3665a8e0ff99928a43a4a98138f450639f311","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-05-30 15:13:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c6f26a5193435e80ae628405e7eae29b7f8e0e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 16:52:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/738359e4d37c423d9af92e2e9f7de4c6 : SUCCESS in 14m 46s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/de65ffccaae943aaa3b242b9097e8e27 : SUCCESS in 1h 13m 18s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/53f73578caae455a83f410a77f5f11c5 : SUCCESS in 57m 45s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/cd3b392f23a741bd898fab435e9dd144 : SUCCESS in 1h 30m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94486d612b4644f5ae36181501f5eca3 : SUCCESS in 1h 06m 44s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c0ae7cacf8b54ba9bbf31f2fcaf5a59a : SUCCESS in 1h 25m 42s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01df652070404dda88fd2f2a6ce3620c : FAILURE in 1h 00m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/302b30ce6fd7486395fdbd91efcfd710 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"da537130b305acce5c23c569f3e3d8b67a873d48","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-05-30 21:56:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"72d4e6cfd5cc0a8228e677aada8b30836a5e761a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 23:58:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8e4fb09f4b943a3aed0c000a1cd0041 : SUCCESS in 16m 36s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6a99350a1214ee88ed9c30b9281618e : SUCCESS in 1h 19m 57s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/62bb3091f1bd4f71a40f296d9a5e065c : FAILURE in 1h 06m 03s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/2efa52c03a6e456f98db1c38fc59285a : SUCCESS in 1h 11m 16s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/627a0dc9b4e341d2b030232a184f90b7 : SUCCESS in 1h 23m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ff8db3a27e3430b8ae3e093543eab14 : SUCCESS in 1h 08m 31s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b004b74b8ec4282aea2d09bb8a575ab : SUCCESS in 1h 55m 02s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f237028ee3ae482696127dfd40b01aef : SUCCESS in 1h 44m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66cde5f6ad5c4e169a8140b0dcedf7f8 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1e1239a1409fc07d33eb0846772f9522dc2e8db9","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-05-31 00:50:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb9902f2e7b60370bfe83dd3af3720fd59091e8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 04:12:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55d85bdc78704049956bf65bb05d3d22 : SUCCESS in 13m 31s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/48a993f99eb241708d9ec0d8f0276d44 : SUCCESS in 1h 36m 29s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac936c85697f449c900845c96ea31805 : TIMED_OUT in 3h 14m 49s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1d55ffe161db4cbaa2574c55175feffd : SUCCESS in 1h 11m 23s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/4b4bfad6fd124a4b92d8cab954b021af : SUCCESS in 1h 26m 38s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/547dacf4df1446c3b18c70eac5c79f67 : SUCCESS in 1h 07m 49s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/3b85271d75d347e6960b5fada1d235e0 : SUCCESS in 1h 27m 44s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15f7e2060ba24459a17aaeb1d43645ee : FAILURE in 1h 08m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6798a90f6014f7887f16a59c487b118 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2f3890925c467b2516c1ea0ab786dcfcb00e9c1b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-05-31 14:29:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3cd6966ea5d46a97c897d139534df6129c30a300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 17:49:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2a5b2ff71aa496ca2177241d41346b6 : SUCCESS in 14m 33s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/948bbcc8007545328bfc4ed9038f5b44 : SUCCESS in 1h 41m 22s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d59b22b1103849ec9cf0dea28afd6ad4 : TIMED_OUT in 3h 15m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/b02c20b545af455eaadf38d8d838b3f9 : SUCCESS in 1h 12m 24s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/da1f82e245444f118579791a4af910fa : SUCCESS in 1h 28m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/831f0e94ab5f4e009a6d127ac886cd29 : SUCCESS in 1h 32m 09s\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a79d262602f407580fb32f8f6d950a3 : SUCCESS in 1h 53m 13s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff5c906b3be648ffb6395d446758ffc5 : SUCCESS in 1h 45m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e34a31cf93094d8f901c573bd105d033 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"dddbe79136ea9a46737b643ee3415dc14408c232","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-06-02 18:12:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b85fadeb3dad9d15254ddd5d2d12798804f7cfef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 20:07:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34481c7278bd4b608496c19d6216b2c6 : SUCCESS in 15m 02s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/54fe51a2e4594f3085adaba33981c8df : SUCCESS in 1h 46m 46s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd6f2029d85841f38392c2e7cf105bc6 : SUCCESS in 1h 12m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/ce34c0ad9254440fb54c8f994ccbb55c : SUCCESS in 1h 11m 00s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/2edb951306c54ab3b44427e68974aae5 : SUCCESS in 1h 22m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1d20527acb94efeabc2f4233fd906f3 : SUCCESS in 1h 31m 03s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99214998bba04f6cafb8d1bb659c3405 : SUCCESS in 1h 22m 35s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a8895e35947443ca8e825edd4705f26 : FAILURE in 1h 41m 35s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b072176440448b2a5f59f5bf0633b4f : SUCCESS in 1h 46m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e74d16f59d94ddebf98a31ac99d86e0 : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"82d1e54cb88ec8fcfa96f9e370812dc7236fbc93","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-03 06:54:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4af42f601153b3ab3388d665478ce341df0775c7","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-06-03 12:22:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"04705d954c35f233f4658d60b2d86379f141b845","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 12:24:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f38cfdee62309080bb4397076d69d76d3e501490","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 14:13:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e28ebf322764953b702e811f224d424 : SUCCESS in 13m 34s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/5587e1cf930548d89fd06e60bc39dcb0 : SUCCESS in 1h 30m 12s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a48d9604c3147bdaeca61ab4a9bcc32 : SUCCESS in 1h 26m 04s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/8a387ab22f3b41df83a336bd618ed0c4 : POST_FAILURE in 1h 34m 07s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/d01bd3d7a99e486f890a71c6e0fff6d9 : SUCCESS in 1h 25m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25139149a20c4bc59411c9b66cead4a9 : SUCCESS in 1h 12m 48s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4645cb7c434e423cadd11d4a0d628802 : SUCCESS in 1h 48m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51fb1ba07651407a96a299e28d83b8b9 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"eb0e0c28894bcc5f9cb9c654b99510242d73620b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-03 15:21:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c0a74b32aed8f5dc1ac0ae7e24afdd93360e804","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 17:19:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e36fec0c7b944d0818f73b97c432cc5 : SUCCESS in 11m 55s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/c8392d9d414c4fb7a9b873310afd2aac : SUCCESS in 1h 17m 17s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a486d51b4eef414ba919523fbab3d3ec : SUCCESS in 1h 16m 27s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/54f9137b196c42c2bbe2d2f6f4bc6483 : SUCCESS in 1h 15m 20s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/76bd8dc4c46843008c5817699140ebec : SUCCESS in 1h 30m 20s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddd2df7f33a0430a8ec41cef18b00a23 : SUCCESS in 1h 15m 11s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef09155446314d43b9a9795d7a2357c4 : SUCCESS in 1h 08m 14s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/f1fb9cdcee154725a368cca2d2d34445 : SUCCESS in 1h 35m 41s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4306df1f22ad41d9be117f37e7387d05 : SUCCESS in 1h 54m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b725db912e84b388e6a1b4730c1efab : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"50f21efe8e7661b7df987d4790ea9acc4d5855ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 17:21:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6f0a7935629534a69dfcdf819384e84f0ad1aea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 19:18:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a92a46f1e584e1eb6200e086e62f1f8 : SUCCESS in 18m 08s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/de10a2e54b344d35a5cf0ba99b349ed5 : POST_FAILURE in 1h 45m 31s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/18f7bf8590444d75b7856b8c57e9696d : SUCCESS in 1h 47m 41s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/f7f86869a9bb44cb8da8970749d2ea8c : SUCCESS in 1h 10m 35s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/24a96fcb6eab4674934c94fb237b9e8b : SUCCESS in 1h 28m 24s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/447a18926d86457db28471675fe156f7 : SUCCESS in 1h 14m 52s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32da6c70991a4749a394da4faf5b98a0 : SUCCESS in 1h 50m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c69c4c847644f618fc50e5ebdca6220 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4a33781511dc8f48f6e49692d090100956105556","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-03 20:16:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"083cdabb3c549e003de7b33f55f16ce134ac130c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 22:10:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f73f9d5c9a214e37baf629ea27360520 : SUCCESS in 14m 25s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b2d6a7eca1f410c91ef1d71e1c36e8a : SUCCESS in 1h 29m 52s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4fc1789194534117b36346f141b4e5d6 : SUCCESS in 1h 23m 52s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/3ac32cf78df749d3846be195f4dd1bdb : SUCCESS in 1h 10m 16s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/52e915eda16c4be8b5941c1d1e1149da : SUCCESS in 1h 24m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84840f177775470f9d70330344946730 : SUCCESS in 1h 17m 30s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac57976ae257457da69073f397e5a027 : SUCCESS in 1h 14m 45s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/bfa843a9da164e83a3c6d59dd6d32a3d : FAILURE in 1h 46m 40s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c650588c247342789d8be61e4fd8ca5a : SUCCESS in 1h 33m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddd12086b49a4fc9b1f4982b06969dc7 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e293b054dc8fbec16056ec955148c9c6fb00ddd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 22:13:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a8cd4f015b553f2324cb69774533b3cb3ca8d50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 00:04:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39ff8c71af2c4cd6aefeec7885465c1b : SUCCESS in 12m 48s\n- openstack-ansible-deploy-infra_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/eac55b25c01b4ff1ad8470db796cf23e : SUCCESS in 1h 12m 32s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/6440b65df1da444f9c81b9d19bce1053 : SUCCESS in 1h 20m 44s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/1956034f91cf443584cb6bb8d78c75fd : SUCCESS in 1h 06m 37s\n- openstack-ansible-deploy-infra_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b5287ee14c41428eb912fd431d50d96f : SUCCESS in 1h 25m 06s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db6413b8787d443eba2c59478aed0b4f : SUCCESS in 1h 06m 11s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02088f8956764b668561cd0e1f264656 : SUCCESS in 1h 23m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d69fe8b0a204691a2dae78a19145a04 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e77bde5ce8235b86a072b5821c79b67dd70b90c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 00:04:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c4b88a97e5cc7b0c81c4e22de94f1c82a1c6f6c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 00:05:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d616ec5a026249dcaa9a5404c2090447 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9ecec103d0f823123baf429a4c9dd1361003096","revisions":{"a9ecec103d0f823123baf429a4c9dd1361003096":{"kind":"REWORK","_number":1,"created":"2022-05-30 15:13:37.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/31/843931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/31/843931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/31/843931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/31/843931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/31/843931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/31/843931/1"}}},"commit":{"parents":[{"commit":"b0fe7fd14d0bbbeff4256fe21b767b39420d27cc","subject":"Switch sphinx language to en","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b0fe7fd14d0bbbeff4256fe21b767b39420d27cc"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-30 15:11:00.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-05-30 15:13:27.000000000","tz":60},"subject":"Restart nginx after removing old repo content","message":"Restart nginx after removing old repo content\n\nThe version of nginx on centos-8 appears to keep file handles open,\npossibly the old /var/www/repo directory persistently. Once the old\ncontent is removed and the new shared filesystem mount is created at\nthe web root, ensure that nginx is restarted to close any file handles\nwhich are now stale.\n\nChange-Id: I941359b1b42aa4a874230a32b438dcefddfb2acb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/a9ecec103d0f823123baf429a4c9dd1361003096"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/a9ecec103d0f823123baf429a4c9dd1361003096"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
