)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~876429","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I5c5d3dd5689ac122781303ad21dacc8a1fa746eb","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"tls-backend","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2023-05-01 21:03:21.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-04-29 09:36:31.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I5c5d3dd5689ac122781303ad21dacc8a1fa746eb","subject":"Add TLS support to repo_server backends","status":"MERGED","created":"2023-03-03 19:35:44.000000000","updated":"2023-05-01 21:04:25.000000000","submitted":"2023-05-01 21:03:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876429-tls-backend","meta_rev_id":"23214cdf1e3550eb4a4adbe5c736e5c59c1a15f3","_number":876429,"virtual_id_number":876429,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-01 21:03:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2023-05-01 19:08:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2023-04-29 09:36:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2023-05-01 19:08:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-03 20:34:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-04 16:43:25.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-04-24 14:59:25.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-05-01 19:08:34.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"a11f13874c4636f88d48ea980debd0451c05af71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-03-03 19:35:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c49334db388e3957e41c40f09dd02da92f527546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 20:34:43.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/871189 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"daa6b5d728f56362f51bce7e7a6b6fd36756a317","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-03-14 22:37:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5743bd3248dad45d2e14d1ad9dd7e161cd26a844","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-14 22:38:05.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/openstack-ansible for 871189,23","accounts_in_message":[],"_revision_number":2},{"id":"458d95cda91cce697e698c835c0196ef1c962172","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-04 07:26:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2f2c54bb3e7eb107cf9f90ba38423b469848814b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 09:12:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebc98a3f16e54c1a8f901baa4369cefc\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a92941aa730a4965bcc5e2db616d6170 : SUCCESS in 15m 55s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/50206518986049a099e90e55a1fb783f : FAILURE in 39m 25s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/071afe5ece1c451ab8ffb69c9b9d5296 : SUCCESS in 1h 37m 17s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/c15f29b7a0544085be31008cd1a29179 : SUCCESS in 1h 13m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2778a0ec7a2e41e4abf03c8d84e8b184 : SUCCESS in 1h 17m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50eae1c5295649b08f0c93f3f6cde187 : SUCCESS in 1h 17m 58s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1dd400220f241edb73e85a45cd103e8 : SUCCESS in 1h 12m 12s\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fc7909291bd4098b4d65090c5b752d3 : FAILURE in 35m 18s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b95f641769044c219580f00c21c1c1e7 : SUCCESS in 1h 35m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3de154a735c4f3183a57983b8a0013b : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":2},{"id":"d12c28791c202e066c9ca9b53b427483c62819d2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-04-04 16:43:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"17fdd486330f3921e8cd39c195d59be481487e1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 16:43:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/69ec2ae32c9b403c886274f4dd710680\n\n\nWarning:\n  Change 871189 in project openstack/openstack-ansible does not share a change queue with 876429 in project openstack/openstack-ansible-repo_server","accounts_in_message":[],"_revision_number":2},{"id":"8b9be873b09b52e2b512cfb96575df9d29bc6bcb","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-04-04 17:35:40.000000000","message":"Removed Workflow+1 by \u003cGERRIT_ACCOUNT_25023\u003e\n","accounts_in_message":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"_revision_number":2},{"id":"b5ecb0a089370a9428a02531f86a16e27d8d89de","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-04-04 17:35:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ea9aa2b4e8d063c2995949d01c5f18d1cef085e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 19:18:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e85849d6c3934ea287c849ec027335df\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4849ef64b88b48c89034df24a66bb8df : SUCCESS in 14m 00s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/0978537ae335476da9fb2ac2f9aeab7a : FAILURE in 41m 38s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/472057527d534ef0b5b11b555b738122 : SUCCESS in 1h 34m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e243d2a6e1594f96b949f226de9d2184 : SUCCESS in 1h 06m 02s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c13c0fbeee2b46049bc05112448b14ff : SUCCESS in 1h 09m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4c840ffd11f4c75b8ab67cd80430016 : SUCCESS in 1h 15m 36s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3fafba43ffe747718049373ec658582c : SUCCESS in 1h 07m 46s\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4d06bc98f5046cf8c807c88c61d177f : FAILURE in 30m 23s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28294414c5b740cca19ddb7ceea4c931 : SUCCESS in 1h 25m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00d9d75b6f104c3d98c0a96045f31aca : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"9f9fa86f09ea7975a5b27325b7a395657191519a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-05 22:59:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"01b5ebf63f094d329e554f1a2777a4f85bbad186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 00:48:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e24379abec1433c94c2da26559d1ef3\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3572b58e27fc4ad59e0a45a4780baf0c : SUCCESS in 9m 35s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/54e3d9e394074a50a9df7a30db95b461 : RETRY_LIMIT in 14m 39s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a07981a271944318a4b88d69c2ce6803 : SUCCESS in 1h 41m 32s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e67bf5f2280f4b6b8af18496e6973f89 : SUCCESS in 1h 07m 32s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/955fddcf9dbc4d6b9de8b5aa02f56596 : SUCCESS in 1h 13m 00s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d869223c67034336b186f4b2886f1888 : SUCCESS in 1h 19m 43s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a4fe476d4194898a3cc015c152728cf : SUCCESS in 1h 23m 39s\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2db206ed1184336834ef8892d9ae51a : FAILURE in 11m 39s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd8cb82243f44c54b27754cb2afeee38 : SUCCESS in 1h 21m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/346a951824a742e7a9f27c88f4b335f0 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":3},{"id":"8c2395643d4d22fb20f5c92f6d5a412f5c5a8030","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-07 21:25:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"4f386266f535b0e8d6afe63601e83a06a827b1a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-07 21:29:02.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"5a23ebd816385b19f559cddf6e404119f3a12214","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 23:27:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf7814f47c644c4794b67079022aeda1\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a36e655ff4c145d98d68867e03bba8b3 : SUCCESS in 11m 49s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a73d607c3d9e40c2ae2466c7db0e2c38 : FAILURE in 24m 09s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/053384ed443647b8a1e19e72a9dad323 : SUCCESS in 1h 27m 00s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/dacc1d67e3274d4f82fe818d330f9851 : SUCCESS in 1h 11m 55s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/629c331462314cdbb22e5bc81366587d : SUCCESS in 1h 10m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51316c23470d4a85877e4f4d21e746fe : SUCCESS in 1h 10m 32s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a61fcbf0084843728becb9ed580f23d9 : SUCCESS in 1h 16m 07s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2c4c3438bde441209c493cbb69cb41ce : SUCCESS in 1h 41m 31s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54078e0184084bf5813744ee02041b23 : SUCCESS in 1h 25m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78a7329f59cb4d77a8d0e79704e93b55 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":5},{"id":"b0b19937afdcdade9ddffb8512178308847c632e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-24 14:59:25.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"49b6d44dbd1e9ca0460d5781f16cf897c0793e0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-24 22:08:13.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"ee1eae8ed26b10338a98d5beb96b7c5d6f08c418","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-24 22:09:05.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"3a95da64f7c4c8218543074cb3d0685fc7bf8be2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 23:59:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93034cae02764f1aacc9b018f83d82e9\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b565d11f72d3437f9c4d00783ac21e04 : SUCCESS in 10m 48s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/31f8d66198214f4da5b33d9866f1fc4c : FAILURE in 39m 59s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1af33197d8f64ff0bfd15bf031648195 : SUCCESS in 1h 36m 20s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/e6de91f8ea9641aaa2ce6cbabf37999b : SUCCESS in 1h 21m 57s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4463af2066f644c6b2bb946a2609032f : SUCCESS in 1h 12m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b9f69aaa53e476dacfeb2a68995f51c : SUCCESS in 1h 23m 13s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0bdd15e3ace4947ba01b6509fae5ec5 : SUCCESS in 1h 17m 46s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ef414c97724e423395455499e517a7f1 : SUCCESS in 1h 44m 56s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2b2f6adaa074f768f3066aee20980a9 : SUCCESS in 1h 30m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42eaa986bdb044499b7c61eeb6ce2ac7 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":6},{"id":"7c1e386b678a508d50713f6adf4cff988813e367","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-25 12:12:33.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"bdfd0183d9d562b51d090179bcf106d449d88a0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-25 13:25:06.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"267652131885d4399024dbd1b47f742ee64bf042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 15:23:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/517447b7de794b78a7cc367a5ab679a6\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8a78ec0a91447be9c0bb861a673c135 : SUCCESS in 11m 15s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/d520613b01e74a8baf89f3ec7cf9ff3a : FAILURE in 42m 55s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5b8800e0196e4e31a48520588e243eb2 : SUCCESS in 1h 38m 43s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/21820a0a41254093a29749e604ba511b : SUCCESS in 1h 27m 37s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf42cfdbffa444c0b5cbbf8f7b3066b1 : SUCCESS in 1h 08m 13s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbf660a3107743d78e84a94df6316200 : SUCCESS in 1h 13m 33s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/279c93c7f9994b2eb8292f796cd32368 : SUCCESS in 1h 15m 45s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/84302f7652cd44eaa06c4af29f916b40 : SUCCESS in 1h 48m 26s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9049560e600c4b538b9917622ce930c7 : SUCCESS in 1h 41m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b96c21a790324209a2d6c2025a4df16a : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":8},{"id":"c54d7995bca55de5c163e622e58489d33fe1b8d6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-25 20:25:27.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e944afdb50e276867636ae16d3635020ab7567db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-28 09:27:16.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"ec8fa4b64911863727d1e2a83f439ff7fcb19885","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-28 09:28:46.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3cfe892673a51d280f8e6227d8d6f5a277bd8227","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-28 09:46:31.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9783e390044c413066c59a5363ef2c2a5a51cc03","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-28 10:30:23.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ca4b436f6f124b4e8391b3e671f125530a0ff1a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 11:11:08.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bceb1b7e80642579d21f38e0d22cfff\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0577d3098d0447d0b00191cdbc14e575 : SUCCESS in 11m 06s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/128ce0e97b444d99aa257cd5f989cfb9 : FAILURE in 35m 33s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f2df815643004e749f3bc0ee7d323094 : SUCCESS in 1h 34m 55s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/382f91345098459fb66ca6bb9b14cf37 : SUCCESS in 1h 06m 59s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/080b97dd6ba749d69014f51d63eeb429 : SUCCESS in 1h 18m 13s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c43f0fa87d0140619cb3eb85568fde6c : SUCCESS in 1h 18m 36s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/352812d10e5747c281b70ad3406f2322 : SUCCESS in 1h 23m 54s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1da1dbdea61a4dd7b0e3b1523eba9e46 : FAILURE in 39m 11s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3fc6a528fc34cfe84d2ba99ebd81319 : SUCCESS in 1h 37m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dad44839d6f4b0b8eb08d264e6779ab : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":9},{"id":"f24689a598bd40a7508c30e6c06a6a81ed94b9c7","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-04-28 11:29:22.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"bfd584c5e5a4c61784fbe63d578a0b1a4254c386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 13:37:04.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fac643035414c3a9bb5b27e389ec034\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57e2bd66d4344dc783faf8e8a5b532e9 : SUCCESS in 14m 03s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/851824eb23414feb999eb8a5936994ef : FAILURE in 43m 17s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/48079850b43646b7952acf52a79493fb : SUCCESS in 1h 32m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/de058e67eb9145a183fde9fc4628c46f : SUCCESS in 1h 17m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cb1dc6e58a94cdeac5b6441ba4c21e5 : SUCCESS in 1h 16m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de5c3e8ecffe4bb084633d6ddf22c210 : SUCCESS in 1h 19m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20d9d25fc607437da6962a61a013af1d : SUCCESS in 1h 32m 44s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/20aff8a542ac45d199d787dc393d45be : SUCCESS in 1h 59m 30s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db7c15c170124e34bb2372b55aa02697 : SUCCESS in 1h 38m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d0df9be4156419388bd22b892a450cb : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":9},{"id":"bb6cabcb54a3431cb16f7942dc0b92c0cd262f16","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-04-29 09:36:31.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"61abd28883f3ba50d5c441179e95a7882753ced3","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2023-05-01 19:08:34.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"7040878f7947133a3671b4b44b2a0f1ca8d838ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 19:10:45.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"d9404b7fadec29897c35c8df7b2bdf491b90bb19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 21:03:21.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0cc8b3321734454b4571968c516402d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85f6320853234305b84b2e3dd0d89c17 : SUCCESS in 11m 39s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/65769e2dd257490e97470e50c7a78194 : SUCCESS in 1h 26m 08s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/dc0e5c98013947c1bff859dfdbf1196e : SUCCESS in 1h 18m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/003b61f839f74a359842d1ec1df484ce : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22abc6a463894b9dbf98087f99e41ff9 : SUCCESS in 1h 11m 08s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8ea25b26975149f3867cd77a7d1f6ee2 : SUCCESS in 1h 46m 21s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31d1ca3a46c24ae3a78448dc6f606a33 : SUCCESS in 1h 31m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04bff39a3ab34adabfa48a4244883f6b : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":9},{"id":"742704ec2f07d72efd36f7fc5326c03af5699b12","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 21:03:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"23214cdf1e3550eb4a4adbe5c736e5c59c1a15f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 21:04:25.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40e0a8d92b864be1a95c2d7694846ebb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ea07c13d2fe49919413d63f28ffcaa1 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"2d0e465fd3bab22790213993eecafa6727785924","revisions":{"70a951a557124dbd2436c4be7e19c783dc3e389b":{"kind":"REWORK","_number":1,"created":"2023-03-03 19:35:44.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/29/876429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/29/876429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/1"}}},"commit":{"parents":[{"commit":"b9d9ccfc8a781759f421d5c8aa1a9276295765b5","subject":"Update tox.ini to work with 4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b9d9ccfc8a781759f421d5c8aa1a9276295765b5"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-03 19:34:19.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-03 19:35:41.000000000","tz":60},"subject":"Add TLS support to repo_server backends","message":"Add TLS support to repo_server backends\n\nBy overriding the variable `repo_backend_https: True` HTTPS will\nbe enabled, disabling HTTP support on the repo_server backend.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/871189\nChange-Id: I5c5d3dd5689ac122781303ad21dacc8a1fa746eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/70a951a557124dbd2436c4be7e19c783dc3e389b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/70a951a557124dbd2436c4be7e19c783dc3e389b"}]},"branch":"refs/heads/master"},"2aecebce3472d998ab802ca4d2ed73f094e4af16":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-14 22:37:30.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/29/876429/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/29/876429/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/2"}}},"commit":{"parents":[{"commit":"3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505","subject":"Turn off absolute_redirect for nginx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-03 19:34:19.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-14 22:36:38.000000000","tz":60},"subject":"Add TLS support to repo_server backends","message":"Add TLS support to repo_server backends\n\nBy overriding the variable `repo_backend_https: True` HTTPS will\nbe enabled, disabling HTTP support on the repo_server backend.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/871189\nChange-Id: I5c5d3dd5689ac122781303ad21dacc8a1fa746eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2aecebce3472d998ab802ca4d2ed73f094e4af16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2aecebce3472d998ab802ca4d2ed73f094e4af16"}]},"branch":"refs/heads/master"},"284fddcc774a87048822f5018891a9f060d15b0c":{"kind":"REWORK","_number":3,"created":"2023-04-05 22:59:22.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/29/876429/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/29/876429/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/3"}}},"commit":{"parents":[{"commit":"3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505","subject":"Turn off absolute_redirect for nginx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-03 19:34:19.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-05 22:58:29.000000000","tz":120},"subject":"Add TLS support to repo_server backends","message":"Add TLS support to repo_server backends\n\nBy overriding the variable `repo_backend_https: True` HTTPS will\nbe enabled, disabling HTTP support on the repo_server backend.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: I5c5d3dd5689ac122781303ad21dacc8a1fa746eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/284fddcc774a87048822f5018891a9f060d15b0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/284fddcc774a87048822f5018891a9f060d15b0c"}]},"branch":"refs/heads/master"},"c7ca27e22a6083e09ed5a63c693457e01673e777":{"kind":"REWORK","_number":4,"created":"2023-04-07 21:25:41.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/29/876429/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/29/876429/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/4"}}},"commit":{"parents":[{"commit":"3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505","subject":"Turn off absolute_redirect for nginx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-03 19:34:19.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-07 21:25:08.000000000","tz":120},"subject":"Add TLS support to repo_server backends","message":"Add TLS support to repo_server backends\n\nBy overriding the variable `repo_backend_https: True` HTTPS will\nbe enabled, disabling HTTP support on the repo_server backend.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: I5c5d3dd5689ac122781303ad21dacc8a1fa746eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c7ca27e22a6083e09ed5a63c693457e01673e777"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/c7ca27e22a6083e09ed5a63c693457e01673e777"}]},"branch":"refs/heads/master"},"386b680d42d4cfba8c37fae5597a342f35897c90":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-04-07 21:29:02.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/29/876429/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/29/876429/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/5"}}},"commit":{"parents":[{"commit":"3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505","subject":"Turn off absolute_redirect for nginx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-03 19:34:19.000000000","tz":60},"committer":{"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","date":"2023-04-07 21:29:02.000000000","tz":0},"subject":"Add TLS support to repo_server backends","message":"Add TLS support to repo_server backends\n\nBy overriding the variable `repo_backend_ssl: True` HTTPS will\nbe enabled, disabling HTTP support on the repo_server backend.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: I5c5d3dd5689ac122781303ad21dacc8a1fa746eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/386b680d42d4cfba8c37fae5597a342f35897c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/386b680d42d4cfba8c37fae5597a342f35897c90"}]},"branch":"refs/heads/master","description":"Edit commit message"},"40e3890d057e64b14457eeb1fb2662a535a3ff3b":{"kind":"REWORK","_number":6,"created":"2023-04-24 22:08:13.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/29/876429/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/29/876429/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/6"}}},"commit":{"parents":[{"commit":"3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505","subject":"Turn off absolute_redirect for nginx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-03 19:34:19.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-24 22:08:11.000000000","tz":120},"subject":"Add TLS support to repo_server backends","message":"Add TLS support to repo_server backends\n\nBy overriding the variable `repo_backend_ssl: True` HTTPS will\nbe enabled, disabling HTTP support on the repo_server backend.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: I5c5d3dd5689ac122781303ad21dacc8a1fa746eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/40e3890d057e64b14457eeb1fb2662a535a3ff3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/40e3890d057e64b14457eeb1fb2662a535a3ff3b"}]},"branch":"refs/heads/master"},"7273f04095fbeae897544465cfb4f6f5e2b6d5a6":{"kind":"REWORK","_number":7,"created":"2023-04-25 12:12:33.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/29/876429/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/29/876429/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/7"}}},"commit":{"parents":[{"commit":"3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505","subject":"Turn off absolute_redirect for nginx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-03 19:34:19.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-25 12:12:25.000000000","tz":120},"subject":"Add TLS support to repo_server backends","message":"Add TLS support to repo_server backends\n\nBy overriding the variable `repo_backend_ssl: True` HTTPS will\nbe enabled, disabling HTTP support on the repo_server backend.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: I5c5d3dd5689ac122781303ad21dacc8a1fa746eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/7273f04095fbeae897544465cfb4f6f5e2b6d5a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/7273f04095fbeae897544465cfb4f6f5e2b6d5a6"}]},"branch":"refs/heads/master"},"33becd8ccbb1150b0b13dc5e7cd1fb59d3e58528":{"kind":"REWORK","_number":8,"created":"2023-04-25 13:25:06.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/29/876429/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/29/876429/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/8"}}},"commit":{"parents":[{"commit":"3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505","subject":"Turn off absolute_redirect for nginx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-03 19:34:19.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-25 13:24:35.000000000","tz":120},"subject":"Add TLS support to repo_server backends","message":"Add TLS support to repo_server backends\n\nBy overriding the variable `repo_backend_ssl: True` HTTPS will\nbe enabled, disabling HTTP support on the repo_server backend.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: I5c5d3dd5689ac122781303ad21dacc8a1fa746eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/33becd8ccbb1150b0b13dc5e7cd1fb59d3e58528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/33becd8ccbb1150b0b13dc5e7cd1fb59d3e58528"}]},"branch":"refs/heads/master"},"2d0e465fd3bab22790213993eecafa6727785924":{"kind":"REWORK","_number":9,"created":"2023-04-28 09:27:16.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/29/876429/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/29/876429/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/876429/9"}}},"commit":{"parents":[{"commit":"3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505","subject":"Turn off absolute_redirect for nginx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-03 19:34:19.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-04-28 09:27:09.000000000","tz":120},"subject":"Add TLS support to repo_server backends","message":"Add TLS support to repo_server backends\n\nBy overriding the variable `repo_backend_ssl: True` HTTPS will\nbe enabled, disabling HTTP support on the repo_server backend.\n\nThe ansible-role-pki is used to generate the required TLS\ncertificates if this functionality is enabled.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/879085\nChange-Id: I5c5d3dd5689ac122781303ad21dacc8a1fa746eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2d0e465fd3bab22790213993eecafa6727785924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2d0e465fd3bab22790213993eecafa6727785924"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
