)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~877429","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I9e55508996d9b24437870f2f23dca5db7827fee1","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"tls-backend","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2023-03-16 13:05:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9e55508996d9b24437870f2f23dca5db7827fee1","subject":"Turn off absolute_redirect for nginx","status":"MERGED","created":"2023-03-14 22:37:30.000000000","updated":"2023-03-16 13:06:12.000000000","submitted":"2023-03-16 13:05:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877429-tls-backend","meta_rev_id":"224cf7e23986b72d658e874100c856faf52843df","_number":877429,"virtual_id_number":877429,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-16 13:05:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-03-16 11:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-16 11:20:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-16 11:20:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-15 00:14:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-16 11:07:29.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-03-16 11:20:54.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"6ef4f846b5aac91ee2bdaeb99b3bca01dfc64891","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-03-14 22:37:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b95126da5ec157fbf61e8ff3b8cd572ebb342b2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 00:14:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/431699a97a624ba5951b1e941eec19a7\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8164646b08bb4acb9322015361964a82 : SUCCESS in 11m 27s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1a1e760521c4165ae2948ff277d581b : SUCCESS in 1h 25m 01s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1554c9b4f31e4d509fc585d26de1dd74 : SUCCESS in 1h 30m 04s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4dd90f05cc3b4d99be157fc88b8e69c6 : SUCCESS in 1h 03m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c660416d45ab4777b3a6c2babb96cf51 : SUCCESS in 1h 08m 12s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3935d54c2f9e4cf7a15565040343d30d : SUCCESS in 1h 14m 50s\n- openstack-ansible-upgrade-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/517484f323f14d4b8e5d2a865b439e79 : FAILURE in 44m 39s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bdf8eff37ad4a7e96c8abfa67830300 : SUCCESS in 1h 15m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73739136b87d4fad9876515e96bb1f93 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"27a4c366da352e4bba60e53cbf02db696651c7f3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-03-16 11:07:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"932d5edcf1083677140d80bb40994ad5b88d539c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-03-16 11:20:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"816f2abe6ae15186da714d0b0f26da6a3fb6985f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 11:23:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c0930a69b85f2d5c4e67959c214ffe2e029f110","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 13:05:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78fcbb051ea44054b5a818186e6b0071\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/934220d1fa5b4c72a4c0fcb2d8b6abe1 : SUCCESS in 12m 39s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e19032147174973960c7788663a2403 : SUCCESS in 1h 35m 31s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e5ecb0ff20914c659075bbafb617df3b : SUCCESS in 1h 28m 29s\n- openstack-ansible-deploy-infra_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a8dd3d814ec24e00af49261c9d09ff55 : SUCCESS in 1h 16m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31252d2e3dce457cbd1cad0ece16e86d : SUCCESS in 1h 16m 11s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43b66bfd48ee4812a97164857570abfd : SUCCESS in 1h 25m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd78a0186c574d34b3359ba8dcaaff14 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ad5a55d50a9d72f3a9753161de139fbe44971f9e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 13:05:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"224cf7e23986b72d658e874100c856faf52843df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 13:06:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea933515b23a420aaf0566707e94bbd3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08c8c32cc81b4f8383df0580e5bed887 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505","revisions":{"3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505":{"kind":"REWORK","_number":1,"created":"2023-03-14 22:37:30.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/29/877429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/29/877429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/877429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/877429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/877429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/29/877429/1"}}},"commit":{"parents":[{"commit":"b9d9ccfc8a781759f421d5c8aa1a9276295765b5","subject":"Update tox.ini to work with 4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b9d9ccfc8a781759f421d5c8aa1a9276295765b5"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-14 22:35:19.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2023-03-14 22:35:19.000000000","tz":60},"subject":"Turn off absolute_redirect for nginx","message":"Turn off absolute_redirect for nginx\n\nNginx adds trailing slashes to the URLs ending with directories.\nSo by default, when accessing http://172.29.236.101:8181/pools, nginx\nwill return 301 redirect to http://172.29.236.101:8181/pools/.\nIt\u0027s an absolute redirect which causes a problem when haproxy frontend\nlistens on HTTPS but its backends listen on HTTP.\nIn this case, when accessing https://172.29.236.101:8181/pools, nginx\nwill return 301 redirect to http://172.29.236.101:8181/pools/ (http)\nthat won\u0027t work.\n\nThis patch changes behavior by disabling absolute_redirects, so when\naccessing https://172.29.236.101:8181/pools, nginx will return a\nredirect to relative location \u0027/pools/\u0027 without changing protocol.\n\nChange-Id: I9e55508996d9b24437870f2f23dca5db7827fee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/3d3f6102458ca7d70ce9eb8dd15e6a49c7e73505"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
