)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~929690","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I6c7503794ec9b66ec54d8dd2ae81ad914aa3d389","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"osa/apache_mpm_alignment","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-09-25 08:51:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6c7503794ec9b66ec54d8dd2ae81ad914aa3d389","subject":"Ensure that selected Apache MPM is enforced","status":"MERGED","created":"2024-09-17 17:56:02.000000000","updated":"2024-09-25 08:52:30.000000000","submitted":"2024-09-25 08:51:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929690-osa/apache_mpm_alignment","meta_rev_id":"e193f1a74a775e65faa86fc4759c17c3e0eb3a90","_number":929690,"virtual_id_number":929690,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-25 08:51:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2024-09-25 07:09:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-23 10:42:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2024-09-25 07:09:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-17 19:37:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-23 10:42:02.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-09-25 07:09:17.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"4f602291981b84ed63ce009c70ac2a26e5cb5de9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-17 17:56:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"672264f3b3990c9ed72cb9062c904f4b95867a79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 19:37:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72e79020543a4ec7912bbe6c9dd9766d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50a1feb8f8d749a8a4520f5505c65fdd : SUCCESS in 7m 21s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cb2379f1bc74cbf9b4387cad256749c : FAILURE in 28m 21s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7ee49c46bb89483f8b1abea7f169dcd0 : FAILURE in 31m 51s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/070602f7747444f494934e59b8b3af00 : FAILURE in 30m 40s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/5eba68dc16ba4cbfba5938ee9130ae40 : SUCCESS in 1h 16m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/860b927e0959427dbced5b7953365196 : SUCCESS in 1h 07m 40s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c5178fc2c114e06b64b3642acdbea99 : SUCCESS in 50m 09s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a8db47e11294123bd1712038e9d3cfa : SUCCESS in 1h 07m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f0fd8dbf839469c8b773abf06eb1954 : SUCCESS in 1h 09m 42s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b48bf9217ee408892dc2b421e7fc7c3 : SUCCESS in 1h 13m 39s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/997e1331402b4fa1b290168bc7b6cd31 : SUCCESS in 1h 14m 00s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/deae951702554081b5ac862bb7105c65 : FAILURE in 1h 09m 49s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b60d3c0e4c24a7ab49d5fab99ca3530 : SUCCESS in 1h 37m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e66e7e770474311b0e212b9d811ed4e : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"baaed260e1a5108f495408f84fae2e8f74a39b4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-17 19:52:02.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6431660ffaf15caf80d9a080795c2d0222d4b357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 21:18:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/efa63119c7f0465aaeb4dc988e64f558\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bef76430ed3a4badbb24b63052fda443 : SUCCESS in 6m 02s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2c29b33b13745d78386ae100e661624 : FAILURE in 36m 31s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a77a0d479c114439b1805b7f705a5896 : FAILURE in 35m 58s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c93ca27e437e4e08b2b4f789b20e8d79 : FAILURE in 32m 41s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8b94ef1cea7e45eab8f17ae948bb330c : SUCCESS in 1h 12m 01s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f1f134b5e6f4770800644f5c32e7b2d : SUCCESS in 1h 03m 01s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f982c55775bd44a3a9abf8859865f558 : SUCCESS in 1h 07m 58s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c2af429c96945b68a868b6d744f7982 : SUCCESS in 1h 09m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29ad59d3c9e24404a602e73736c6f082 : SUCCESS in 52m 18s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8426f1903c2042a1a97b90950ea4d71b : SUCCESS in 1h 18m 29s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bb00b6e0476419a9fb16febd24f5afb : SUCCESS in 1h 12m 47s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/031b1484761f41619f8f463898c89122 : FAILURE in 1h 09m 55s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bca13a11081546fb93a4f5e5903a05f2 : SUCCESS in 1h 05m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e68d346285124cce86fbacc0c0d1ff1d : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"6284cf48b06d584ec06d863995ef5b79de2652e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-18 08:26:35.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c3e8fe0dee40211a6192f757bd5269eb081683e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 09:53:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ff90970be17416e8deec5ba7e78a156\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0cd7fb272c2491d978ff8d0c2a528be : SUCCESS in 7m 14s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d0c1a709f104842baea93a66f0f4f9f : FAILURE in 28m 06s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7ca4681639074aa186dd264cad29681e : FAILURE in 28m 53s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/06f055cfdd8b466989548514f75a8b64 : FAILURE in 32m 23s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/ae5529281f20443e98d4dd63f26f0b9f : SUCCESS in 1h 08m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/163d5a0a965c4809b1a09d4e3c7e4c0e : SUCCESS in 1h 12m 04s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/958778ae4bbc4295825e181726debc70 : SUCCESS in 1h 15m 01s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e13ad0dca0e4e47a199d49b06889b77 : SUCCESS in 1h 13m 18s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd63a7ba67194173b441108b912cf7b7 : SUCCESS in 1h 16m 22s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f5f34d7a12540a0891eab645840cf1f : SUCCESS in 1h 14m 35s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9225cb75209f4a68ab007e26dc686a7e : SUCCESS in 51m 19s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cf306b5c182d44b395312cf9a3e804a2 : FAILURE in 1h 23m 41s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77b003783b5c4024bcef6ccefbf91c06 : SUCCESS in 1h 05m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d948a2f641f94201a690438237503b5d : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":3},{"id":"13ab91ac22da2258ab4c2abe30017fe41e0e51c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-09-18 13:28:41.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9d98b42228a03e7f840a11e1d805cce35c6b63de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 15:30:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e013646f0a3d424e96eeecaca31309f5\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d57ca7365cec4bd189448402dabacaaf : SUCCESS in 5m 53s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fe5863f8a584f75abd292d9a1cd9dee : FAILURE in 1h 03m 16s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/38258fa3cef042d1857669e1edee8c19 : SUCCESS in 1h 30m 53s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/130a3acb5e6048e88a44ec337bb701ad : SUCCESS in 1h 37m 22s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4d89107e147945d5a74e97961ca646a4 : SUCCESS in 49m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2443f177c2f24e24827ccd736c1d8e95 : SUCCESS in 1h 28m 00s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/834e7aba566342edb10cf8aa2e2361b5 : SUCCESS in 1h 12m 02s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2693593c0d8c413fbc60fec8c40b7a1e : SUCCESS in 50m 45s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99dd3afa458e43ee8b4ca82a06e15383 : SUCCESS in 1h 34m 21s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2929b915fd04d6f8b88ba72ff89d432 : SUCCESS in 1h 21m 55s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/912d08dd6c264dc3b3c4f4148f4404bd : SUCCESS in 1h 18m 33s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a961479d4e074d259d36bdcf7b024875 : SUCCESS in 1h 36m 28s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2477cb5ac6244238bae173d294736f0 : SUCCESS in 1h 55m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c39428d58dba40349884faeac7f211e1 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":4},{"id":"a59ceb8c3477da30f2f676f79cabc5e124e312d6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-09-23 10:42:02.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2d6f8641662a1e900d539a34852b5b8bf4566c02","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-09-25 07:09:17.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0051a77dd149c44676cde1a43f5111892632d53d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 07:11:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a9e6a968bfb66de1d120aa4639165db82f02bb8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 08:51:33.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ace4c8a43b8f4bd5a040b85aea7ed745\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/546ef44f4bb94b8b89501f8087a1285f : SUCCESS in 6m 44s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/50daf5cfa7a94a8a919ce1f230197d3d : SUCCESS in 1h 26m 35s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6cffb9c0a5b941d3bf8a5789c28d4974 : SUCCESS in 1h 28m 38s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/94ea285c63d444b8b1a5bcbaaaf8162d : SUCCESS in 1h 14m 40s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25515abed70241a7a0ce7e9d3b3218a9 : SUCCESS in 1h 16m 04s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de828d42b7c54be0ae9038d9181c0df8 : SUCCESS in 1h 05m 01s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d758ba19ad364cee80c2c467ca9a599f : SUCCESS in 1h 20m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4410e74cd61d4d73bac75c5f8fef6e91 : SUCCESS in 1h 19m 17s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99c2ee1042824903b8c2e0731c4634a9 : SUCCESS in 1h 00m 15s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1537cce0a6c49b6a93544c04b69bf11 : SUCCESS in 1h 01m 27s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8dde15f3e76148af9914fa0250860aed : SUCCESS in 1h 26m 15s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/273aec6270604b2494da9e3e2243fe34 : SUCCESS in 1h 16m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0696d3ca2a8344098db38181529ed598 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":4},{"id":"66f6f72ff354a876bc2a7a174604e1bc3a5e0e03","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 08:51:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e193f1a74a775e65faa86fc4759c17c3e0eb3a90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 08:52:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c2f9aa77ccb47b8afc0dd9534d5cbaf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/505569b732f94c058c7490a286a08a5d : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b9c8ad83760a523febfbf72398dcea785ce0a1e9","revisions":{"75a12becf9c62ca9ea00f2c0498d74c8af45aff1":{"kind":"REWORK","_number":1,"created":"2024-09-17 17:56:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/929690/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/90/929690/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/929690/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/929690/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/929690/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/929690/1"}}},"commit":{"parents":[{"commit":"2a6f7579f864675eef24b9a92d9db0a7147c6ca3","subject":"Remove references to lsync and rsync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2a6f7579f864675eef24b9a92d9db0a7147c6ca3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 17:55:51.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 17:55:51.000000000","tz":120},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nChange-Id: I6c7503794ec9b66ec54d8dd2ae81ad914aa3d389\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/75a12becf9c62ca9ea00f2c0498d74c8af45aff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/75a12becf9c62ca9ea00f2c0498d74c8af45aff1"}]},"branch":"refs/heads/master"},"39c9fb00802866b0195c44e9528bc1d62b0e1b24":{"kind":"REWORK","_number":2,"created":"2024-09-17 19:52:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/929690/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/90/929690/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/929690/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/929690/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/929690/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/929690/2"}}},"commit":{"parents":[{"commit":"2a6f7579f864675eef24b9a92d9db0a7147c6ca3","subject":"Remove references to lsync and rsync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2a6f7579f864675eef24b9a92d9db0a7147c6ca3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 17:55:51.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 19:52:02.000000000","tz":0},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nChange-Id: I6c7503794ec9b66ec54d8dd2ae81ad914aa3d389\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/39c9fb00802866b0195c44e9528bc1d62b0e1b24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/39c9fb00802866b0195c44e9528bc1d62b0e1b24"}]},"branch":"refs/heads/master"},"28e2aef72f8928c90828aa6388d2db3acf7ad6c2":{"kind":"REWORK","_number":3,"created":"2024-09-18 08:26:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/929690/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/90/929690/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/929690/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/929690/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/929690/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/929690/3"}}},"commit":{"parents":[{"commit":"2a6f7579f864675eef24b9a92d9db0a7147c6ca3","subject":"Remove references to lsync and rsync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2a6f7579f864675eef24b9a92d9db0a7147c6ca3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 17:55:51.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-18 08:26:35.000000000","tz":0},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nChange-Id: I6c7503794ec9b66ec54d8dd2ae81ad914aa3d389\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/28e2aef72f8928c90828aa6388d2db3acf7ad6c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/28e2aef72f8928c90828aa6388d2db3acf7ad6c2"}]},"branch":"refs/heads/master"},"b9c8ad83760a523febfbf72398dcea785ce0a1e9":{"kind":"REWORK","_number":4,"created":"2024-09-18 13:28:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/90/929690/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/90/929690/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/929690/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/929690/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/929690/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/90/929690/4"}}},"commit":{"parents":[{"commit":"2a6f7579f864675eef24b9a92d9db0a7147c6ca3","subject":"Remove references to lsync and rsync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/2a6f7579f864675eef24b9a92d9db0a7147c6ca3"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-17 17:55:51.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-18 13:28:41.000000000","tz":0},"subject":"Ensure that selected Apache MPM is enforced","message":"Ensure that selected Apache MPM is enforced\n\nAt the moment services might have different MPM selected while all\noperating the same Apache setup, ie on metal setup.\nThis results in failures to set selected MPMs, so eventually second run\nof roles after initial deployment will end up in failure (ie upgrade).\n\nThis patch ensures that all except selected MPMs are disabled and do\nrole get\u0027s the desired state of deployment.\n\nWe also need to align selected MPM across all roles to avoid\nfuture conflicts.\n\nChange-Id: I6c7503794ec9b66ec54d8dd2ae81ad914aa3d389\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b9c8ad83760a523febfbf72398dcea785ce0a1e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b9c8ad83760a523febfbf72398dcea785ce0a1e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
