)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~938270","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~I8cf760fe59842746758f4f430038cfe6e64e936e","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"osa/httpd_role","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-01-21 16:55:19.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-01-21 16:55:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8cf760fe59842746758f4f430038cfe6e64e936e","subject":"Use standalone httpd role","status":"MERGED","created":"2024-12-26 16:45:18.000000000","updated":"2025-01-21 16:56:18.000000000","submitted":"2025-01-21 16:55:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"938270-osa/httpd_role","meta_rev_id":"d358004845fa5ea32940ee059008e6ff25b61460","_number":938270,"virtual_id_number":938270,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-21 16:55:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2025-01-21 14:47:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-20 21:42:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2025-01-21 14:47:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-26 17:29:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-15 16:54:17.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2025-01-21 14:47:36.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"3e93f6c557118c16b5cbb324e580b990d1929fc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-12-26 16:45:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2e72e482a796bb78cf067e90b019624fbe12116","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-12-26 16:45:47.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6add3b24cba6b13fb3a9c790a325c8dec463beee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-12-26 16:59:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"24dcab4471c7a279b7b74122eb7527b5c8ecd4f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-26 17:29:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2925bee53124806ad06ed44271ffeac\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4906d50386c84f6f838de84c0cf27905 : CANCELED\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad180a35152840a49277d2f9967a5a74 : CANCELED (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ea3696412fe842718fbe2ab3bab13482 : CANCELED\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/723738427aa94c46a5157d25b2095d27 : CANCELED\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/2a182658873d46c6babeaa7e9bb3a80a : CANCELED\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/382462e6a10c4429a4c11520c82f011d : CANCELED\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d3fa80e30db4d78a2ebfb6355b1783e : CANCELED\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55f9ccb889c342e983482317f9ae2d02 : CANCELED\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4baa55406eb64163906024cb6a5171b6 : FAILURE in 26m 08s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a19ec3b851d4f25b20e113383eb6c03 : CANCELED\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2368670b97e743fa960e41866f0f4b58 : CANCELED\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f1b18c680f024331948dee2f543eb840 : CANCELED\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a1debb3790948f7ae6a0d0896df9799 : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7dd785cfd8b4768a4784fb1433996c8 : CANCELED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39150b06f8b34cf9be5da8c5f845a8e3 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":3},{"id":"e757283fdaff30d51e413cf9462296d1a2f914f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-12-26 17:35:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2a3e8e170d2f3744d7d9f11e8021322c2a3cce38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-26 18:22:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4bc6d86e98eb414abc5cd32d28e12877\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1af1a66cc4ba45ac8e794aafd59b446c : SUCCESS in 5m 42s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff62e162f5cb4b8286ddff589ff49f4b : FAILURE in 42m 21s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/57d8302d0282462090a590e1c96d4c8f : CANCELED\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/18cfdbe43f694802957614aae8581b95 : FAILURE in 43m 20s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a9ee6588c5684c398b3ca9e944c27807 : CANCELED\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f2f8a733ce244c6a7fd2570230fa3fe : CANCELED\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93702226e6574a1397d0df34e5694a30 : CANCELED\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1dafdadd81f64a96832ab3c3408921a4 : CANCELED\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ad21db9a1f247cc9c32d4c1f20ef1f5 : CANCELED\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43e29d7bfe804467988126cc6d32403b : CANCELED\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe8cf58aa9a241fd9e4c28fc55f5517c : CANCELED\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/36f02f5e23a1490b89bfc3f06d189682 : CANCELED\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d44a955e7cc543f790e005695b081245 : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f69bbea856974781a8904a7fb377b23d : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d5591644bc3477d9dcb522933b12535 : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":4},{"id":"40f6b21060dd203121fa05adeb4352ab09131ae3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-12-26 18:52:30.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"42952cc56d4f4ebd9dcd1ae8491bca8d7ab98238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-26 19:41:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e27179b0c3554fae9f7f00acf4755191\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1109c7cab5f24befa0ab92925dc00306 : SUCCESS in 7m 14s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc4c556a031944418523ca16bbd54033 : CANCELED (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2b77d5800dff43ea9c95bcaba59a5f8b : CANCELED\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/40d8231a246e41c397abc46c0db259f7 : CANCELED\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e6b356d852bd4a97930565b56669c0da : CANCELED\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7f5984a59a04cdf9a639759dfaecd17 : CANCELED\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/539325c4f9d34c8b9be1427123773d8e : CANCELED\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f99b347ca9b647669e75d377213537a2 : CANCELED\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8526d5a582cb495cbcd4508d5348de96 : CANCELED\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd76ee2d249e4d6497aeab349c00f176 : CANCELED\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8b78a0161744084a2923f5fb9782fcc : CANCELED\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/931e6c653fea4afa9edc322e673b887e : CANCELED\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97b62cf469974f53af94db7c0736435c : FAILURE in 45m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c218c326b564a3d9cf0d5ba2cfc1442 : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/933b951bcac2426fbd85f8e68fb5c447 : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":5},{"id":"199c2eab4470e2c7f8851eed10ba941394896196","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-12-26 20:36:26.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"1c5332e4579a2301c51ddc475b9570d27102ce57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-26 21:32:39.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/022d4e7b1397499499a7d8d7799cd8f8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb03e286b9eb4477b639068a3ce0b509 : SUCCESS in 8m 03s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f17e1d30cb7d46738db0964157fcf96f : CANCELED (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/58c3ace8b8094548888afac1d36f5003 : CANCELED\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e8228a9c757143d68cccc1e776660b72 : CANCELED\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/687c62f2a617497992541812ea881f53 : CANCELED\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2e51fcfd11f404d94a5ca185a9bab6b : CANCELED\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7917a9a17d8e4dc9a73d435eab1b1dd5 : CANCELED\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd33c8af723149279e67f0b32f0c456f : CANCELED\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76c09911f5484e4f98d9676c952ae816 : CANCELED\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ec377f2989a4217a4514c6ff1eae03c : CANCELED\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4e597a35a7d4f06ba29b356381bdae9 : CANCELED\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e9eb54512e8e46109af1a4855576a6c9 : CANCELED\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17a1e7f4c1c34644a59889b927ca9631 : FAILURE in 52m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79d54f0c813a4364a92b752cf82c5257 : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5ef3aeaf4f14a33b063b65ebca0155d : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":6},{"id":"c98dfd58bc05cbb85acbafacf8949b5d5a541d2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-12-26 23:48:17.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"0281e4fc2fb8fd0f7348338b23db9ce866438b6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-12-26 23:50:09.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"14f0806c2e66d249d66573382e9f7ac44b36a461","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 01:48:59.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f56321cffac44e192506bc80afc97dd\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2d8dc3ed0a847fb83b5f67d4b187658 : SUCCESS in 8m 00s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/cebe651a4ff0446c819c81f9bd0272da : SUCCESS in 1h 02m 08s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eaae8bd6b19a48769b6bb72b243b84c9 : SUCCESS in 1h 05m 52s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cc0a34bf83384aaab0de6284fba5c655 : FAILURE in 51m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/60e9d128b73a459980982888c8f6ec80 : SUCCESS in 1h 10m 49s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbb7a1f31ada4736be084cb6738b4d23 : SUCCESS in 51m 33s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8880be7b31a148faa42f67ac6fd7ba75 : SUCCESS in 1h 12m 37s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8f05a5f0bbf404ba0ef52a7bfb8a36b : FAILURE in 47m 35s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08da37eb6eec48658ab0f3a492020f0f : SUCCESS in 1h 05m 58s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4962088b0a80436cb51efccdf064f407 : SUCCESS in 1h 04m 21s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43ea31db99c644b9b041cf67d9c6faff : FAILURE in 49m 09s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a51a8e5f55ab48cc92ce1ae8afd7a2a5 : SUCCESS in 1h 28m 16s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/083cd784a0c74868aef4288789c5b825 : SUCCESS in 1h 41m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eeab68139134af197d4ce2c6e87f9ba : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b27616f033d144f8beaacbb4475434eb : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":8},{"id":"346a549c96dc087b290a2f882f15b070d5d74e1a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-12-27 12:37:54.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8f37f15a0b71fa026ee1e0fbf8d47152ebbc0eb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 12:38:49.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/938275 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"6a2c34fcc347aad3937ca968943928121f72bed8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-12-27 12:39:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7a1183cd7b65d590caec88055afff1a9a666923d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 14:50:10.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38488fcd219e489faeb3fe3dc79549a8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bc313bbd19e434aadb6df3791540432 : SUCCESS in 8m 31s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/db7f7a7b427149a09e37ebec70ea39a6 : SUCCESS in 1h 21m 08s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/efc88908e6164b9cbc71f4f34024f5b7 : SUCCESS in 1h 26m 28s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/830500052226481b8042b61c78a14eb4 : SUCCESS in 1h 19m 17s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/a5d078273a44474690d2acac47201f6f : SUCCESS in 1h 13m 15s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67a2fde30960411fb3cab20f505f9ed9 : SUCCESS in 1h 10m 33s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1420a2bfb8b4cc1b5fbc3d1417053a5 : SUCCESS in 1h 07m 34s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e864c6931fdb4057b70d4e9d9f3b2b9d : SUCCESS in 1h 10m 08s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e62c3ecd89f74689948f2aaa0f1cc7b9 : SUCCESS in 1h 16m 27s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0be32288a04a41ed986ad6a7f96c097e : SUCCESS in 1h 14m 29s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1ce345e78b240438849ed4f5ec682b9 : SUCCESS in 1h 14m 05s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a134f63828b9425b8b2fc98de5f268cd : SUCCESS in 2h 02m 49s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5de59c84ef44bcfa5f637eb604bde3c : SUCCESS in 1h 28m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a210d787c1ec47478a39f6d810d437d5 : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ae642778f8146dbac896811cac2d8be : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":8},{"id":"074af1eeb9270342090a5db92da3a05fe7453837","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-01-06 09:47:59.000000000","message":"Patch Set 8: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8b2667e4bf0df7acb9ab69eecc79dde16a89892d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-01-06 09:48:17.000000000","message":"Uploaded patch set 9.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"6509a1a136437c8076417264ecc416eba3f0102a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-01-06 09:52:24.000000000","message":"Patch Set 9: -Backport-Candidate","accounts_in_message":[],"_revision_number":9},{"id":"8ae3259704db86244166f2a9ea96bda0b075f37d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 11:24:31.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2842f1471444d8ea233959830a33681\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/718c5b3ffff64f0daea6854c2b9ee749 : SUCCESS in 6m 53s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/27083eddec4e44aba823267c0ac2c87c : SUCCESS in 1h 16m 19s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d2d645713ce349ceae978b960aa770ea : SUCCESS in 1h 26m 49s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/db0a02ae21ec477d96f9156ef50e18e4 : SUCCESS in 1h 26m 08s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/52b71043e88647ceb7568f6921971861 : SUCCESS in 1h 05m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d060d658b70841b3b97bf09078b9624c : SUCCESS in 1h 10m 39s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae33b70486244e13a1a86405ae2c4cc5 : SUCCESS in 1h 15m 38s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44d78b034f1c4a2c95ca71c4413d157f : SUCCESS in 1h 03m 39s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/553986666e1d4568b9f805657741c874 : SUCCESS in 1h 18m 06s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67d1d448642e4acc985a06c7f59e501c : SUCCESS in 1h 13m 01s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50ecb6310b6d4f978cb0ab2e883f3804 : SUCCESS in 1h 23m 01s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/823e4f21a47248e58ebf25b9900609a8 : SUCCESS in 1h 06m 45s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4492e87be68c4e3baadbfb8e0f672c32 : SUCCESS in 1h 15m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdc4ce2540ec44c8a83b6bb2ba220e33 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/553541ab2d564e2fa9f8247415dd4a09 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":9},{"id":"f7f063171238c45a2844377b2332c277c58cbb21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-01-06 12:32:23.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"184509137c3af014dbc863b020d0982538212886","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 14:39:20.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e1d2c6eb8314b5885491ec8eed5952b\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba6bc541ff9f419382c6d120babf46e6 : SUCCESS in 6m 28s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/3daa246a7dcb4e2bbbd5f2c371060ac6 : SUCCESS in 1h 12m 10s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cf4f51f4e9e448d78c6783258ea98d43 : SUCCESS in 1h 14m 50s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c5de89e08f204d7d8602c7eca3240ee7 : SUCCESS in 1h 21m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/81bb5341fddf4305b1a35fbf098ab0d3 : SUCCESS in 1h 08m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e82a238db6348fe9a3493d57931de71 : SUCCESS in 1h 12m 25s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3b7716022314854afac4481ff481b96 : SUCCESS in 1h 08m 59s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc98a45553824439ad0bd206776be864 : SUCCESS in 43m 48s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cef448cb20a403b8f8012d41a8ee368 : SUCCESS in 1h 14m 29s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b74908f50af144199418733f2fdc85ae : SUCCESS in 1h 17m 11s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0f50e64bee6430096635e7b3a191a85 : SUCCESS in 45m 33s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8b81257ba9e940ee967063f008d99c46 : SUCCESS in 1h 40m 21s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5f12ea6381d04a11a5523513fd133ed3 : SUCCESS in 1h 57m 26s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2edde8f6e8e64cfb8cfbc559913e4b02 : SUCCESS in 53m 05s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afcc5aa586804992a39b4ab8429fbd87 : SUCCESS in 58m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38deff0fca2e4be6bfcdb34546728c04 : SUCCESS in 3m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b27bbd8022ee4fb0bab7ad6417d0065b : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":10},{"id":"3c1abf19366d27cd4191912debe411924e27da0d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-01-15 16:54:17.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"d2a2a70b93c209544e5a9bbae409fae56ef04bb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-01-20 10:39:40.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"18dcc697870807f8ab32cb31f576e649abb07f96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 12:49:02.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/632751e51264477ca33a08eed7149ce8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1190c90db72a4bc8b38735b429798c1b : SUCCESS in 7m 14s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/94484be4568849fa94ae36bff9688f70 : SUCCESS in 1h 19m 31s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dea78bd936e44d3f9179576ff9572f64 : SUCCESS in 1h 22m 58s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/635ed34e95a3492db13654b3bfbad756 : SUCCESS in 1h 36m 09s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/6bc9c28e6b1b4f1387cb061b1bce2c8a : SUCCESS in 1h 15m 22s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb0a909196104f1c865a2f62f7f7d506 : SUCCESS in 1h 37m 37s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de8542dea74e47ceb61100280ae3f274 : SUCCESS in 1h 33m 40s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a98ba1c81c364c86b29a083d9826a82b : SUCCESS in 1h 40m 51s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55091411b65d4d1a9db9b5ecc0d43cb6 : SUCCESS in 1h 45m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1e3ec2728da40e38167d72cd8684e6e : SUCCESS in 1h 32m 27s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13c781137a7c42cba43ff0c28d01b5f1 : SUCCESS in 1h 44m 19s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/594b253e903246d2b549c0cc1a7ff8a2 : SUCCESS in 1h 29m 56s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cc7f53bcfb044ca8b13616ca2abb212a : SUCCESS in 1h 51m 48s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/630f53a80ac54c018ebf482946772981 : SUCCESS in 2h 00m 57s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c16990c469e47a5b276d20c205d223f : FAILURE in 1h 26m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e95c3ad553e840659e675aff177f6d37 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9178463da9534134a084bc8d7d7c1620 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":11},{"id":"a7a3d3ad2fa8a5342bde2de1a8bf51683decfebe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-01-20 19:02:59.000000000","message":"Patch Set 12: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"31cd2fdb650818357e6da2e53ada93542123a297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-20 21:02:28.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3923cd3f2ac4c15a6d1f1bb16a61420\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aff2ed2b8a684e7092fcfdaf61a5bd1f : SUCCESS in 6m 35s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5860efc20fb448fbab22cf6b6205ba4d : SUCCESS in 1h 23m 42s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/842b4e0294024b3ea2eb2bdfd0d9bdc5 : SUCCESS in 1h 29m 08s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/96004d14dbfd42e1b9e7a46547934d85 : SUCCESS in 1h 31m 48s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/22b7781d7e30427a9ed5abc1fa6d1f56 : SUCCESS in 1h 13m 26s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06dafe661c7b479f94a004e55ca37c49 : SUCCESS in 1h 19m 51s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1cd4d0cd9e64bd6be1c0e4b4c42e2e1 : SUCCESS in 1h 11m 40s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1032e475fb144548f1f5157ad26e1a2 : SUCCESS in 1h 12m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da10bf9081b1449bb2b12efa196e16d6 : SUCCESS in 1h 17m 01s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1259446f0af40baba28966e92a6a03a : SUCCESS in 1h 13m 30s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9dade832e6b48ac9c5c91fa5c7b5cbe : SUCCESS in 1h 14m 09s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b0dc817a4ce84a67bb32352d09829b82 : SUCCESS in 1h 44m 50s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1ba7bda5ad61485da0d19a79be75aa04 : SUCCESS in 1h 52m 13s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/540fa81b159442e19081ab83097e2bdf : SUCCESS in 1h 42m 40s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46e30493facd4791ba5c061d553879a3 : SUCCESS in 1h 37m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bff2ca03f85488098f716e8f82e941d : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b03125ecd5fe4a4d91a361a24a488a4e : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":12},{"id":"3ecf2a12d37a0d5ef5ad41413baa36ebc6f0036e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-01-20 21:42:19.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"c6eb55f023387f7808a8dd35b58be9926fb9b19f","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2025-01-21 14:47:36.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"8eb052396dd832e0ce0f3b391ab2c8a2c617cd31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 14:50:32.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"fbda2410f0cd1cd558f19eaf41b4b0927157b0a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 16:55:19.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05b75b0152814d02a99fd252f46c1d8c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19a4fc1fb5d24ade91a1e615c36ae957 : SUCCESS in 6m 27s\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5df527650174476b81fd7db4571f87d2 : SUCCESS in 1h 33m 24s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/59864a7b198946bbb8a4c0a8c85403e3 : SUCCESS in 1h 25m 35s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1b7e493c03694014a36c5623b0567b3c : SUCCESS in 1h 15m 46s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0488d0d7c6f4468bf64fe93abfbea10 : SUCCESS in 1h 23m 59s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ee11581cac54a23bb46339e5712c889 : SUCCESS in 1h 19m 28s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b0717fc8477477ba6a765127f2fc9ca : SUCCESS in 1h 15m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/757ce7f7148040e28f1a092e975310e4 : SUCCESS in 1h 20m 38s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb68b98f7139415c849ddfdda2f0a412 : SUCCESS in 1h 20m 56s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb6d642c5ab44527a5f5b187fc1e2d58 : SUCCESS in 1h 17m 20s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e185ad1f8b994f49b56176c9fe0ce806 : SUCCESS in 1h 59m 01s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/094f81c5058348488efb48ec2bca6cd9 : SUCCESS in 1h 50m 34s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/083e3d0ee3ed4fcd953dce514f5f0487 : SUCCESS in 1h 35m 50s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67765d3f6edf40c3a6c6777c380ee08b : SUCCESS in 1h 41m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4c040a7fd2f46b2b4b40c3d3a4b30da : SUCCESS in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57e7f1c644b7414995e0106c18928fd9 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":12},{"id":"379e7989c5baf0b053e027f676e1a2fe17b296d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 16:55:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"d358004845fa5ea32940ee059008e6ff25b61460","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-21 16:56:18.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aff4967ee322445a89b73cf63ab0ff60\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1e8c93c188e412abff0795e441af335 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"73dcf7c341e4cda17da9392f11e3cbf2b355599d","revisions":{"a6007239b503382f62aa67c1f2811fc98e309a58":{"kind":"REWORK","_number":1,"created":"2024-12-26 16:45:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/938270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/70/938270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/1"}}},"commit":{"parents":[{"commit":"4de71336b66880b99b31ec812b4eba37a121d136","subject":"Fix tags usage for the role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/4de71336b66880b99b31ec812b4eba37a121d136"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 16:45:10.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 16:45:10.000000000","tz":60},"subject":"Use standalone httpd role","message":"Use standalone httpd role\n\nIn order to unify approach for managing Apache Web server, we migrate\nto usage of standalone `httpd` role instead of managing apache\nseparately inside service roles.\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-httpd/+/938245\nChange-Id: I8cf760fe59842746758f4f430038cfe6e64e936e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/a6007239b503382f62aa67c1f2811fc98e309a58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/a6007239b503382f62aa67c1f2811fc98e309a58"}]},"branch":"refs/heads/master"},"fb7b11a8e5a14a1f49419c1b0dc14c1ab4d72b15":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-12-26 16:45:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/938270/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/70/938270/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/2"}}},"commit":{"parents":[{"commit":"4de71336b66880b99b31ec812b4eba37a121d136","subject":"Fix tags usage for the role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/4de71336b66880b99b31ec812b4eba37a121d136"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 16:45:10.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 16:45:47.000000000","tz":0},"subject":"Use standalone httpd role","message":"Use standalone httpd role\n\nIn order to unify approach for managing Apache Web server, we migrate\nto usage of standalone `httpd` role instead of managing apache\nseparately inside service roles.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/938271\nChange-Id: I8cf760fe59842746758f4f430038cfe6e64e936e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/fb7b11a8e5a14a1f49419c1b0dc14c1ab4d72b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/fb7b11a8e5a14a1f49419c1b0dc14c1ab4d72b15"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b7f2487dbced6a6324b1177a2c476641845f525a":{"kind":"REWORK","_number":3,"created":"2024-12-26 16:59:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/938270/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/70/938270/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/3"}}},"commit":{"parents":[{"commit":"4de71336b66880b99b31ec812b4eba37a121d136","subject":"Fix tags usage for the role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/4de71336b66880b99b31ec812b4eba37a121d136"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 16:45:10.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 16:59:23.000000000","tz":60},"subject":"Use standalone httpd role","message":"Use standalone httpd role\n\nIn order to unify approach for managing Apache Web server, we migrate\nto usage of standalone `httpd` role instead of managing apache\nseparately inside service roles.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/938271\nChange-Id: I8cf760fe59842746758f4f430038cfe6e64e936e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b7f2487dbced6a6324b1177a2c476641845f525a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b7f2487dbced6a6324b1177a2c476641845f525a"}]},"branch":"refs/heads/master"},"33e972790c1e2d149e5b18150680d812cdddf926":{"kind":"REWORK","_number":4,"created":"2024-12-26 17:35:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/938270/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/70/938270/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/4"}}},"commit":{"parents":[{"commit":"4de71336b66880b99b31ec812b4eba37a121d136","subject":"Fix tags usage for the role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/4de71336b66880b99b31ec812b4eba37a121d136"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 16:45:10.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 17:35:29.000000000","tz":60},"subject":"Use standalone httpd role","message":"Use standalone httpd role\n\nIn order to unify approach for managing Apache Web server, we migrate\nto usage of standalone `httpd` role instead of managing apache\nseparately inside service roles.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/938271\nChange-Id: I8cf760fe59842746758f4f430038cfe6e64e936e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/33e972790c1e2d149e5b18150680d812cdddf926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/33e972790c1e2d149e5b18150680d812cdddf926"}]},"branch":"refs/heads/master"},"59b03ef75b00f9b8468d0d5202004fde004a9dea":{"kind":"REWORK","_number":5,"created":"2024-12-26 18:52:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/938270/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/70/938270/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/5"}}},"commit":{"parents":[{"commit":"4de71336b66880b99b31ec812b4eba37a121d136","subject":"Fix tags usage for the role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/4de71336b66880b99b31ec812b4eba37a121d136"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 16:45:10.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 18:52:30.000000000","tz":0},"subject":"Use standalone httpd role","message":"Use standalone httpd role\n\nIn order to unify approach for managing Apache Web server, we migrate\nto usage of standalone `httpd` role instead of managing apache\nseparately inside service roles.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/938271\nChange-Id: I8cf760fe59842746758f4f430038cfe6e64e936e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/59b03ef75b00f9b8468d0d5202004fde004a9dea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/59b03ef75b00f9b8468d0d5202004fde004a9dea"}]},"branch":"refs/heads/master"},"1a6271ffa8734a268b80a211fa15e985fcbeeea7":{"kind":"REWORK","_number":6,"created":"2024-12-26 20:36:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/938270/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/70/938270/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/6"}}},"commit":{"parents":[{"commit":"4de71336b66880b99b31ec812b4eba37a121d136","subject":"Fix tags usage for the role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/4de71336b66880b99b31ec812b4eba37a121d136"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 16:45:10.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 20:36:09.000000000","tz":60},"subject":"Use standalone httpd role","message":"Use standalone httpd role\n\nIn order to unify approach for managing Apache Web server, we migrate\nto usage of standalone `httpd` role instead of managing apache\nseparately inside service roles.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/938275\nChange-Id: I8cf760fe59842746758f4f430038cfe6e64e936e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/1a6271ffa8734a268b80a211fa15e985fcbeeea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/1a6271ffa8734a268b80a211fa15e985fcbeeea7"}]},"branch":"refs/heads/master"},"94a31d00fef56b78f30fd8f623c19c56ad47c0f4":{"kind":"REWORK","_number":7,"created":"2024-12-26 23:48:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/938270/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/70/938270/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/7"}}},"commit":{"parents":[{"commit":"4de71336b66880b99b31ec812b4eba37a121d136","subject":"Fix tags usage for the role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/4de71336b66880b99b31ec812b4eba37a121d136"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 16:45:10.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 23:48:07.000000000","tz":60},"subject":"Use standalone httpd role","message":"Use standalone httpd role\n\nIn order to unify approach for managing Apache Web server, we migrate\nto usage of standalone `httpd` role instead of managing apache\nseparately inside service roles.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/938275\nChange-Id: I8cf760fe59842746758f4f430038cfe6e64e936e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/94a31d00fef56b78f30fd8f623c19c56ad47c0f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/94a31d00fef56b78f30fd8f623c19c56ad47c0f4"}]},"branch":"refs/heads/master"},"a839e8dd73f899c57feacdbbc5b179b4ca976455":{"kind":"REWORK","_number":8,"created":"2024-12-26 23:50:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/938270/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/70/938270/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/8"}}},"commit":{"parents":[{"commit":"4de71336b66880b99b31ec812b4eba37a121d136","subject":"Fix tags usage for the role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/4de71336b66880b99b31ec812b4eba37a121d136"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 16:45:10.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 23:50:09.000000000","tz":0},"subject":"Use standalone httpd role","message":"Use standalone httpd role\n\nIn order to unify approach for managing Apache Web server, we migrate\nto usage of standalone `httpd` role instead of managing apache\nseparately inside service roles.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/938275\nChange-Id: I8cf760fe59842746758f4f430038cfe6e64e936e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/a839e8dd73f899c57feacdbbc5b179b4ca976455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/a839e8dd73f899c57feacdbbc5b179b4ca976455"}]},"branch":"refs/heads/master"},"e8ac1526a88fdbb1a3870aaa79dc9693439a8a7d":{"kind":"REWORK","_number":9,"created":"2025-01-06 09:48:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/938270/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/70/938270/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/9"}}},"commit":{"parents":[{"commit":"6dd3dfed09bd10420f356d5860a2d2e04002fa6a","subject":"Fix tags usage for the role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/6dd3dfed09bd10420f356d5860a2d2e04002fa6a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 16:45:10.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-01-06 09:48:07.000000000","tz":60},"subject":"Use standalone httpd role","message":"Use standalone httpd role\n\nIn order to unify approach for managing Apache Web server, we migrate\nto usage of standalone `httpd` role instead of managing apache\nseparately inside service roles.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/938275\nChange-Id: I8cf760fe59842746758f4f430038cfe6e64e936e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/e8ac1526a88fdbb1a3870aaa79dc9693439a8a7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/e8ac1526a88fdbb1a3870aaa79dc9693439a8a7d"}]},"branch":"refs/heads/master"},"da45ae8c5b75dcd9498560e2558464b8caa552f2":{"kind":"REWORK","_number":10,"created":"2025-01-06 12:32:23.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/938270/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/70/938270/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/10"}}},"commit":{"parents":[{"commit":"6dd3dfed09bd10420f356d5860a2d2e04002fa6a","subject":"Fix tags usage for the role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/6dd3dfed09bd10420f356d5860a2d2e04002fa6a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 16:45:10.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-01-06 12:32:23.000000000","tz":0},"subject":"Use standalone httpd role","message":"Use standalone httpd role\n\nIn order to unify approach for managing Apache Web server, we migrate\nto usage of standalone `httpd` role instead of managing apache\nseparately inside service roles.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/938275\nChange-Id: I8cf760fe59842746758f4f430038cfe6e64e936e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/da45ae8c5b75dcd9498560e2558464b8caa552f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/da45ae8c5b75dcd9498560e2558464b8caa552f2"}]},"branch":"refs/heads/master"},"ac693b14282699ddbca428c6acff74f275247b8c":{"kind":"REWORK","_number":11,"created":"2025-01-20 10:39:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/938270/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/70/938270/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/11"}}},"commit":{"parents":[{"commit":"6dd3dfed09bd10420f356d5860a2d2e04002fa6a","subject":"Fix tags usage for the role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/6dd3dfed09bd10420f356d5860a2d2e04002fa6a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 16:45:10.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-01-20 10:39:40.000000000","tz":0},"subject":"Use standalone httpd role","message":"Use standalone httpd role\n\nIn order to unify approach for managing Apache Web server, we migrate\nto usage of standalone `httpd` role instead of managing apache\nseparately inside service roles.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/938275\nChange-Id: I8cf760fe59842746758f4f430038cfe6e64e936e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/ac693b14282699ddbca428c6acff74f275247b8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/ac693b14282699ddbca428c6acff74f275247b8c"}]},"branch":"refs/heads/master"},"73dcf7c341e4cda17da9392f11e3cbf2b355599d":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2025-01-20 19:02:59.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/70/938270/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/70/938270/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/70/938270/12"}}},"commit":{"parents":[{"commit":"6dd3dfed09bd10420f356d5860a2d2e04002fa6a","subject":"Fix tags usage for the role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/6dd3dfed09bd10420f356d5860a2d2e04002fa6a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-12-26 16:45:10.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-01-20 19:02:59.000000000","tz":0},"subject":"Use standalone httpd role","message":"Use standalone httpd role\n\nIn order to unify approach for managing Apache Web server, we migrate\nto usage of standalone `httpd` role instead of managing apache\nseparately inside service roles.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/938275\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_networkd/+/939640\nChange-Id: I8cf760fe59842746758f4f430038cfe6e64e936e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/73dcf7c341e4cda17da9392f11e3cbf2b355599d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/73dcf7c341e4cda17da9392f11e3cbf2b355599d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
