)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~973136","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~If6019cceef3799183548a9488293c608efb57631","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"osa/bindep","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-04-22 11:28:34.000000000","reason":"Change was abandoned"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-04-15 13:33:05.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-04-22 11:28:34.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If6019cceef3799183548a9488293c608efb57631","subject":"Update outdate information in bindep.txt","status":"ABANDONED","created":"2026-01-12 17:10:46.000000000","updated":"2026-04-22 11:28:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1c020770952e9d6945cd2608112a9c5c3c5848dc","_number":973136,"virtual_id_number":973136,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-15 12:44:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-12 19:35:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-15 20:56:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-04-15 20:56:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"0c32f35a0839e78b8f18eaf83d7101c595395843","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 17:10:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af86c7a1da1694a21332a0fb28041339d26a088f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 17:12:29.000000000","message":"Topic set to osa/bindep","accounts_in_message":[],"_revision_number":1},{"id":"7927a27bd5210a7ee2ecee3c5ae06f9ad5140562","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 17:12:54.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"df4351a913adbc1dbc80e9e47d96f594c11e0581","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 17:12:58.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"0f0f220ce98007fca6e66576a61779bbe9cda38f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 19:35:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/052c1023efe9444681438f089b9bf65f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e305d74b95eb4441b06ba8a0b380cb42 : SUCCESS in 12m 13s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e346ded5fbc646df8bd944de06729dc1 : FAILURE in 22m 30s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c3abb0df1b6a4ea9b551588470556945 : FAILURE in 1h 22m 37s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/b2df3a715f6646d0b679678fe35c5f69 : SUCCESS in 1h 24m 47s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9c2fd2c68a244c04be5cec85690ea8cf : FAILURE in 34m 08s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/45ce3d58753340048d088bef338fe4f5 : FAILURE in 54m 17s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3c3ce7772f54ecd94a48ec314db8290 : SUCCESS in 1h 24m 19s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2973bc7c139145f8aa5dfeeda911ad0a : SUCCESS in 1h 05m 14s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/486606fbe84141a887d9e8d53fb7222b : SUCCESS in 1h 00m 25s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b0a967d618640edac48cbd0af63ec01 : SUCCESS in 1h 52m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f578d7c9b3fc47b584568c9e45cc33ed : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ef68d636c3541cb88dcb89b7eb398f3 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"6b0270f0f08daa782c44b87d79e0654704660f32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-15 10:26:23.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"8089d25347f38e1551b510309868406f3c844eaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 12:44:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5488cf43be4b418faf2eafbd738d2391\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/881cd94607e9482ab87228338c0f03c7 : SUCCESS in 8m 45s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e25298e548ea46ab87635d4c94b6f339 : SUCCESS in 1h 27m 26s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/3fbb0bced40c44c79b0580ebdfbb5c4b : SUCCESS in 1h 18m 29s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/0212730bdef64f6c93e28ab4c9af9461 : SUCCESS in 1h 19m 20s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1604f62dd29c4e3a9115e0088e9f7208 : SUCCESS in 1h 33m 40s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0d37b3f750f94252b6f32ee692825567 : SUCCESS in 1h 21m 16s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f5249332c724136a784a7af248cc06c : SUCCESS in 1h 22m 23s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/697d091c1b38410694f2d3e46182df96 : SUCCESS in 47m 12s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/085a9d3a97f048baa2eefde7a767e5e7 : SUCCESS in 1h 31m 45s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16625b3aefb744b9937d51f53e371267 : SUCCESS in 2h 08m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2118539f1aab42c085c97c700576e894 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/259cd31f93354e0aab5cf4b332a3412c : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"1c020770952e9d6945cd2608112a9c5c3c5848dc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-22 11:28:34.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d0b44a43a29d36468769b06cf8021f3f72958d5a","revisions":{"91e6c24d3eabae90d34f4e03493936357d43a4a3":{"kind":"REWORK","_number":1,"created":"2026-01-12 17:10:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/973136/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/36/973136/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/973136/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/973136/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/973136/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/973136/1"}}},"commit":{"parents":[{"commit":"52329489bd8a95ffb2413d6070962848c37ed31c","subject":"Merge \"Remove redundant vars line\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/52329489bd8a95ffb2413d6070962848c37ed31c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 17:10:46.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 17:10:46.000000000","tz":0},"subject":"Update outdate information in bindep.txt","message":"Update outdate information in bindep.txt\n\nChange-Id: If6019cceef3799183548a9488293c608efb57631\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/91e6c24d3eabae90d34f4e03493936357d43a4a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/91e6c24d3eabae90d34f4e03493936357d43a4a3"}]},"branch":"refs/heads/master"},"82ed2ceed5a6d5d1e0d570e70060912535c686ef":{"kind":"REWORK","_number":2,"created":"2026-01-12 17:12:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/973136/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/36/973136/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/973136/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/973136/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/973136/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/973136/2"}}},"commit":{"parents":[{"commit":"52329489bd8a95ffb2413d6070962848c37ed31c","subject":"Merge \"Remove redundant vars line\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/52329489bd8a95ffb2413d6070962848c37ed31c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 17:10:46.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 17:12:54.000000000","tz":0},"subject":"Update outdate information in bindep.txt","message":"Update outdate information in bindep.txt\n\nChange-Id: If6019cceef3799183548a9488293c608efb57631\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/82ed2ceed5a6d5d1e0d570e70060912535c686ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/82ed2ceed5a6d5d1e0d570e70060912535c686ef"}]},"branch":"refs/heads/master"},"d0b44a43a29d36468769b06cf8021f3f72958d5a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-15 10:26:23.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/973136/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/36/973136/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/973136/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/973136/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/973136/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/36/973136/3"}}},"commit":{"parents":[{"commit":"5539e20b0d05a0dd19d7b503091fe1b3c527b0c7","subject":"Remove unused Makefile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/5539e20b0d05a0dd19d7b503091fe1b3c527b0c7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 17:10:46.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-15 10:26:23.000000000","tz":0},"subject":"Update outdate information in bindep.txt","message":"Update outdate information in bindep.txt\n\nChange-Id: If6019cceef3799183548a9488293c608efb57631\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d0b44a43a29d36468769b06cf8021f3f72958d5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d0b44a43a29d36468769b06cf8021f3f72958d5a"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"82ed2ceed5a6d5d1e0d570e70060912535c686ef","theirs":"5539e20b0d05a0dd19d7b503091fe1b3c527b0c7","contains_conflicts":false}}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
