)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~979731","triplet_id":"openstack%2Fopenstack-ansible-repo_server~stable%2F2025.1~Ic2f42506df66e50bd754ba1ae3b78862d6c1882a","project":"openstack/openstack-ansible-repo_server","branch":"stable/2025.1","topic":"osa/pin-setuptools-2025.1","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-03-10 15:59:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic2f42506df66e50bd754ba1ae3b78862d6c1882a","subject":"Pin setuptools via global-requirement-pins.txt","status":"MERGED","created":"2026-03-09 21:10:28.000000000","updated":"2026-03-10 16:01:44.000000000","submitted":"2026-03-10 15:59:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979731-osa/pin-setuptools-2025.1","meta_rev_id":"23ce2043cf4ea695852e3fe9bc242e786e9799c0","_number":979731,"virtual_id_number":979731,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-10 15:59:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2026-03-10 15:35:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-10 10:32:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-03-10 15:35:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 23:31:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-10 10:32:09.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-03-10 15:35:23.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"e60b42abad9ff9429e9932557261187bd1f9a2dc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-09 21:10:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1672002fcaad022efd030a634bb9ce1b01f7888a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 23:31:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/834fa7a8315c4a67aae1752a70aeacb3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/fdf0346440f947519c4c2b782f9f9e97 : SUCCESS in 3m 00s\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed6d98823e2748bb859ce316c29e74f0 : SUCCESS in 11m 35s\n- openstack-ansible-deploy-infra_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/85c0b5681b0d4974a4ec8060babdc37d : SUCCESS in 1h 33m 30s (non-voting)\n- openstack-ansible-deploy-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/87edbcaaa55e4798a6503e35fbcb9b63 : SUCCESS in 1h 25m 40s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9e68ed70d26a4207be35a87914926c36 : SUCCESS in 1h 15m 41s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4e8f44ca8a0b431b890ef39187599c99 : SUCCESS in 1h 32m 04s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/acf5dad390164aa195a9ead5afd48533 : SUCCESS in 1h 37m 18s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/328e8a61a91844ef9e44215ed967feb2 : SUCCESS in 1h 24m 57s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/1ae9323832ab475fa42284c99d011378 : SUCCESS in 49m 00s\n- openstack-ansible-deploy-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16e4294d5e6e4eb48f6538c3d81f5df3 : SUCCESS in 45m 09s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7d08969cbdc49558dd1accc3f4b9003 : SUCCESS in 50m 06s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfdd5fc7a01643cca884f36e03bc4122 : SUCCESS in 1h 06m 33s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/297ac5c2b25548629ea339c39a907b89 : SUCCESS in 42m 44s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a20fdf8b78b4b9bb11681cd4e07f5a7 : SUCCESS in 1h 25m 22s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be20ea20218e4405a29eb973668b7b7f : SUCCESS in 43m 07s\n- openstack-ansible-upgrade-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f2f97835a3f141e298df9307ddfe5345 : SUCCESS in 2h 08m 55s\n- openstack-ansible-upgrade_2024.1-infra_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/48483e0580d84ab08c31286ae68f1146 : SUCCESS in 1h 02m 19s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92751bb88cd04f15b9f769e27d634192 : SUCCESS in 1h 17m 57s\n- openstack-ansible-upgrade_2024.1-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8cb5053077f440be8021d742d9a77032 : SUCCESS in 54m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a15489aa9267430281263b05ae70d66e : FAILURE in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63db6fc61be9467b943e323ff14bb381 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"339c5ee1eed982b74ad4ba99fc68aa6d1b40d19b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-10 07:01:38.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"086565411f97231e0b9f83b835faf06afc3719eb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-10 07:01:54.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b4135f6ed7dae40392f4e003cd5e4e5d82b15e0d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-10 07:01:56.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"0f891c69e6d84c4abe285b84868ebd2e1f8f7d77","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-10 07:02:01.000000000","message":"Topic set to osa/pin-setuptools-2025.1","accounts_in_message":[],"_revision_number":3},{"id":"358c620d9c3184b3745558ac6f72d7ba23fd1e22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 07:11:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ca7fc4a776e45dfb20704fcf49d1d07\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfae71d2adc8403cb349f17bc41bd615 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfab31edde0344fba5ef87074c04a421 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a94b97bb22514b6e90fdac4102a7ae21 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":3},{"id":"5fc1021ae9a07a21f731356ffdc35acdb44a760c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-03-10 10:32:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a13c9e777ebf63e8f1d99068976f8b02903117ee","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-10 15:35:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9130ac624f8b8439601f597216c67cc791664c3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 15:40:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"782810e315ea7f66e1ef93ac2c852287084bb97a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 15:59:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f239e4437fd443d9bbd6a8b187d767f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c99eead43f44024bcedf4bdadaa9664 : SUCCESS in 13m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e6c2fece781421c9813c0c42f36e1c7 : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/132adcd5826f4e3c96dbbb381588675c : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":3},{"id":"5cbf91bdab9f13f983d58125aa8022d6dc08d99d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 15:59:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"23ce2043cf4ea695852e3fe9bc242e786e9799c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 16:01:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5658606f1a164337a20f6bfd9e7dc151\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0471046d98f4f08bf41723ae4d676f3 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ef6db3efb3155f19bca3228f7e9f341596dadebf","revisions":{"ad19fde1a279b1c01681c67721e07bb94439e892":{"kind":"REWORK","_number":1,"created":"2026-03-09 21:10:28.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/31/979731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/31/979731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/31/979731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/31/979731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/31/979731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/31/979731/1"}}},"commit":{"parents":[{"commit":"b4b9ad0c1d8754be739da0171aa350b66c0e6018","subject":"Fix role idempotency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b4b9ad0c1d8754be739da0171aa350b66c0e6018"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-09 21:10:28.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-09 21:10:28.000000000","tz":0},"subject":"CI: check status","message":"CI: check status\n\nChange-Id: Ic2f42506df66e50bd754ba1ae3b78862d6c1882a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/ad19fde1a279b1c01681c67721e07bb94439e892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/ad19fde1a279b1c01681c67721e07bb94439e892"}]},"branch":"refs/heads/stable/2025.1"},"d66f697f1872e2bbd30734a1f25a7eb5631fb5ac":{"kind":"REWORK","_number":2,"created":"2026-03-10 07:01:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/31/979731/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/31/979731/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/31/979731/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/31/979731/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/31/979731/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/31/979731/2"}}},"commit":{"parents":[{"commit":"b4b9ad0c1d8754be739da0171aa350b66c0e6018","subject":"Fix role idempotency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b4b9ad0c1d8754be739da0171aa350b66c0e6018"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-09 21:10:28.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-10 07:01:38.000000000","tz":0},"subject":"CI: check status","message":"CI: check status\n\nChange-Id: Ic2f42506df66e50bd754ba1ae3b78862d6c1882a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d66f697f1872e2bbd30734a1f25a7eb5631fb5ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/d66f697f1872e2bbd30734a1f25a7eb5631fb5ac"}]},"branch":"refs/heads/stable/2025.1"},"ef6db3efb3155f19bca3228f7e9f341596dadebf":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-03-10 07:01:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/31/979731/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/31/979731/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/31/979731/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/31/979731/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/31/979731/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/31/979731/3"}}},"commit":{"parents":[{"commit":"b4b9ad0c1d8754be739da0171aa350b66c0e6018","subject":"Fix role idempotency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/b4b9ad0c1d8754be739da0171aa350b66c0e6018"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-09 21:10:28.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-03-10 07:01:54.000000000","tz":0},"subject":"Pin setuptools via global-requirement-pins.txt","message":"Pin setuptools via global-requirement-pins.txt\n\nsetuptools 82.0.0 dropped support for the legacy \u0027pkg_resources\u0027\nlibrary.\n\nChange-Id: Ic2f42506df66e50bd754ba1ae3b78862d6c1882a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/ef6db3efb3155f19bca3228f7e9f341596dadebf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/ef6db3efb3155f19bca3228f7e9f341596dadebf"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
