)]}'
{"id":"openstack%2Fopenstack-ansible-repo_server~989979","triplet_id":"openstack%2Fopenstack-ansible-repo_server~master~Icce1cb98632ed2dff5dbe9392ba941324db9b602","project":"openstack/openstack-ansible-repo_server","branch":"master","topic":"osa/resolute","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-06-01 21:37:36.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Icce1cb98632ed2dff5dbe9392ba941324db9b602","subject":"Filter upper-constraints content","status":"MERGED","created":"2026-05-25 15:30:51.000000000","updated":"2026-06-02 02:22:05.000000000","submitted":"2026-06-02 02:01:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989979-osa/resolute","meta_rev_id":"1af3f73b5edf1a80f7f4a61f290c7ee3d4eddf49","_number":989979,"virtual_id_number":989979,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 02:01:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2026-05-29 15:14:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-28 17:19:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2026-05-29 15:14:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-25 18:08:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-28 17:19:57.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-05-29 15:14:28.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"32f8f8f1d10899ce5edb9032b89fab1893808638","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-05-25 15:30:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87d8fea0e72b00197964d80b1908c8e0027aa63a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 18:08:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70ba268dbe354abf9ced0566a5b34ae8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2022ed6f2c2b4954b1d2d565293f5bd3 : SUCCESS in 7m 56s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/5308368c495c4e38aa3bd05d673f6a53 : SUCCESS in 1h 08m 28s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/3bc268a3140146939841105e76044457 : SUCCESS in 1h 00m 28s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d453e75f5f1f4077b67159b42ce2f482 : SUCCESS in 1h 04m 51s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1499df5b6857433db65b8d2e6d2923cf : SUCCESS in 1h 40m 40s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5492428c4f6240f585951b68229dcf50 : SUCCESS in 1h 42m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6d3516dab7c476d9e82fa6363b8e7df : SUCCESS in 1h 32m 34s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f189a6541ce438a8b8e60a8e0ed9b55 : SUCCESS in 1h 50m 55s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20a3dfa40ecb4c5d9d7e04452b1222a8 : SUCCESS in 1h 55m 26s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd0117dc632b4e9c813eb0247ec394dd : SUCCESS in 2h 26m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be61d444e920456783c08ef6c21f78dd : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af1f63d8d76747acb9de934a97a24fca : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"37f67130de94ac98b1f63f878745c487ec71c2b1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-05-28 17:19:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d7c9f23d6f414eccb5fab9138137b54c5dc53b84","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-05-29 15:14:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"caedb9880ef4f887769d8cfc87586c442bc84b6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 15:17:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe0c2c83c0b929022560767524c525261d6c03ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 17:35:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dae78b1f3f9e4bb5bd13a8c81f0c7b6d\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab0281b9059d4458ad7ebc68be66bc42 : SUCCESS in 9m 01s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/50ac865ac27b4e1384d3a3e634b38ef4 : SUCCESS in 1h 11m 45s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/e5174344b4da42fc81d47da9ce0bffa5 : SUCCESS in 1h 50m 52s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/51813b49cb224b5fa2426388dab263ce : FAILURE in 1h 14m 00s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/32772bc626e9432db8a967b39ea3b8d9 : FAILURE in 47m 36s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a6b6f6aacddd45adb383b063f8a828a7 : FAILURE in 40m 43s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1bc5b9b6a7543beba2b5a7e8c7d9f4d : SUCCESS in 1h 14m 08s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/513d559139ab429aa990a39f678b93bc : SUCCESS in 1h 13m 37s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8299976d69d54cad8464b33670d32c43 : SUCCESS in 2h 02m 37s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7895064657084a0a8352db3f34b01267 : SUCCESS in 2h 12m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66d5c6e5c1b24d15875a41828cc1c719 : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e902fd3f123e456aad9ce3427c265e76 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f910fe877ec8c473c9ae291127886e7233838a87","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-01 11:54:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00877ae7f604d5cb1aee71d7c10c6335e73d5324","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 13:40:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d713b549028416f889f2864810509eb\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91496d6196fa4361911f1a47717d6ee5 : SUCCESS in 10m 58s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/6f5bfac223e941fab23dd1c1fc387d2c : FAILURE in 58m 53s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/64ab2dd659c54b4f99e4da3726029baa : FAILURE in 1h 09m 07s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/9f98f4ddd0e645aa8b32b650c0641c25 : FAILURE in 1h 15m 06s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9b1821c16f0a49f4a0610ff1f6b2f276 : FAILURE in 1h 04m 12s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c8b756525c244a7e975b31832215d90d : FAILURE in 1h 21m 36s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec7891242f89434fa6718737f5813dd2 : FAILURE in 1h 09m 13s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d0478ebbcbe43f9b208847fb084efe9 : FAILURE in 1h 02m 19s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8bece5c51034ccebc4f358b368a8118 : FAILURE in 1h 11m 42s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7e12ed5933248ce8d42f95cc0a20b25 : FAILURE in 1h 42m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cdd98a362d64caab73867dda72c69f9 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c78728fd7f9a4ae5bd5af64abd9ba06a : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"487b99a30e43767d73c0f52d2eeea6374896d7dc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-01 21:37:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e294e1c80a84fc961b78a6f9064b76c18d94002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 23:49:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0b56d1fb9814611b0c45159b36907fd\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4b375160f024778b40f5f3d5b4bb287 : SUCCESS in 12m 36s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/2f133b2c0bfb4ce4a432b44e022fd2da : SUCCESS in 1h 36m 51s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/674120eb328646c9b7670308dd7fffd3 : SUCCESS in 1h 35m 09s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/d26b19776c4b4d7192c8f764df349214 : SUCCESS in 1h 29m 53s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4f17a85124e84d5e93ac510a293c2a75 : SUCCESS in 1h 21m 46s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/07bc5a14db544de3b2c85ff02a59c574 : SUCCESS in 1h 03m 28s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f45705edc474bce8f9a0df1eec03982 : SUCCESS in 1h 36m 01s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b3dbdd154064bd5ae3c8567df7ef87d : SUCCESS in 1h 16m 42s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4aca88b3b71d48d29f919133fbe40edc : SUCCESS in 1h 17m 24s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/690c543595dc40ad98d5a28c8ce72306 : SUCCESS in 1h 57m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/770dd0b9219c4b468928b1a006610f19 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6eea71acb3694fb9808c5877849e662a : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"7280669c4d9fada0e11d9024162768c142271191","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 23:51:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4fc69936892af4363a6531269eb54e82ec61dc87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 02:01:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19db71318b5848f98d7d77b4104d9d68\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff967d39bbb9440ab9e0194d17c646e8 : SUCCESS in 7m 55s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/5cbb96da05234a44a3b828fbae50fccc : SUCCESS in 54m 36s\n- openstack-ansible-deploy-infra_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/d77da0d76c264a289c008c0da3180db9 : SUCCESS in 1h 24m 32s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/eeb869470e1b4d35a68feddc62e933e0 : SUCCESS in 55m 32s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dcea5a71a9db4297a055014fafabe745 : SUCCESS in 1h 16m 30s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/11edd1b0564a4272b472578d343e20cd : SUCCESS in 1h 17m 56s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/857ba88fbde24c2994ae306d8afb5cb0 : SUCCESS in 1h 28m 47s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1b237c1f60842f6b7184f5c99bd9272 : SUCCESS in 1h 10m 39s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5056461606e84d90be1ab1877b2d4750 : SUCCESS in 52m 39s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb006e8c115f4876a67e5b377cbd107a : SUCCESS in 2h 04m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25e8a1bd4a0a4dc590faae1ddc9f1ef7 : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/295633dcbbb6460b8b26e139913bb1f6 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"5ac14a78670d36ea7811a38822d763bf23c79044","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 02:01:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1af3f73b5edf1a80f7f4a61f290c7ee3d4eddf49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 02:22:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3cda992c5fd4233a18a3b3eb17fec37\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1149274b688d48a1a819a1c79f675ab3 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81a25c2080b2d9d261f03254af2e10c7500c4bf2","revisions":{"81a25c2080b2d9d261f03254af2e10c7500c4bf2":{"kind":"REWORK","_number":1,"created":"2026-05-25 15:30:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/79/989979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-repo_server","ref":"refs/changes/79/989979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/79/989979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/79/989979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/79/989979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-repo_server refs/changes/79/989979/1"}}},"commit":{"parents":[{"commit":"a063a51859135bdf9eeee27d04e5f3812f798d79","subject":"Updated from OpenStack Ansible Integrated repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/a063a51859135bdf9eeee27d04e5f3812f798d79"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-25 15:30:42.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-05-25 15:30:42.000000000","tz":120},"subject":"Filter upper-constraints content","message":"Filter upper-constraints content\n\nAs part of the deployment process, we store upper-constraints\ncontent on the repo server, while caching it locally on the deploy host.\n\nWith that, we maintain a global-requirements, which may be conflicting\nwith upper-constraints at some points.\n\nIn order to avoid such conflicts, we can filter out our cached version\nof upper-constraints to not contain any records from\nthe global-requirements, and allow global-requirements to be\na source of truth.\n\nThis way we cal solve a numpy conflict for Ubuntu 26.04, as it has\nincompatible version pinned in u-c, but also puts us in control for any\nfurther usecases.\n\nChange-Id: Icce1cb98632ed2dff5dbe9392ba941324db9b602\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/81a25c2080b2d9d261f03254af2e10c7500c4bf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-repo_server/commit/81a25c2080b2d9d261f03254af2e10c7500c4bf2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
