)]}'
{"id":"openstack%2Fopenstack-ansible-rsyslog_client~664380","triplet_id":"openstack%2Fopenstack-ansible-rsyslog_client~stable%2Fpike~I19129fb8224307615bd52eaada627639992e7e01","project":"openstack/openstack-ansible-rsyslog_client","branch":"stable/pike","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I19129fb8224307615bd52eaada627639992e7e01","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:47:14.000000000","updated":"2019-06-10 18:17:35.000000000","submitted":"2019-06-10 18:17:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664380-1560190655661-e232e0c4","meta_rev_id":"786be88bbc1e001aed69e70d27a241553c34505c","_number":664380,"virtual_id_number":664380,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-10 18:17:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-06-10 17:54:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-06-10 17:54:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:54:33.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-10 18:17:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8148f5e8c8ef9c3c8bc5cb682cfe3fecfe49068d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:47:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"305df4a669f2611552bc92859fbe37d93ceb1253","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:37:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/80/664380/1/check/build-openstack-sphinx-docs/7b33474/html/ : SUCCESS in 4m 05s\n- openstack-ansible-linters http://logs.openstack.org/80/664380/1/check/openstack-ansible-linters/f4fbfae/ : SUCCESS in 8m 34s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/80/664380/1/check/openstack-ansible-functional-centos-7/7f53cb6/ : SUCCESS in 6m 52s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/80/664380/1/check/openstack-ansible-functional-opensuse-423/c096967/ : SUCCESS in 5m 25s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/80/664380/1/check/openstack-ansible-functional-ubuntu-xenial/5500d23/ : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a54bfc04ac02fc83b8afeabc520af837b750fcb2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-10 17:54:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b28192c146f12466eeb7581ef31db7be167e9421","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:54:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"204ebaa4572be04105f3c9aeaa5120ef70a5f39b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:17:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3eaabe01c4d7763efa8f8a33468215fc764fcc54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 18:17:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/80/664380/1/gate/build-openstack-sphinx-docs/2a56221/html/ : SUCCESS in 4m 18s\n- openstack-ansible-linters http://logs.openstack.org/80/664380/1/gate/openstack-ansible-linters/a53fb54/ : SUCCESS in 4m 47s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/80/664380/1/gate/openstack-ansible-functional-centos-7/031be2e/ : SUCCESS in 8m 45s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/80/664380/1/gate/openstack-ansible-functional-opensuse-423/a27d2c7/ : SUCCESS in 6m 26s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/80/664380/1/gate/openstack-ansible-functional-ubuntu-xenial/9124b48/ : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d11d1d6f564c3f9b624f8c4021b06e7c6a161178","revisions":{"d11d1d6f564c3f9b624f8c4021b06e7c6a161178":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:47:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/664380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rsyslog_client","ref":"refs/changes/80/664380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rsyslog_client refs/changes/80/664380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rsyslog_client refs/changes/80/664380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rsyslog_client refs/changes/80/664380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rsyslog_client refs/changes/80/664380/1"}}},"commit":{"parents":[{"commit":"ef6fe4dac17a87c377b3e7d8201773e5dba78f81","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rsyslog_client/commit/ef6fe4dac17a87c377b3e7d8201773e5dba78f81"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:47:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:47:13.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I19129fb8224307615bd52eaada627639992e7e01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rsyslog_client/commit/d11d1d6f564c3f9b624f8c4021b06e7c6a161178"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rsyslog_client/commit/d11d1d6f564c3f9b624f8c4021b06e7c6a161178"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
