)]}'
{"id":"openstack%2Fopenstack-ansible-rsyslog_client~729716","triplet_id":"openstack%2Fopenstack-ansible-rsyslog_client~master~I4a2468490b4b1a4c33f06d708a4e471f9c2c2ba6","project":"openstack/openstack-ansible-rsyslog_client","branch":"master","topic":"pygments_style","hashtags":[],"change_id":"I4a2468490b4b1a4c33f06d708a4e471f9c2c2ba6","subject":"Update docstheme for style","status":"MERGED","created":"2020-05-20 16:53:35.000000000","updated":"2020-06-03 18:24:42.000000000","submitted":"2020-06-03 18:23:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729716-1591208585780-9e89036d","meta_rev_id":"3843ae385d7f659b58ba27332dd857677f5df2fa","_number":729716,"virtual_id_number":729716,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-03 18:23:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2020-06-03 18:11:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-20 17:21:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2020-06-03 18:11:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 17:21:10.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-06-03 18:11:16.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-06-03 18:23:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9958cd27fad0c19cad842bcf0e26c007e3fb4df5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-20 16:53:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc36e635654759609e0d33cf2bb79f840e1588e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 17:08:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/efbb84aaf59647f99239615dc18f95ab : SUCCESS in 10m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ea1446e9f0142feaf8a6d6b62b0c0ae : SUCCESS in 8m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83272c001b784f5a9390eb07c585e276 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"945ba13876837ce8f6e9d28f48cbcf977d866177","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-20 17:21:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"217938d94b149a7f2d08c0ed78ae2f2149e95549","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-03 18:11:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2efb8fb8abef95fba9316b6f9bc677d9399b2f3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 18:11:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd560c003f63a393a264b1f213afe7d08b2ab1b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 18:23:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3fb81e6c1bfb4fc291e92c7371bb8bb5 : SUCCESS in 9m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12055e6fde78456cb268f83b35d18f83 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d543f13a982466d8cd8552665acc628 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"8ba1fa8d9dafd58ce2b64590a736e3d8e4b8fb43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 18:23:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2a928b752df15bcfcf3547636c46019179bdf5dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 18:24:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/653e2c58976140798913ad5d9779f267 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9c30bf93b504f6881175a84ecd2af2d : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a23f2d96c5f4d1ee77a955834bf0eb0c3cbe8ea9","revisions":{"a23f2d96c5f4d1ee77a955834bf0eb0c3cbe8ea9":{"kind":"REWORK","_number":1,"created":"2020-05-20 16:53:35.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/16/729716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rsyslog_client","ref":"refs/changes/16/729716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rsyslog_client refs/changes/16/729716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rsyslog_client refs/changes/16/729716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rsyslog_client refs/changes/16/729716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rsyslog_client refs/changes/16/729716/1"}}},"commit":{"parents":[{"commit":"9b2012b5250486195a7891159f16a3af364dba78","subject":"Merge \"[ussuri][goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rsyslog_client/commit/9b2012b5250486195a7891159f16a3af364dba78"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-20 16:53:30.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-20 16:53:30.000000000","tz":120},"subject":"Update docstheme for style","message":"Update docstheme for style\n\nNew version of openstackdocstheme (Victoria+) respects pygments_style.\nSince this repo is using now Victoria (master) requirements but has\nnot branched for Ussuri yet, it uses the new version.\n\nChange pygments_style to \u0027native\u0027 since old theme version always used\n\u0027native\u0027 and the theme now respects the setting and using \u0027sphinx\u0027 can\nlead to some strange rendering.\n\nChange-Id: I4a2468490b4b1a4c33f06d708a4e471f9c2c2ba6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rsyslog_client/commit/a23f2d96c5f4d1ee77a955834bf0eb0c3cbe8ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rsyslog_client/commit/a23f2d96c5f4d1ee77a955834bf0eb0c3cbe8ea9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
