)]}'
{"id":"openstack%2Fopenstack-ansible-rsyslog_client~772057","triplet_id":"openstack%2Fopenstack-ansible-rsyslog_client~master~I24ec156c4b6d06542c16c61c658c6f0effacfee9","project":"openstack/openstack-ansible-rsyslog_client","branch":"master","topic":"reno_build","hashtags":[],"change_id":"I24ec156c4b6d06542c16c61c658c6f0effacfee9","subject":"[reno] Stop publishing release notes","status":"MERGED","created":"2021-01-22 16:29:08.000000000","updated":"2021-01-26 18:46:37.000000000","submitted":"2021-01-26 18:20:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772057-reno_build","meta_rev_id":"b2c7565928b7d93e1a05cfdc218a0a16b7e8c3b3","_number":772057,"virtual_id_number":772057,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-26 18:20:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-01-26 16:25:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-26 09:58:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-01-26 16:25:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-22 19:43:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-25 08:36:47.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2021-01-26 09:58:03.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-01-26 16:25:52.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"32b184a7ca0202178fd834f04f555cc0ccbacbbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-22 16:29:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"834a7dcc74543dcd114fcaa808a36dcdf624f6c4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-22 16:34:51.000000000","message":"Topic set to reno_build","accounts_in_message":[],"_revision_number":1},{"id":"30035d9f46c1e2139fb0e2ff59dded661cdbbeed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 19:43:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ed047e865c5a4a30b5566fa7a9042a6a : SUCCESS in 8m 41s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9ebe0c99fc1945f09336b01718720075 : SUCCESS in 8m 33s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/9b3fe29d4bdc4fd0ac492844f5fe262a : SUCCESS in 10m 34s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3982d819b6cf490cb648787ab4098999 : SUCCESS in 8m 18s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/464ad2b5ec4b43038832c42c0f7ab64f : SUCCESS in 10m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/16c2f862240d4da1bbc9de83c255238e : SUCCESS in 1h 13m 21s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5638fd5604904860904ea860c52ea28b : SUCCESS in 1h 20m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77e7cf21834a464688b302a1f5e8a67e : SUCCESS in 1h 22m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e86e293fc54435c83b6c6905de12a7d : SUCCESS in 1h 28m 21s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d529658c7a31470680960cd3e95699b4 : SUCCESS in 1h 38m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16978c421f2e47d58c01de6941ab9f1f : SUCCESS in 1h 13m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f6224dd773724d0e8c31264599790cbd : SUCCESS in 1h 01m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bca8c5f7d16c44e3aae78e183a060275 : FAILURE in 22m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65df711ecfd24a2d93524d989f9e99a7 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a7ce1bed90d6c2007e5ed99830f2cecfcb20c436","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-25 08:36:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9b42a56951babb645030325896ecefb055f408c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 10:41:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/adc814484e0242769ca022612eaf62c7 : SUCCESS in 7m 28s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/1db1ab325afc44c7b67f24cff46cba7a : SUCCESS in 9m 01s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/504d7a862c854bd1bfcc9799d1edd76d : SUCCESS in 8m 27s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e09d8b13e994a51b1c73d5fb7df89fb : SUCCESS in 10m 24s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a01edd89aa5443129ce43d4c1fa0a379 : SUCCESS in 7m 52s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9935132717864595b83aa59c51e6313e : SUCCESS in 1h 15m 46s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/489588a75e4a4395bb31d5837e6a87e2 : SUCCESS in 1h 23m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54dc346def194392bfb881d630dcb4d7 : SUCCESS in 1h 20m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e01f7f6b2ee4e6080b5184eb5d63ac8 : SUCCESS in 1h 09m 28s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d86e90cd91e24f2a991c8beb3b879c93 : SUCCESS in 1h 53m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be0465d8b73543dca2ddb4d88aced779 : SUCCESS in 1h 33m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/48d1f8ed9edb413b92e758cf2ff2b3f3 : SUCCESS in 1h 05m 00s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c850f19b665849e29096222151f1ab1f : SUCCESS in 1h 01m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f725754f166543e494b4e3e38beac6cc : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c92a807e2a05c774f90de651007c516c529f0f95","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-26 09:58:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3df743bba895011efe7df7ec0c0a5e11e5d94137","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-26 16:25:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f814971d92a7cfd2b3b2abcec8ed8ad54c05935","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 16:26:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95d95543e176a278c325342bf8f79366926d5c67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 18:20:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1620257f0740440689ac83cee3cda69e : SUCCESS in 8m 35s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/91431f3fb1b240d1b6f2c30a1f608a98 : SUCCESS in 7m 23s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66f58820e87a44bd9e67dbf707be0303 : SUCCESS in 10m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/34a86b7ccd424fc3b765bca2d165e55a : SUCCESS in 1h 06m 28s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9d49e68707db45fbb2da92dea58ec7f9 : SUCCESS in 1h 17m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b21be955124404fa111c2d5f2dfbfff : SUCCESS in 1h 14m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d42f8ca3ded4433a138890fb33cb557 : SUCCESS in 1h 23m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9ff127275e2d4994b4323b03e6d56cb0 : SUCCESS in 57m 00s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f39442ee419a4d7aadbdcb195ab37ca1 : SUCCESS in 59m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eba14c756b394fcfb05cfcce142c93b4 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6eb063722b7c527bca926cb6926f441d2a22c9c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 18:20:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b2c7565928b7d93e1a05cfdc218a0a16b7e8c3b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 18:46:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec67871307bd45759c7b8a1a4ebfeb42 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06593fb6eeab700f47b3ea2ada6ab43531794a7d","revisions":{"06593fb6eeab700f47b3ea2ada6ab43531794a7d":{"kind":"REWORK","_number":1,"created":"2021-01-22 16:29:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/57/772057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rsyslog_client","ref":"refs/changes/57/772057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rsyslog_client refs/changes/57/772057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rsyslog_client refs/changes/57/772057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rsyslog_client refs/changes/57/772057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rsyslog_client refs/changes/57/772057/1"}}},"commit":{"parents":[{"commit":"271be6c8d93aed12bb5e77119573144a64122c61","subject":"Remove examples for variables which no longer exist","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rsyslog_client/commit/271be6c8d93aed12bb5e77119573144a64122c61"}]}],"author":{"name":"dmitriy","email":"dmitriy@dmitriy-XPS-13-9380","date":"2021-01-22 16:29:02.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-01-22 16:29:02.000000000","tz":120},"subject":"[reno] Stop publishing release notes","message":"[reno] Stop publishing release notes\n\nSince we copy all release notes to the integrated repo there is not need\nin publishing release notes for each repository. We should only verify their\nvalidity and linting.\n\n\nChange-Id: I24ec156c4b6d06542c16c61c658c6f0effacfee9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rsyslog_client/commit/06593fb6eeab700f47b3ea2ada6ab43531794a7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rsyslog_client/commit/06593fb6eeab700f47b3ea2ada6ab43531794a7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
