)]}'
{"id":"openstack%2Fopenstack-ansible-rsyslog_server~441331","triplet_id":"openstack%2Fopenstack-ansible-rsyslog_server~master~I67849272f833b5be88164f7742e0ed5df077703d","project":"openstack/openstack-ansible-rsyslog_server","branch":"master","topic":"python2-pyOpenSSL","hashtags":[],"change_id":"I67849272f833b5be88164f7742e0ed5df077703d","subject":"Install python2-pyOpenSSL package on CentOS","status":"MERGED","created":"2017-03-03 18:50:15.000000000","updated":"2017-03-03 20:31:45.000000000","submitted":"2017-03-03 20:31:45.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"be88307d83643acdbaa262d57b7fe1595db8d2db","_number":441331,"virtual_id_number":441331,"owner":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-03-03 20:31:45.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-03-03 18:52:06.000000000","_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":2,"date":"2017-03-03 20:27:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":1,"date":"2017-03-03 20:27:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-03-03 18:52:06.000000000","updated_by":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"reviewer":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"state":"REVIEWER"},{"updated":"2017-03-03 20:27:53.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2017-03-03 20:31:45.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"b379e79b8784da4887e1376966897533135dac2f","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2017-03-03 18:50:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"828b2e53bc5cfeeeca2105effbc759e7deae8be8","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2017-03-03 18:52:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8e27966622a2c4b341ef9828e84fe3e610cd084a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-03 18:58:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-ansible-rsyslog_server-docs-ubuntu-xenial http://docs-draft.openstack.org/31/441331/1/check/gate-openstack-ansible-rsyslog_server-docs-ubuntu-xenial/4576634//doc/build/html/ : SUCCESS in 1m 07s\n- gate-openstack-ansible-rsyslog_server-linters-ubuntu-xenial http://logs.openstack.org/31/441331/1/check/gate-openstack-ansible-rsyslog_server-linters-ubuntu-xenial/4b38546/ : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6b4a5ff86566fc67546d076c817057eb319acade","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2017-03-03 20:27:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"67ddf3cf43f7bc029e91a82383389bd9906fcf2d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-03 20:28:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"a3246ac7b9f8e7da2bb21c10bf67505cf41c2018","date":"2017-03-03 20:31:45.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"ac003161a2a6526c21a9636987905aab1792c493","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-03 20:31:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-openstack-ansible-rsyslog_server-docs-ubuntu-xenial http://docs-draft.openstack.org/31/441331/1/gate/gate-openstack-ansible-rsyslog_server-docs-ubuntu-xenial/824790a//doc/build/html/ : SUCCESS in 56s\n- gate-openstack-ansible-rsyslog_server-linters-ubuntu-xenial http://logs.openstack.org/31/441331/1/gate/gate-openstack-ansible-rsyslog_server-linters-ubuntu-xenial/3f7492d/ : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b933c87f60cd302041f62a3f792ef9dcc7e2ab2f","revisions":{"b933c87f60cd302041f62a3f792ef9dcc7e2ab2f":{"kind":"REWORK","_number":1,"created":"2017-03-03 18:50:15.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/31/441331/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rsyslog_server","ref":"refs/changes/31/441331/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rsyslog_server refs/changes/31/441331/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rsyslog_server refs/changes/31/441331/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rsyslog_server refs/changes/31/441331/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rsyslog_server refs/changes/31/441331/1"}}},"commit":{"parents":[{"commit":"021e2e499c7f04be58c47d80a10e5552fe0b972e","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rsyslog_server/commit/021e2e499c7f04be58c47d80a10e5552fe0b972e"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2017-03-03 18:48:06.000000000","tz":-360},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2017-03-03 18:50:15.000000000","tz":-360},"subject":"Install python2-pyOpenSSL package on CentOS","message":"Install python2-pyOpenSSL package on CentOS\n\nPackage pyOpenSSL is obsoleted by python2-pyOpenSSL.\n\nChange-Id: I67849272f833b5be88164f7742e0ed5df077703d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rsyslog_server/commit/b933c87f60cd302041f62a3f792ef9dcc7e2ab2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rsyslog_server/commit/b933c87f60cd302041f62a3f792ef9dcc7e2ab2f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"021e2e499c7f04be58c47d80a10e5552fe0b972e","is_merged_in_target_branch":true,"change_id":"I252aa01b3c6f82d2e6d78cb3e123f20224d770c7","change_number":440438,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
