)]}'
{"id":"openstack%2Fopenstack-ansible-rsyslog_server~517972","triplet_id":"openstack%2Fopenstack-ansible-rsyslog_server~stable%2Fnewton~If1f536b0022f3a352854c0d0043631a6d22655a7","project":"openstack/openstack-ansible-rsyslog_server","branch":"stable/newton","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"If1f536b0022f3a352854c0d0043631a6d22655a7","subject":"Updated from OpenStack Ansible Tests","status":"ABANDONED","created":"2017-11-06 12:28:22.000000000","updated":"2017-11-07 10:33:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c677790a509dfb50587fa775c3d5d78c1c40a0bc","_number":517972,"virtual_id_number":517972,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2017-11-06 12:38:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-11-06 12:38:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c06c6eab71a4d2ec1923b54f7411ff633d59dc26","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-11-06 12:28:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1a8d66b022378c13fe20163919b98b514e0e0a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-06 12:38:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/72/517972/1/check/build-openstack-sphinx-docs/c0b04a8/html/ : SUCCESS in 2m 08s\n- openstack-ansible-linters http://logs.openstack.org/72/517972/1/check/openstack-ansible-linters/07906e6/ : SUCCESS in 3m 45s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/72/517972/1/check/openstack-ansible-functional-ubuntu-xenial/7d5dceb/ : FAILURE in 4m 54s\n- openstack-ansible-functional-ubuntu-trusty http://logs.openstack.org/72/517972/1/check/openstack-ansible-functional-ubuntu-trusty/3f6ba28/ : FAILURE in 2m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a8624683d4fa567d1808cbbdb4e8adecee3a979e","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2017-11-07 10:33:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e24f9cb8e1d11ceb5b88f73fe2312bf9c0cd73fd","revisions":{"e24f9cb8e1d11ceb5b88f73fe2312bf9c0cd73fd":{"kind":"REWORK","_number":1,"created":"2017-11-06 12:28:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/517972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-rsyslog_server","ref":"refs/changes/72/517972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-rsyslog_server refs/changes/72/517972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-rsyslog_server refs/changes/72/517972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-rsyslog_server refs/changes/72/517972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-rsyslog_server refs/changes/72/517972/1"}}},"commit":{"parents":[{"commit":"3c06e80213d21a60b76420557e00fb3cddc5e5cd","subject":"Initial OSA zuul v3 role jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rsyslog_server/commit/3c06e80213d21a60b76420557e00fb3cddc5e5cd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-11-06 12:28:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-11-06 12:28:20.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: If1f536b0022f3a352854c0d0043631a6d22655a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rsyslog_server/commit/e24f9cb8e1d11ceb5b88f73fe2312bf9c0cd73fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-rsyslog_server/commit/e24f9cb8e1d11ceb5b88f73fe2312bf9c0cd73fd"}]},"parents_data":[{"branch_name":"refs/heads/stable/newton","commit_id":"3c06e80213d21a60b76420557e00fb3cddc5e5cd","is_merged_in_target_branch":true,"change_id":"Ibce728f863d5e21addd93b2bf36c8b4c54bb8bd2","change_number":514036,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/stable/newton"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
