)]}'
{"id":"openstack%2Fopenstack-ansible-tests~588191","triplet_id":"openstack%2Fopenstack-ansible-tests~master~Ia1657a181ab00310c5b186a16cd91fccf35661c1","project":"openstack/openstack-ansible-tests","branch":"master","topic":"move-mq-setup-into-roles","hashtags":[],"change_id":"Ia1657a181ab00310c5b186a16cd91fccf35661c1","subject":"Prevent common MQ task set from synchronizing","status":"MERGED","created":"2018-08-02 09:41:32.000000000","updated":"2018-08-03 13:20:32.000000000","submitted":"2018-08-03 13:20:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"588191-1533302432174-ad70efe2","meta_rev_id":"a6085712a1841d0cfbf3588297d272c27c20fb90","_number":588191,"virtual_id_number":588191,"owner":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2018-08-03 13:20:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2018-08-03 12:32:02.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"date":"2018-08-03 12:28:25.000000000","_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":1,"date":"2018-08-03 12:32:02.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-08-03 12:32:02.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2018-08-03 13:20:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3bda9dc73c0faabaf00e8ef79bbfdccb4ac2bda7","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-08-02 09:41:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f4e4b20aaec2c0f0c31842588eff9e49cff30e9","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-08-02 09:54:20.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"ec8f937654562da23823a2c19dd3fa8e8bc8a65a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-02 10:35:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/91/588191/1/check/build-openstack-sphinx-docs/9697baf/html/ : SUCCESS in 3m 01s\n- openstack-ansible-linters http://logs.openstack.org/91/588191/1/check/openstack-ansible-linters/1bfe56b/ : SUCCESS in 6m 33s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/91/588191/1/check/openstack-ansible-functional-centos-7/806c4e4/ : SUCCESS in 51m 55s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/91/588191/1/check/openstack-ansible-functional-opensuse-150/7efd4ff/ : FAILURE in 19m 10s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/91/588191/1/check/openstack-ansible-functional-opensuse-423/b7cf5de/ : SUCCESS in 42m 18s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/91/588191/1/check/openstack-ansible-functional-ubuntu-bionic/29dd4e8/ : FAILURE in 9m 26s (non-voting)\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/91/588191/1/check/openstack-ansible-functional-ubuntu-xenial/2a798e5/ : SUCCESS in 38m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b96f0517544660bd0711dff2b1f6641e49539ede","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2018-08-03 12:28:25.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"7939cd9c0ebdea3883ba7c069db9502cc7275d6d","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2018-08-03 12:32:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a6a71d731909863f4a4ef33b8e9a914b18bad7f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-03 12:32:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"564e370e3a0249259a70b3ccd9af5bec960670e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-03 13:20:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/91/588191/1/gate/build-openstack-sphinx-docs/068b412/html/ : SUCCESS in 2m 46s\n- openstack-ansible-linters http://logs.openstack.org/91/588191/1/gate/openstack-ansible-linters/692d45b/ : SUCCESS in 7m 33s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/91/588191/1/gate/openstack-ansible-functional-centos-7/666aead/ : SUCCESS in 46m 42s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/91/588191/1/gate/openstack-ansible-functional-opensuse-423/d47242f/ : SUCCESS in 34m 18s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/91/588191/1/gate/openstack-ansible-functional-ubuntu-xenial/23175d7/ : SUCCESS in 34m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4754dffd15f3c012b67a5679ebf038ac28613ae4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-03 13:20:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9abc3018e1a5afa2bdeece48a91ac776a50606d","revisions":{"e9abc3018e1a5afa2bdeece48a91ac776a50606d":{"kind":"REWORK","_number":1,"created":"2018-08-02 09:41:32.000000000","uploader":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"ref":"refs/changes/91/588191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-tests","ref":"refs/changes/91/588191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-tests refs/changes/91/588191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-tests refs/changes/91/588191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-tests refs/changes/91/588191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-tests refs/changes/91/588191/1"}}},"commit":{"parents":[{"commit":"ca68eafc063e943fddf6a3b6d84a4c3efdef7452","subject":"Merge \"Implement clouds.yaml for root user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-tests/commit/ca68eafc063e943fddf6a3b6d84a4c3efdef7452"}]}],"author":{"name":"Jesse Pretorius","email":"jesse.pretorius@rackspace.co.uk","date":"2018-08-02 09:39:25.000000000","tz":60},"committer":{"name":"Jesse Pretorius","email":"jesse.pretorius@rackspace.co.uk","date":"2018-08-02 09:41:12.000000000","tz":60},"subject":"Prevent common MQ task set from synchronizing","message":"Prevent common MQ task set from synchronizing\n\nAs we need to update all the roles for the new mechanism,\nwe rename the task file to prevent changes in the tests\nrepo from reverting the changes in the roles.\n\nNeeded-By: https://review.openstack.org/588190\nChange-Id: Ia1657a181ab00310c5b186a16cd91fccf35661c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-tests/commit/e9abc3018e1a5afa2bdeece48a91ac776a50606d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-tests/commit/e9abc3018e1a5afa2bdeece48a91ac776a50606d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
