)]}'
{"id":"openstack%2Fopenstack-ansible~1003511","triplet_id":"openstack%2Fopenstack-ansible~master~I05515010a2a8510c11e065649ca8ae94cc261277","project":"openstack/openstack-ansible","branch":"master","topic":"osa/ansible.mariadb","attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-09 16:10:28.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I05515010a2a8510c11e065649ca8ae94cc261277","subject":"Remove community.mysql dependency","status":"NEW","created":"2026-09-02 11:17:57.000000000","updated":"2026-09-18 09:16:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d68097a20a30285f0beb9e3086d9e4c7d34f039e","_number":1003511,"virtual_id_number":1003511,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-03 16:36:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":2,"date":"2026-09-18 09:16:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-09 16:10:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 12:49:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-09 16:10:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-09-18 09:16:17.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"828ea778a41e29ea5fbc0cb23c886b3ae8cec5b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-02 11:17:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e59e2690c90e107cc4300767c3d9fa266abe96b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 12:49:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1fe1bbe36f334ca5b1d3f0442714f996\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75efb742b5b24fd6894296a2f5984ef2 : SUCCESS in 13m 47s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/d36ec201c78c444797848d62d2ae41eb : SUCCESS in 15m 42s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/d3d8418a8ad74c4981040650965fd135 : SUCCESS in 15m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59602c18e8894295891ba7bdf61d3029 : SUCCESS in 9m 29s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/496fb47cafef4fe693d134415574f348 : FAILURE in 36m 58s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/416e587d803e47c6a94ed78999d44c6e : FAILURE in 42m 00s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b725889dd7554f41bdc13ebc14379fa4 : FAILURE in 32m 49s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16683ffb75554ce8b9b5da9866cd4485 : FAILURE in 43m 36s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/29a603a8ee8147da93263dc39061a03a : FAILURE in 29m 25s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/4e495ef8434147a7b8bfb4ccf38080ce : FAILURE in 19m 27s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/cf37f47441704ccbb2f3a391a0aef0d0 : FAILURE in 29m 11s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cf0834733256439c8100bd79c593a864 : FAILURE in 20m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eed9b12ea4884f549eba7c67603873d8 : FAILURE in 29m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d61b916f7604bd5bee02dc5cd46f77c : FAILURE in 19m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a82fb59f6eb04c8f8605101d364c185c : FAILURE in 18m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10cdc5058c69468b84bb8e95b715a9f7 : FAILURE in 31m 33s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6dc445bc289b43eeb165e24ff5ac221f : SUCCESS in 1h 20m 42s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1664b6d2a0d24e8d90ae1d37881594a8 : SUCCESS in 1h 16m 26s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b98df763f91949b8930ee48f67f5d663 : SUCCESS in 1h 18m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcdacaa8a8624e3484ac757c2b5c3d55 : FAILURE in 21m 30s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a170429d9ed54ceba5bb55c79a9912c6 : FAILURE in 21m 09s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a4e17fdece144f9a58a05a0a4231975 : FAILURE in 43m 04s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eabc6d8fd95e454bb3ec4404aba4930f : FAILURE in 29m 50s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31cba7467b7d4f129c7854a737268709 : FAILURE in 47m 23s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e253a7f31e18485fb731e0ffe68649b6 : FAILURE in 21m 08s\n- openstack-ansible-deploy-aio_metal_openbao-ubuntu-noble https://zuul.opendev.org/t/openstack/build/775a990b2499456eb067932f76f700f7 : FAILURE in 31m 24s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/bd8dd8977e664ae3a28c33ccaf851854 : SUCCESS in 6m 24s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/434830e2651245698a30bb553202a071 : SUCCESS in 6m 45s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/c18f4b543a2840c1b87327bba30133e9 : SUCCESS in 5m 13s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/79858831299247ada172c4cb4304219c : SUCCESS in 9m 02s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e850559a7e70479e961bc8354e253251 : SUCCESS in 4m 32s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cf3048b419f474a93fe76fbeadd2e2c : SUCCESS in 13m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d8fa118015269f50163c00dee2484be1df6cb166","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-03 13:19:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71a3ad0c227fb67b4c84703fcdee498f4f4d2570","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:36:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43d6d5bf769a47739eb1ff6767eb41ef\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7e0d03fc88740ee8e40cdfad7ab3f29 : SUCCESS in 17m 21s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/ac763fc274a74a55ade8cca1f1b7bd2f : SUCCESS in 22m 22s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/a9e6cfcdc9224b88942afd612fccb7a4 : SUCCESS in 16m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f8df91e7e5b4afdb83d845804fe2e5f : SUCCESS in 9m 57s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/03745f3c793a44e3bcde87e94852cc71 : SUCCESS in 1h 59m 45s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/262824cd692e423489e8fa23a66f0144 : SUCCESS in 2h 15m 50s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6645e643561846a1aab2700cdbb3fda7 : SUCCESS in 2h 15m 16s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/281139f251f8411cbae2c669cb97ec55 : SUCCESS in 2h 13m 49s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/1612019bb58b4b4d83588a905a1e1dad : SUCCESS in 1h 53m 00s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/0da6d5c0d16141a381f1fc3dc1e8a5a8 : SUCCESS in 1h 23m 38s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/9572592b9e5b43aa884b6b4eeb79d9ba : SUCCESS in 1h 23m 01s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/759b0bda19404eba8248c51fc91c1612 : SUCCESS in 1h 00m 53s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/47dd70c8341347038e8e1875d8af39ca : SUCCESS in 1h 21m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fb5bc9440b44197a3ad860a1373ab4f : SUCCESS in 1h 26m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/cd68b4ccd24c44b1817749f71315477a : SUCCESS in 1h 24m 39s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46cbd58b193144d0a61adfc963b3cf7c : SUCCESS in 1h 41m 32s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/da530ceda61948789b64d3288ca7c304 : SUCCESS in 2h 28m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0833f78f9e30408fa042d1f3810340a1 : SUCCESS in 2h 26m 37s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/655f2685da724416817c09814501b8c5 : SUCCESS in 2h 34m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f07166aef3742b9a8c79fa9076e2166 : SUCCESS in 1h 25m 56s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/139d03ceda3640c58249c6df35258d9f : SUCCESS in 1h 27m 07s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cfdf9f7fc494029bc9e52649922bb71 : SUCCESS in 2h 51m 49s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/371053949fa84e608dee14f960436747 : SUCCESS in 1h 18m 13s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76da166f6de8445f8cc7f0543549fb0c : SUCCESS in 2h 30m 31s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26517390209e40628898647e70597179 : SUCCESS in 1h 40m 39s\n- openstack-ansible-deploy-aio_metal_openbao-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6702a83e1d8408bb76b1238a5c92f86 : SUCCESS in 1h 20m 45s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/a5e5230327a94d31a3f9db10f6fa5492 : SUCCESS in 5m 32s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48722b46b65a4bb8814faca3d4365ec7 : SUCCESS in 8m 32s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/466f133e88434d7ea8643a42385df5e0 : SUCCESS in 6m 24s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/1bc76196df054862aac7c4a19bf66080 : SUCCESS in 13m 51s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6fe278d8827647a6b5496385803dced8 : SUCCESS in 13m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a258fcce3b984b45a069799d7fa7fad5 : SUCCESS in 14m 22s","accounts_in_message":[],"_revision_number":1},{"id":"88058e5fe2d8f64b7aa05f8f5111100b9a0bdc94","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-09 10:13:07.000000000","message":"Topic set to osa/ansible.mariadb","accounts_in_message":[],"_revision_number":1},{"id":"1f26319f62622e26558d160349aad60bb97fb173","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-09 16:10:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d68097a20a30285f0beb9e3086d9e4c7d34f039e","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-09-18 09:16:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8de09cd1af8b9b9298fbf23326ade875fa3e2b1e","revisions":{"8de09cd1af8b9b9298fbf23326ade875fa3e2b1e":{"kind":"REWORK","_number":1,"created":"2026-09-02 11:17:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/11/1003511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/11/1003511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/11/1003511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/11/1003511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/11/1003511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/11/1003511/1"}}},"commit":{"parents":[{"commit":"a409fd0eafa5ff6c792befd3aa33c642ff51e0b8","subject":"Add ansible.mariadb to collections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/a409fd0eafa5ff6c792befd3aa33c642ff51e0b8"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-02 11:16:02.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-02 11:17:46.000000000","tz":120},"subject":"Remove community.mysql dependency","message":"Remove community.mysql dependency\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/1003500\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-galera_server/+/1003510\nChange-Id: I05515010a2a8510c11e065649ca8ae94cc261277\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/8de09cd1af8b9b9298fbf23326ade875fa3e2b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/8de09cd1af8b9b9298fbf23326ade875fa3e2b1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
