)]}'
{"id":"openstack%2Fopenstack-ansible~1003877","triplet_id":"openstack%2Fopenstack-ansible~master~I3424aa1d79b96938b22f28f7d241f6bb0f4733ad","project":"openstack/openstack-ansible","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-09-07 09:41:36.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-07 09:41:40.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3424aa1d79b96938b22f28f7d241f6bb0f4733ad","subject":"Do not disable NetworkManager in AIO","status":"ABANDONED","created":"2026-09-03 19:56:37.000000000","updated":"2026-09-07 09:41:40.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ef2938fd79808069b92bedef4575c19561d855d5","_number":1003877,"virtual_id_number":1003877,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-04 09:25:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"disliked":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":-1,"date":"2026-09-07 09:27:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-04 10:51:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2026-09-04 10:51:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 22:58:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-04 10:51:40.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-09-04 10:51:40.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"a4c5e38609b01296cc1d827236a2bf33acb453b0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-03 19:56:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c73d6e6d7e7e7164ee2318fbf62eeae4f386543","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-03 20:00:16.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"86fed66cbc361e99d8cee51858e78677733ae17d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-03 20:00:18.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"cfd12ef95e36df9462095596a0ba17698dadba85","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-03 20:01:00.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"824e9c165286101ccf4b486e12ee791b1d87040f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-03 20:02:12.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0cd49738e15a7f7b2554c405efc76545e4e920de","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-03 20:02:14.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"755f40ff79d28cf7cee661f93f18917f75bfc641","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 22:58:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4691a7a2c5046908cf1e5dc6eba4298\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e74373cb789436abcb471bb172f9665 : SUCCESS in 13m 25s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/42f91dd2f2d948b19ebc69250bc927d1 : SUCCESS in 11m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6545d3af51f4ca29a44bac884ab5726 : SUCCESS in 10m 11s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/95aa68f6abd4497789339ab8914f8999 : SUCCESS in 1h 08m 15s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c0b125eb41cd4fc4930fa5d0c387d436 : SUCCESS in 2h 07m 21s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c9c411e5e7d4473e81d3b5d216b90cb1 : SUCCESS in 1h 58m 54s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d27064a5385d41778ad61669fcf08085 : SUCCESS in 1h 36m 31s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ee102dcbbcd2482d8f54e8dc0567708d : SUCCESS in 1h 05m 48s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/1a5039aeb1b743afbe6476e3a0d0464c : SUCCESS in 1h 01m 22s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/942d8614263747afac9c20d8ac88af7f : SUCCESS in 1h 16m 37s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/86da4a268e3c44e4bef0bfe366abef2c : SUCCESS in 46m 20s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/952e06ede1714c8ca8115320ca2b0c15 : SUCCESS in 1h 08m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cca21aa7d4164859a43a509aca616a6c : SUCCESS in 1h 24m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/87f347e4c22a46cab7f66e9e66836e3d : SUCCESS in 1h 21m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5663587b2734d0e8c8d9b9c7303dff7 : SUCCESS in 1h 21m 49s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5845dce5ecf54557acb20fdf900e0853 : SUCCESS in 1h 17m 27s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71826e2b85eb4a9daf60c6c850ef7165 : SUCCESS in 2h 16m 53s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cd9ecb0e84147e7a5fa5c2202c03652 : SUCCESS in 1h 46m 24s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b36a86e3b2384327a9ff985aa881c715 : SUCCESS in 1h 18m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/c40636164e7c4ca693d55996fabf40f0 : SUCCESS in 48m 03s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19397e6053a347de8c7c6e9223844869 : SUCCESS in 2h 53m 22s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/246f683d60d44f5b8d7d27942838af34 : SUCCESS in 1h 52m 40s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d5110baa65a41c8ae48b40b20abee8d : SUCCESS in 2h 22m 58s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a31f74f4a15f4b6b8c39ebe5c2a807ee : FAILURE in 18m 56s\n- openstack-ansible-deploy-aio_metal_openbao-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee67afbe116640e3b32b68538adae58c : SUCCESS in 1h 03m 43s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/738af5542e8e45d9a49436fddf3ea69a : SUCCESS in 5m 37s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99b84ab93e3447148403a31c8cae0301 : SUCCESS in 6m 03s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e5282c7b2f6a41d58ce05bc7fda20e77 : SUCCESS in 5m 35s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/f94d648eac8e40ac8c89e495adaf23a4 : SUCCESS in 9m 05s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b8bb75f4f4b04628a43ad9ae22a0cfab : SUCCESS in 5m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/336a481862e941d5bc4d346239f475ca : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":3},{"id":"b6e99c5cffa06fc51a7710c28821e52e24d1065a","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-04 06:44:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0d7fc9e14a87c1966c7f6ff784811a0024b5c3b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 09:25:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e310785a53e495baf74952eee9baf47\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d025286b20594cd191a23067c179650f : SUCCESS in 9m 39s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/7e0d5b564c3845be9b26f3592d82d25b : SUCCESS in 11m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd826d921aa469981af577e024c5622 : SUCCESS in 7m 00s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/4bbe4688abe34a55ac34b99e9fe10914 : SUCCESS in 1h 21m 46s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/3599bbb9ef0444849f8d88990f7cee28 : SUCCESS in 1h 23m 06s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a2e01397aaa347bdb9dbf2347db62062 : SUCCESS in 2h 33m 17s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21e77394d7d345a4acf2210ea28b3c81 : SUCCESS in 1h 18m 35s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/78b189ef5c4e4f919fc748cf5e8740a0 : SUCCESS in 1h 14m 26s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c2c4500640b2480db4de85161a3dd3b5 : SUCCESS in 1h 05m 43s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/2f420cc85c104264bec2bc757e29f892 : SUCCESS in 50m 26s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9e63616bb2ca4b9b85aafe0e5ab84985 : SUCCESS in 55m 50s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/48497977025d4360b6db01cc648d2028 : SUCCESS in 52m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/786415c9163f44188096f14fd5ba120d : SUCCESS in 45m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f27d550538ff403eb26a915f733e13bf : SUCCESS in 44m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/280d96709655461a94fdc6b9a50729ab : SUCCESS in 50m 52s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/388338addab648288d941accc059cacf : SUCCESS in 1h 24m 27s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d498a2ddd73141a3852e872b11a4a465 : SUCCESS in 1h 51m 27s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8ae8e436cbe4394b492b9b3aa119cc1 : SUCCESS in 2h 30m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24dad3affbd748549d4483ead3eaaac5 : SUCCESS in 1h 19m 41s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b8e4342bc0844b00bd5b981ddce73274 : SUCCESS in 1h 21m 26s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0961aa96cecf4a6ba24a89e93e1ce7cf : SUCCESS in 1h 48m 38s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66f96fad26124b8c95e3f436254140b7 : SUCCESS in 1h 07m 23s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0de629b008645d3b0dd17f6ad2260ea : SUCCESS in 2h 18m 57s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a2d4f711bbe48fb8229a5b2d71e3bb0 : SUCCESS in 45m 45s\n- openstack-ansible-deploy-aio_metal_openbao-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95184ec1296045f8a9f0d38ddb2602dd : SUCCESS in 42m 34s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/ff9538849ec844d0873edd17b4c7f0cc : SUCCESS in 5m 05s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32f4c70e11ef42fa803a602482b0c318 : SUCCESS in 5m 19s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/45a071b24da84039802d8d9abba0702b : SUCCESS in 3m 40s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/89a3ccd0e40e44469f3a094f5fba764f : SUCCESS in 6m 09s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/925d629fbc8545a489aad3f3b9f5489e : SUCCESS in 5m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/412586d8ecd44b84970e2c0e6521393e : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":3},{"id":"9ac18d99fd64bab78dcf087171a3abb877fd6883","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-04 10:51:40.000000000","message":"Patch Set 3: Backport-Candidate-1 Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"be3efa590d08b5e703fb34c3f3da772adc9db227","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-07 09:27:40.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f02bae06634f4eccddb002b69890814f964dcb83","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-07 09:41:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ef2938fd79808069b92bedef4575c19561d855d5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-09-07 09:41:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b93edfce36f8dfeb5657bab048f34ce561e96ada","revisions":{"861230340aa8f0293418d997254a7bde3e77af4a":{"kind":"REWORK","_number":1,"created":"2026-09-03 19:56:37.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/77/1003877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/77/1003877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/77/1003877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/77/1003877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/77/1003877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/77/1003877/1"}}},"commit":{"parents":[{"commit":"8d9c53dcc498de77b0521ed1d736ef4023397353","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/8d9c53dcc498de77b0521ed1d736ef4023397353"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-03 19:56:37.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-03 19:56:37.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I3424aa1d79b96938b22f28f7d241f6bb0f4733ad\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/861230340aa8f0293418d997254a7bde3e77af4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/861230340aa8f0293418d997254a7bde3e77af4a"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"1f9484eee7f6295420107309ace1f1350fa7adcf":{"kind":"REWORK","_number":2,"created":"2026-09-03 20:00:16.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/77/1003877/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/77/1003877/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/77/1003877/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/77/1003877/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/77/1003877/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/77/1003877/2"}}},"commit":{"parents":[{"commit":"8d9c53dcc498de77b0521ed1d736ef4023397353","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/8d9c53dcc498de77b0521ed1d736ef4023397353"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-03 19:56:37.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-03 20:00:11.000000000","tz":0},"subject":"Do not disable NetworkManager in AIO","message":"Do not disable NetworkManager in AIO\n\nAs we\u0027ve disabled by default NetwokManager by running \nsystemd-networkd role, let\u0027s leave it enabled for AIO\ndeployments, as we don\u0027t define main interface.\n\nChange-Id: I3424aa1d79b96938b22f28f7d241f6bb0f4733ad\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/1f9484eee7f6295420107309ace1f1350fa7adcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/1f9484eee7f6295420107309ace1f1350fa7adcf"}]},"branch":"refs/heads/master"},"b93edfce36f8dfeb5657bab048f34ce561e96ada":{"kind":"REWORK","_number":3,"created":"2026-09-03 20:02:12.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/77/1003877/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/77/1003877/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/77/1003877/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/77/1003877/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/77/1003877/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/77/1003877/3"}}},"commit":{"parents":[{"commit":"8d9c53dcc498de77b0521ed1d736ef4023397353","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/8d9c53dcc498de77b0521ed1d736ef4023397353"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-03 19:56:37.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-09-03 20:02:08.000000000","tz":0},"subject":"Do not disable NetworkManager in AIO","message":"Do not disable NetworkManager in AIO\n\nAs we\u0027ve disabled by default NetwokManager by running \nsystemd-networkd role, let\u0027s leave it enabled for AIO\ndeployments, as we don\u0027t define main interface.\n\nChange-Id: I3424aa1d79b96938b22f28f7d241f6bb0f4733ad\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/b93edfce36f8dfeb5657bab048f34ce561e96ada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/b93edfce36f8dfeb5657bab048f34ce561e96ada"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
