)]}'
{"id":"openstack%2Fopenstack-ansible~1003949","triplet_id":"openstack%2Fopenstack-ansible~stable%2F2025.2~If751eb2be90382459a37c15a04f5350280160dac","project":"openstack/openstack-ansible","branch":"stable/2025.2","topic":"bump_osa","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-09-18 10:56:38.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-18 10:56:38.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-09-18 10:56:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If751eb2be90382459a37c15a04f5350280160dac","subject":"Bump SHAs for 2025.2","status":"MERGED","created":"2026-09-04 12:49:56.000000000","updated":"2026-09-18 10:58:15.000000000","submitted":"2026-09-18 10:56:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003949-bump_osa","meta_rev_id":"a515f5b7fe94bc025e2874e8c5ceea8c23600ccb","_number":1003949,"virtual_id_number":1003949,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-18 10:56:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-09-18 08:07:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-17 14:23:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2026-09-18 08:07:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 15:33:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 14:23:55.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-09-17 14:24:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-09-17 14:24:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-09-18 08:07:09.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"291aca2f5945ab9a77c4b77c58373f222ad7bf2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-04 12:49:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b2c41da9b5dbd2786db11f3f4018752f6f43e47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 15:33:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aabf1e53d01b4437a20e283eb67a13c3\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc870438488b4d6e9b7a9c7d0d976fb9 : SUCCESS in 15m 38s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/c312bf2b69034daab791a17c18c47f66 : SUCCESS in 6m 33s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/ef236d0ae2ed4ce1847556627bff0983 : SUCCESS in 11m 05s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/1d2fb153daa94c1999afe515753fc875 : SUCCESS in 14m 52s\n- openstack-ansible-tox-inventory https://zuul.opendev.org/t/openstack/build/4cdac7e4e2594508b4136827b7cdc7e2 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cdf8d3d7a93415187447a6678e8950e : SUCCESS in 10m 18s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9a95a32143c04353a52443b07dc48478 : SUCCESS in 2h 04m 28s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/7925411efed04e5a977e2b8f19784ded : SUCCESS in 2h 10m 09s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/12f1c13dc61a4d7ebb930f09a061fa06 : SUCCESS in 1h 52m 47s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/24c9ee7033bf4d77b5850b881b58a977 : SUCCESS in 1h 49m 26s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48e3423d7f8c4727b60c1a7c238ceeef : SUCCESS in 2h 18m 37s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3229cccd5b4b4b88a95f379e25f87ef4 : SUCCESS in 1h 19m 20s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5848f04cb0154c9e8decdc971fb5e1a1 : SUCCESS in 1h 32m 25s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/cb99e772020543ff830cfcb994d64cb4 : SUCCESS in 46m 38s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4bd7681149c345f191c94b154eeff749 : SUCCESS in 1h 38m 16s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fdd83bd2d2f6436492900b350d8012db : SUCCESS in 1h 04m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f08d54da73704ec5b90b246e919fea08 : SUCCESS in 1h 29m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/534d64d7211d458fbd1e2fb82128a1b3 : SUCCESS in 1h 27m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56e21e55192d4da08deca0bb44ab171b : SUCCESS in 2h 02m 35s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/febe3198d11440e2a0643927f4df1c27 : SUCCESS in 1h 59m 18s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0d8ae4ab3a7640df8e955908d316e7f8 : FAILURE in 15m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a03deda80384226b972bad0bc6f731f : FAILURE in 44m 00s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/706be1b7de494217af2f70f78e593c76 : SUCCESS in 1h 25m 28s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3840a916b5e4423a7d8a51e1b754962 : SUCCESS in 2h 00m 42s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a0e0d3fa3bb4723baa4e414f01eb21e : SUCCESS in 2h 20m 56s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/166624592210419ea9f6ed8878892297 : SUCCESS in 1h 34m 54s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/3ee11b901f4e4dfdb08cfd242a318373 : SUCCESS in 6m 18s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/a9e798167e4d4b458702914741573420 : SUCCESS in 7m 09s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8d2152c10d143c287017902b0711a6c : TIMED_OUT in 31m 02s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/e1c0e313c5d74e8fbea6f4ef24e2a43b : RETRY_LIMIT in 2m 52s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8577070ce7054a84ad909bd9a800af7b : SUCCESS in 6m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de5b69aa5efa4035b50b28576042e492 : SUCCESS in 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"7d0aa328df294aff790fbae30d089b1b69db7f5a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-04 19:57:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54f4c674f2190447010cf206d65d1479b0417408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 22:34:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86a3337cd24e456f8f07c81beb0cb5ad\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1abd5e55c2434208abdeb0ebf0804f7c : SUCCESS in 8m 38s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/be9d0e93cff142269a96b2b9f142a758 : SUCCESS in 8m 39s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/1bfcd7b6ca43458ab92127319f68412b : SUCCESS in 13m 05s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/79d326060790423dad0d9ccaa7f241a8 : SUCCESS in 9m 30s\n- openstack-ansible-tox-inventory https://zuul.opendev.org/t/openstack/build/56ba4d8e3eba4bcc9c69b039703864bc : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90e19d356f91493385a96ff6536a87ca : SUCCESS in 8m 09s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/cf16ca16f3d6410d80ee3451dd92acc0 : SUCCESS in 1h 19m 49s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/f5f292d07a5f428c93b29e14c97024cf : SUCCESS in 1h 04m 41s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/717b2885be1247efa739d0a0b1996e88 : SUCCESS in 1h 50m 12s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2d4c18f0631640e9b61f71a50bea4a6f : SUCCESS in 1h 09m 33s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3e91647dd4b4f10a0804253bf894d73 : SUCCESS in 2h 13m 26s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/32f36f132e964a6c9035033d0eaba21f : SUCCESS in 43m 14s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/de438490bf3f488e9faf2d65912db2a0 : SUCCESS in 1h 26m 43s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/0b445edae7174818afd6f12a8cef1cff : SUCCESS in 50m 20s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/07ae6f8ffaa14bed89ce8987ebb05ec3 : SUCCESS in 1h 20m 37s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e3719e484be34778a05a2aa088ea466c : SUCCESS in 52m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/921e1ab5c6b44bfd88704edafa505434 : SUCCESS in 1h 17m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8350b470cd134a3f849b2d611c9de496 : SUCCESS in 48m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad3d8650eea749e18900a6829f863395 : SUCCESS in 1h 52m 13s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e28aac5fe874731ac69102b34d9c0f6 : SUCCESS in 2h 17m 15s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/99967fcfdb2e49a4b1821b58f4136fd1 : FAILURE in 12m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3f405e372cf4d09bc71d933bbec01ee : FAILURE in 55m 24s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d17148e13ad14d1286ed57718843ba93 : SUCCESS in 2h 28m 57s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c66933551d9e4f70ac274b288e03812b : SUCCESS in 2h 21m 14s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54fc9fe50e6c46e1b534ef0c391f435f : SUCCESS in 2h 07m 14s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d82a48f5e5124772b411df5a56617d4a : FAILURE in 1h 22m 56s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/973bf7b691ee4e74915d7a570997ce56 : SUCCESS in 5m 15s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/480aed2b88b64f909a868bf83ff6e39b : SUCCESS in 5m 47s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52ac69eabb014f3180effb2287ce342e : SUCCESS in 7m 30s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/ef78faec6da5414ca05ccb93f65cf1b7 : SUCCESS in 8m 57s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7c90e610eb66450c8994a398f38d8528 : SUCCESS in 5m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba0671fa819a482e9eac342b6d694f42 : SUCCESS in 13m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b6fa86ee50ae178f5b0a0f118383892da0e4edc1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-07 09:15:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c83b0155029e9edfaea784ccbd31b9646af62c3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 12:03:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2da6e7223c04a5389a201b6f6e0dfe1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ccb73c4184654883bdf9d598c3a7cde8 : SUCCESS in 14m 32s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/0f5e550f050a46babc815475bdad0870 : SUCCESS in 7m 19s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/b558023f375c4437b0738f8a76852446 : SUCCESS in 15m 37s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/dafaa825cb0346e69c2f5424ecbde940 : SUCCESS in 14m 37s\n- openstack-ansible-tox-inventory https://zuul.opendev.org/t/openstack/build/e12f2cc0376c4d37a871ff51c3c89a60 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3364e0e249094a028fa6dcc45faa7309 : SUCCESS in 9m 08s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/bcc2d348549947e0b9a37a0313713477 : SUCCESS in 1h 54m 29s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/8dda992825f14988935eea47d3b91a51 : SUCCESS in 2h 02m 09s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/2d7a7ae8118b4bc093d3d17bac122e41 : SUCCESS in 2h 28m 53s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/523a8d9b554b4077ab3f51df058d98c1 : SUCCESS in 2h 25m 46s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/758d367f51b34d75a036fb0890a503ff : SUCCESS in 2h 15m 54s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/af69338217c14ea8ad95e89d49beb999 : SUCCESS in 1h 17m 27s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/047f9abd418b4d39ab1560f8fe662632 : SUCCESS in 1h 29m 05s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/399a0ba4fe1644e6bb5849395365ca5d : SUCCESS in 1h 27m 36s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c5ef1e90fd3d413eafad4183e4d0c6dd : SUCCESS in 1h 32m 47s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a42ed74efd644f2d95dca37b14f2b257 : SUCCESS in 1h 27m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c07b7380d0cb4bf1b34cd0343d72f46d : SUCCESS in 1h 41m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb019f41e2fe49eb8d94f5669c0e7d30 : SUCCESS in 1h 13m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8905b44684f4445a90b48ce1dd6b8110 : SUCCESS in 2h 29m 01s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/819e74be89a34e0080ea23bb5c508206 : SUCCESS in 2h 17m 29s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cde1ce197f7d41aba75394b5cf279196 : FAILURE in 11m 02s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f2a8a026973454d9618b46446f1da1f : FAILURE in 57m 38s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67ef92c8d0484dcbae88331a086d80fb : SUCCESS in 2h 13m 26s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ac17487d6884a54a27ce164a88d2e17 : SUCCESS in 2h 18m 49s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b26893c8dd8468fa49143a0aa03948d : SUCCESS in 2h 21m 05s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07a5deec7a8b4967a4565289192c3a55 : SUCCESS in 1h 30m 28s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/bd7bd5ac834f4ecf93a5d4c2de75a086 : SUCCESS in 5m 35s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/51831e769765494dae39299e9b68407f : SUCCESS in 6m 30s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff009a5cb2ca43b5a0a1859d74485403 : SUCCESS in 7m 08s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/1e605ea94c534191812f8baf9e2a6e78 : SUCCESS in 5m 58s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4581c6c389624c168135765aa958a4e9 : SUCCESS in 5m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21af502542f74e6fa3de129c8764903a : SUCCESS in 14m 31s","accounts_in_message":[],"_revision_number":1},{"id":"4c3d561ff8db9a36af580c557cdfb3679711b525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-14 09:37:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"196d654525fadb9089b06f5bc58232f2f800f9ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 12:20:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc2949587d8c4fb0aa81de42f44f47c0\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46a341cb0c8749df8cb8df6bd380e01b : SUCCESS in 15m 17s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/39a3f863c82c407e8cfb692d4aba64b1 : SUCCESS in 6m 33s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/dbacf27482fa46f1af0010451de84839 : SUCCESS in 13m 35s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/6ac211e2478b4924a830b9ab2da8d6df : SUCCESS in 13m 13s\n- openstack-ansible-tox-inventory https://zuul.opendev.org/t/openstack/build/52ae84d96c284bd2abec8ffab6533cff : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5547e1e3dda43d0a691ef45f953bf08 : SUCCESS in 8m 08s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/dbd9f61ffba04b8e9872773114008c20 : SUCCESS in 1h 36m 30s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/4fe282017a804c9db20a085b43e86911 : SUCCESS in 2h 10m 37s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c9a1eaaa97494f22a05569129d359e2d : SUCCESS in 2h 18m 23s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/132900e2c0694fe495b2915c25ce9118 : SUCCESS in 2h 25m 27s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e80bb0a941d4153a5b17fa980a1ce24 : SUCCESS in 1h 45m 07s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/2f682e5acea44bf1a3bed97726ab115c : SUCCESS in 1h 19m 12s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e779300df5334f87b0c7681fbb72e1ab : SUCCESS in 1h 13m 37s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/2eb7a191593646e687949c65a4421bc6 : SUCCESS in 1h 22m 43s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/56871331120c4023aa0a172cc634fb18 : SUCCESS in 1h 21m 11s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/52bf9b1c08bf4d1e8228905e134471ed : SUCCESS in 1h 16m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fab47497547d403297c18fdb80c6a200 : SUCCESS in 1h 37m 31s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/660fac99e6f24f858612f98a1e0b81a6 : SUCCESS in 1h 34m 34s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02192650aea04cb4b0583f7dcabeafc7 : SUCCESS in 2h 11m 33s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01b80a4016704c259c49b1a4daad55be : SUCCESS in 2h 15m 53s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/43c143a6445149eb8e4ae01598b4d891 : FAILURE in 9m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00e2192ef9c94fcd82ddd4e1906d1fb1 : FAILURE in 59m 02s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff5b8ffd49fe496ab795ec5b687d2600 : SUCCESS in 2h 25m 57s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3abfa597dd854fbab459439f1399a27f : SUCCESS in 2h 10m 48s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/914622f53914489d999fea64b2e78131 : SUCCESS in 2h 20m 32s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7150fb60185248d7a23e0c0c671f4591 : SUCCESS in 1h 37m 45s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/8f843575f11b40738b6bb64993caed15 : SUCCESS in 6m 05s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/bd938ff21ee5427c9077e72c3f5b96c5 : SUCCESS in 7m 16s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/218c2e5078164a58bca0a3201702e78e : SUCCESS in 6m 50s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/66a60665d3fb4619a0247bf2b700e1dd : SUCCESS in 7m 23s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/879ca5018b214646acb08df820125f86 : SUCCESS in 8m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0b1d4358413455ab387f4b73288b8c9 : SUCCESS in 10m 17s","accounts_in_message":[],"_revision_number":2},{"id":"e9b641b76f049ef56f80b20b143a04e08e36792e","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-17 14:23:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"447fd5d7e3a6f0ab0ed01bc5b72660270bc97944","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-09-18 08:07:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4e0c733570f2e50b2f7173381e1562701a0a88af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 08:12:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a94121341bdd97ff7d68e3adddc8790440a8ddd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 10:56:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ade47fb26354b86b305ab9aa1c49654\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d32fc871be7946068a6cb2658b9a47ea : SUCCESS in 13m 49s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/b64fefca965e49c18ac98a0e00e0211f : SUCCESS in 6m 13s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/1a782816abad4905b27dfbcbc269d890 : SUCCESS in 12m 05s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/651f1af1bc6f43499eefd23c368da940 : SUCCESS in 18m 35s\n- openstack-ansible-tox-inventory https://zuul.opendev.org/t/openstack/build/bbb29d261ff34b38bc969842e9ae8ee5 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe8f6e0f362c44458e1750aa6722c08f : SUCCESS in 7m 33s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/b4892f27d97e405b82c90fc46cffeb57 : SUCCESS in 1h 52m 53s\n- openstack-ansible-deploy-aio_lxc-debian-bookworm https://zuul.opendev.org/t/openstack/build/c2582f294cfc4a64a51f076005d76877 : SUCCESS in 1h 52m 15s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/84d2e78dd2db44f9a5b213b02a9435ed : SUCCESS in 2h 16m 35s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2a0520c0962d4a509420f6e64759e344 : SUCCESS in 1h 17m 59s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b41508dda0c94dcfb5a8f7a37a472ac8 : SUCCESS in 2h 26m 10s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/e4e816e8a7e64004818ed4198faec825 : SUCCESS in 49m 20s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8a625dcff7b147e6b576f809ae21159b : SUCCESS in 48m 24s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/499ba1e344f44f7f9f340842ed2483be : SUCCESS in 1h 29m 19s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1d308dc7d7ed47d0b30665b2f9a4f505 : SUCCESS in 55m 25s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5cf37983879242c292289caa656b6cd7 : SUCCESS in 52m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2846263ef4804d4bb2395a6b1177a445 : SUCCESS in 53m 15s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2880a2920f244954827a662fe36082f3 : SUCCESS in 1h 36m 13s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee7391a2ecff4aeca2d59bac43824ffc : SUCCESS in 2h 38m 33s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/639414d75e994d5c8434fdbcfacb62f8 : SUCCESS in 2h 09m 20s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e8292874d314f919f96d35fa6076105 : SUCCESS in 1h 22m 35s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b478c1f862845caab1052986d0df96d : SUCCESS in 1h 37m 39s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/5452772ceaa34a39a1bd98e9c8650f72 : SUCCESS in 4m 52s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/71e13098b2b84999ac48b4392c538a8b : SUCCESS in 5m 14s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e70dafaadc9846dba1dd60df1609b9ab : SUCCESS in 4m 55s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/4c35f86733034d258c5a79aaf50d8037 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c64a483bf769407f9abcb27c919605a3 : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":2},{"id":"090f968e71b9aa0e23144b6b3ab1e46ddca976e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 10:56:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a515f5b7fe94bc025e2874e8c5ceea8c23600ccb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 10:58:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1c69988fbb3480aa252f11324aad0c2\n\n- promote-deploy-guide https://zuul.opendev.org/t/openstack/build/632fbc1720914e43a66acda9776b40ed : SUCCESS in 1m 01s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ba1f883c6b2475fb690b780f080e2d6 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6b45fdf3d4b4c6ca8793883a7c6a8c0 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bace99391fc241376690e20f882303e6d84f4384","revisions":{"6b79bacee585b13fb58b27e671857d472589b25b":{"kind":"REWORK","_number":1,"created":"2026-09-04 12:49:56.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/1003949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/49/1003949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/49/1003949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/49/1003949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/49/1003949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/49/1003949/1"}}},"commit":{"parents":[{"commit":"3e0b2253e7c73d3e85738591660c7b316448e661","subject":"Bump SHAs for 2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/3e0b2253e7c73d3e85738591660c7b316448e661"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-04 12:49:48.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-04 12:49:48.000000000","tz":120},"subject":"Bump SHAs for 2025.2","message":"Bump SHAs for 2025.2\n\nCloses-Bug: #2165108\nChange-Id: If751eb2be90382459a37c15a04f5350280160dac\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/6b79bacee585b13fb58b27e671857d472589b25b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/6b79bacee585b13fb58b27e671857d472589b25b"}]},"branch":"refs/heads/stable/2025.2"},"bace99391fc241376690e20f882303e6d84f4384":{"kind":"REWORK","_number":2,"created":"2026-09-14 09:37:15.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/49/1003949/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/49/1003949/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/49/1003949/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/49/1003949/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/49/1003949/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/49/1003949/2"}}},"commit":{"parents":[{"commit":"3e0b2253e7c73d3e85738591660c7b316448e661","subject":"Bump SHAs for 2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/3e0b2253e7c73d3e85738591660c7b316448e661"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-04 12:49:48.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-14 09:37:04.000000000","tz":120},"subject":"Bump SHAs for 2025.2","message":"Bump SHAs for 2025.2\n\nCloses-Bug: #2165108\nChange-Id: If751eb2be90382459a37c15a04f5350280160dac\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/bace99391fc241376690e20f882303e6d84f4384"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/bace99391fc241376690e20f882303e6d84f4384"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
