)]}'
{"id":"openstack%2Fopenstack-ansible~1004131","triplet_id":"openstack%2Fopenstack-ansible~master~I94ab8230d68e68f584db78ab2d8fb812d7bd8b69","project":"openstack/openstack-ansible","branch":"master","attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-09-05 20:05:17.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-04 21:14:09.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-06 23:46:16.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I94ab8230d68e68f584db78ab2d8fb812d7bd8b69","subject":"Drop VMWare NSX-specific from inventory","status":"NEW","created":"2026-09-04 21:12:47.000000000","updated":"2026-09-06 23:46:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f6f35b53d2265b4d6b00a81acb5c372e06938600","_number":1004131,"virtual_id_number":1004131,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-05 00:30:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2026-09-04 21:14:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 21:14:09.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-09-05 00:30:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-05 20:05:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"4942bb695aae9ff70275be10f7f5f1eb9720aa47","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-04 21:12:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5646f95923ebbe9139990210ada8cb09205e846e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-04 21:14:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e6df03ae173b46539ff04a4df4e9d72057a3c520","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-04 21:14:09.000000000","message":"Patch Set 2: Backport-Candidate-1\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"5b656d24148f25ed6c19bf8a183c0312bd50db6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 00:30:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a63cf92d57124fe38fa14894489f5ac1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/daef583602f14ad2b21adeeb031fc4f6 : SUCCESS in 9m 13s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/ab419839191f4c6bb16509a81c8e1930 : SUCCESS in 18m 03s\n- openstack-ansible-tox-inventory https://zuul.opendev.org/t/openstack/build/9392f3e69de442b2be49f9a61a7b5fa0 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f762fb8155094c4fa88b7fdc35af858f : SUCCESS in 9m 24s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/18a5c2beda2a4dc3b7566ebbbe9b047f : SUCCESS in 1h 07m 55s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/6baf0d85a85b4be29bd88fb5274795a4 : SUCCESS in 2h 16m 59s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/42a1388ae0cd4bf4b7f59c001ea5859f : SUCCESS in 2h 41m 32s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f92f42123cc54ff08dc7989dfd393a61 : SUCCESS in 2h 21m 45s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e4e3c4c456764ae28893faa9e280ccba : SUCCESS in 1h 18m 03s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/a575d0d63eed46f1b9b6992909ba5857 : SUCCESS in 52m 42s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/13adcd8c020e43cb904459535c8485c2 : SUCCESS in 1h 25m 50s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/53c5386f585047888be49981cf54b283 : SUCCESS in 1h 16m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f310060016b14dfa83797a0992a49d3c : SUCCESS in 52m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d6cb9a1cb3947cf98e021a79731bd5a : SUCCESS in 1h 22m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/29a711f70a354a7389fd5e18197dcdf2 : SUCCESS in 56m 37s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4b8b355583d4ab98bcf73fe67392b4b : SUCCESS in 1h 31m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/de04c91192584e98ae25bc841967576c : SUCCESS in 2h 23m 33s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0478f9920c074f138762049dca4036fd : SUCCESS in 2h 38m 37s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11464c9f8802434da8db0250a46bba87 : SUCCESS in 2h 32m 56s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a7358f814e84f7baaf4cd401277bbae : SUCCESS in 1h 21m 53s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/8f8c0e70d59f4df0a0b71c996b5d779c : SUCCESS in 1h 18m 16s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7adc49a2ab184a1aa8abda6f084b0a10 : TIMED_OUT in 3h 11m 09s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3c9c4cfcdfc4e3e9b8813843fc678a9 : SUCCESS in 2h 28m 11s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca45bbd2ad9042cbae4ff48424783d12 : SUCCESS in 1h 07m 11s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea7a38e185244c50848483aaf2649f71 : SUCCESS in 1h 28m 32s\n- openstack-ansible-deploy-aio_metal_openbao-ubuntu-noble https://zuul.opendev.org/t/openstack/build/692d72cdeaf04715afb810c84381065a : RETRY_LIMIT in 13m 09s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/72bfd36d89264be99bc2ee9a794aec34 : SUCCESS in 8m 32s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cd74e3ccdc74ffb8e72f932e4c14da3 : SUCCESS in 6m 34s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/eac65074260147cd86db834179beea5a : SUCCESS in 7m 12s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/a136cda661064e6bba442b7fac9a985d : SUCCESS in 6m 43s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d406a608f4b34fb59cc6a7190944b1ea : SUCCESS in 6m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abdc357dea764f31ada33790bad928a8 : SUCCESS in 10m 20s","accounts_in_message":[],"_revision_number":2},{"id":"34ed8830dd64b735be12993664748320b0f70bad","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-05 20:05:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ac626252c08771a2efef9dc79278da08b3e098fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 23:19:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/adabf292974345f1ac78fc7422fb363e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ecf4964db2c4abca2fb96ba4db6a9c1 : SUCCESS in 10m 40s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/7a60b2b778114e19997b8dbdb6954b62 : SUCCESS in 14m 29s\n- openstack-ansible-tox-inventory https://zuul.opendev.org/t/openstack/build/5b72f33d903240dab34b4653e1b0e7e7 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eed0957e0ed947de99fb23116fd9de2b : SUCCESS in 9m 24s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/0e4a72913e69495db04af5cb40b0692c : SUCCESS in 1h 40m 21s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c38540cae5b24e4bb20c93ec154919d4 : SUCCESS in 1h 15m 44s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bacc48d63640407a8505476783b85de4 : SUCCESS in 2h 30m 21s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17041924108e46de9a56512dc4c9ec0c : SUCCESS in 1h 43m 49s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/071094a2b46c482e980596f30f88b477 : SUCCESS in 2h 09m 08s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/62ef489f8469415c98d1674ad4b4e198 : SUCCESS in 1h 06m 12s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/6a6e7bb4c6bb40c59a50f73894d36618 : SUCCESS in 1h 19m 56s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5b83e38811e647c28975c222facf2001 : SUCCESS in 1h 27m 10s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/12fb7707ab7047fca1111804b3eac7c3 : SUCCESS in 1h 09m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5eadec68f2a146058f4180513630091a : SUCCESS in 1h 10m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b5ce501df46b4e61a17679ef35629961 : SUCCESS in 1h 20m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39146a77a01a46a5b56739e52757a153 : SUCCESS in 1h 08m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e1f1f7a0544546c18255c15ed2db6584 : SUCCESS in 1h 54m 38s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2837488c93af4c1f81f1bf13a47cd7fb : SUCCESS in 2h 22m 39s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38b48b1938ad42c79d113ecbc67e72ab : SUCCESS in 2h 21m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8866b18b41044178ae322aae8f11ffad : SUCCESS in 1h 19m 32s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/063b66a4582a47f596a582fc698b4db3 : SUCCESS in 46m 59s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6db8a0677fb345c797069becc0a983c4 : TIMED_OUT in 3h 08m 20s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32d45ec788bb4728bd80c974eb58250c : SUCCESS in 2h 19m 51s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c4b8b3b376746b794731ef15f70b2a2 : SUCCESS in 2h 23m 08s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a129e72abbfd4a7895795549c25cb047 : SUCCESS in 1h 29m 18s\n- openstack-ansible-deploy-aio_metal_openbao-ubuntu-noble https://zuul.opendev.org/t/openstack/build/401f69fab89f465eb0c67aee3d55fe38 : SUCCESS in 1h 23m 34s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/a04e5bd954a4408eb093e460b864eff1 : SUCCESS in 5m 17s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc049da1955f4c238e0b6189698fe121 : SUCCESS in 4m 49s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d11e8373a98d4906abb6f42804d50dd5 : SUCCESS in 4m 52s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/646a59854bfd40cf8753f08d5dc875a4 : FAILURE in 6m 00s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e98d9bba4afd47fa9f2a7f52869d714e : SUCCESS in 5m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84aef8db923b4a3682debb8dc90bf928 : SUCCESS in 12m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d5f4579e9eb1e39160cb461717330af09d1a375c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 20:10:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f6f35b53d2265b4d6b00a81acb5c372e06938600","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 23:46:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21508d507197400cb42ed1fdf8019796\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b9279cdc12a4888800dc38d486a8cae : SUCCESS in 11m 54s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/85076d887fad40cab6fd5d5d22cea8ae : SUCCESS in 11m 52s\n- openstack-ansible-tox-inventory https://zuul.opendev.org/t/openstack/build/922a9adf92a7441bace314812a3ad6c7 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56bf2b83669747b9aeb0bb2671e39e78 : SUCCESS in 5m 43s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/88fe08a439db456fa497709eda896e83 : SUCCESS in 2h 18m 07s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/da7879da63db40c480456b1ec19a0007 : SUCCESS in 2h 15m 07s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2c37216178fd479384e701ad15c8cf4d : SUCCESS in 1h 18m 12s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76b77cc3063f4277bdda94d6cbf634a6 : SUCCESS in 1h 11m 04s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5f821d14115a491a829e37650cc3e926 : SUCCESS in 1h 40m 58s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/89557a57cdc145bbaa29dc448ac47e6d : SUCCESS in 45m 17s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/157bfe12fba74da7ab0f21ddcd0e017d : SUCCESS in 49m 10s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8cf7a1c9d0a04100b7ed62f26c20ad94 : SUCCESS in 50m 45s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3765e63e33674fceb0cf6c896865f35b : SUCCESS in 1h 14m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d3f33e3d4224b22839a44e358102292 : SUCCESS in 1h 32m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/9744bb7b390048949d075e8c5d620a04 : SUCCESS in 1h 18m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8462c1b3b35e44c2b9acf1e0703c8865 : SUCCESS in 1h 26m 09s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b021080d7954489a9f8eda7524b296e0 : SUCCESS in 1h 48m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25a088a2f888467ca7e13a6e7e7ddc06 : SUCCESS in 2h 12m 39s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a09dcd07f3448629b633e0187c1164e : SUCCESS in 2h 24m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f0df91fb363405facbb3c06f542b97c : SUCCESS in 55m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f615d9018af64bce9eb2aa38a070cb62 : SUCCESS in 49m 24s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d490705e6f1f41aca9420108929153ac : TIMED_OUT in 3h 26m 55s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf4f47b9e5a84ed9a81f89d0cff07c31 : SUCCESS in 1h 10m 44s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40a9a6c5d9e346db9eeeb4995541ba3f : SUCCESS in 2h 29m 07s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86b712b8fa07478ba33cf6bf2f144c4d : SUCCESS in 1h 26m 12s\n- openstack-ansible-deploy-aio_metal_openbao-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb1083d382684d6790575472632bdda6 : SUCCESS in 1h 16m 50s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/0cd5f5bee0ba48d38208bd51f7a61dff : SUCCESS in 5m 31s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f175f1c9b0424c478d0654f89a25873d : SUCCESS in 5m 14s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/06b9e0fde2e94347b9006d1725d9b11c : SUCCESS in 5m 17s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/16e280a020cb47d1af8f8413697df411 : SUCCESS in 15m 14s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9a41e8f4dd3d4cf5aa782b9e5093e36e : SUCCESS in 4m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c44eb7e3af734593a850b624f577b343 : SUCCESS in 9m 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d3b005105220d0bb1b5611080a0a2fac38485636","revisions":{"9a1578bab4004ea6b0b04b0cfd6de85829ddf6d8":{"kind":"REWORK","_number":1,"created":"2026-09-04 21:12:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/31/1004131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/31/1004131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/31/1004131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/31/1004131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/31/1004131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/31/1004131/1"}}},"commit":{"parents":[{"commit":"8d9c53dcc498de77b0521ed1d736ef4023397353","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/8d9c53dcc498de77b0521ed1d736ef4023397353"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-04 21:12:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-04 21:12:47.000000000","tz":0},"subject":"Drop VMWare NSX-specific from inventory","message":"Drop VMWare NSX-specific from inventory\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_neutron/+/991971\nChange-Id: I94ab8230d68e68f584db78ab2d8fb812d7bd8b69\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/9a1578bab4004ea6b0b04b0cfd6de85829ddf6d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/9a1578bab4004ea6b0b04b0cfd6de85829ddf6d8"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"d3b005105220d0bb1b5611080a0a2fac38485636":{"kind":"REWORK","_number":2,"created":"2026-09-04 21:14:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/31/1004131/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/31/1004131/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/31/1004131/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/31/1004131/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/31/1004131/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/31/1004131/2"}}},"commit":{"parents":[{"commit":"8d9c53dcc498de77b0521ed1d736ef4023397353","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/8d9c53dcc498de77b0521ed1d736ef4023397353"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-04 21:12:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-04 21:14:02.000000000","tz":0},"subject":"Drop VMWare NSX-specific from inventory","message":"Drop VMWare NSX-specific from inventory\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-os_neutron/+/991971\nChange-Id: I94ab8230d68e68f584db78ab2d8fb812d7bd8b69\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/d3b005105220d0bb1b5611080a0a2fac38485636"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/d3b005105220d0bb1b5611080a0a2fac38485636"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
