)]}'
{"id":"openstack%2Fopenstack-ansible~1006174","triplet_id":"openstack%2Fopenstack-ansible~master~Idcb12085514166e87ebb174376b5e197ea451d3b","project":"openstack/openstack-ansible","branch":"master","attention_set":{"32053":{"account":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"last_update":"2026-09-18 12:09:08.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Idcb12085514166e87ebb174376b5e197ea451d3b","subject":"Extend test var for hashi_vault install_ca test","status":"NEW","created":"2026-09-18 09:04:11.000000000","updated":"2026-09-18 12:09:08.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ed974ec6ceb7f5a420b792562d41e8b8e977acd0","_number":1006174,"virtual_id_number":1006174,"owner":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-18 12:09:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 12:09:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"17d66615dc41dca7c011e9e27030b9d2bef09ae3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"date":"2026-09-18 09:04:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed974ec6ceb7f5a420b792562d41e8b8e977acd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 12:09:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/891c26a538c24057bbc5bea3869fbd39\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29dff20e93ae4e6794ca6ba9afc2c429 : SUCCESS in 19m 46s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/a37d9fd2aa4f47cdafe169d208665b16 : SUCCESS in 19m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e7cf74805984db39dc7c2a84003e093 : SUCCESS in 12m 50s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/686aee92061c49f29e8ecc349f83da53 : SUCCESS in 2h 01m 57s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/a1c013db96d943b89ee4ea84a7e470ec : SUCCESS in 2h 11m 53s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3ed8df7deb8b4718aefe698c1ddfb666 : FAILURE in 35m 46s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6971515910744e68f5e98bbbc5ecb78 : SUCCESS in 2h 16m 48s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/33a307abf6e44ba280d1af0fc7af6136 : SUCCESS in 2h 20m 23s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/2a2296ee110140de88a672607bf27d28 : SUCCESS in 1h 12m 39s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/95a74bcfa43f4d0eae9214d3757379c5 : SUCCESS in 1h 28m 47s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/24b0c5a073db4fa782a3bd489b673818 : SUCCESS in 1h 26m 13s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/86c148d7b3e848d4aff190c0ce322384 : SUCCESS in 1h 28m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bfb08637b7c41428e9be566f00860e8 : SUCCESS in 1h 34m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/50b7de9fa00d4ba092e97d1d3ccf6afb : SUCCESS in 1h 26m 08s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/431467534f3f40efaab980a748817c7c : SUCCESS in 1h 33m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a88de9da223b47e5bbed93277e2316e6 : SUCCESS in 1h 56m 11s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47aba9bb4e6846e2a05587cc3a09efe5 : SUCCESS in 2h 23m 40s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b32ba9075a5944b4b10142084838694e : SUCCESS in 2h 17m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a7eb231b38a4aa4abfb0731e79a7ad3 : SUCCESS in 1h 21m 58s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b14cb79002064d9ba0887cc9fb9f7101 : FAILURE in 1h 14m 18s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2be41d4d7714427c962a63cbac7537a9 : SUCCESS in 2h 39m 58s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78006c8c285c42ceb52529d003b70ca9 : SUCCESS in 2h 21m 37s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8538b9d9323e4997840358553a488e09 : SUCCESS in 2h 03m 43s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c41507a7d624cbea8ae727c352d006d : SUCCESS in 1h 31m 22s\n- openstack-ansible-deploy-aio_metal_openbao-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3de3e003fb9b4fed991cc1cf21c80c75 : SUCCESS in 1h 38m 04s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/c4829f6cfc134d2bb01ec76ba5e8d4ab : SUCCESS in 6m 12s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a84b47eb7b645f0af65df9317cfc995 : SUCCESS in 6m 26s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/3bd2ad080cbc4010a115dd3dd03db14b : SUCCESS in 6m 40s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/6f4b2e6729b8424698c5f2a12dc307f0 : SUCCESS in 11m 13s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9d330939e2284b81a85e71c87dcdd11e : SUCCESS in 6m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95c6f5d8e329428b8f52fabbc7c1a0a7 : SUCCESS in 12m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b6656c1ba59b9972a3f5d0dd6088287038d6ec1","revisions":{"5b6656c1ba59b9972a3f5d0dd6088287038d6ec1":{"kind":"REWORK","_number":1,"created":"2026-09-18 09:04:11.000000000","uploader":{"_account_id":32053,"name":"Denys Mishchenko","email":"denis@mischenko.org.ua","username":"arddennis"},"ref":"refs/changes/74/1006174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/74/1006174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/74/1006174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/74/1006174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/74/1006174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/74/1006174/1"}}},"commit":{"parents":[{"commit":"8d9c53dcc498de77b0521ed1d736ef4023397353","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/8d9c53dcc498de77b0521ed1d736ef4023397353"}]}],"author":{"name":"Denys Mischenko","email":"denys.mishchenko@cleura.com","date":"2026-09-18 08:57:43.000000000","tz":120},"committer":{"name":"Denys Mischenko","email":"denys.mishchenko@cleura.com","date":"2026-09-18 09:04:06.000000000","tz":120},"subject":"Extend test var for hashi_vault install_ca test","message":"Extend test var for hashi_vault install_ca test\n\nIn order to allow installation of ca certifice for the cases when we do\nnot create ca, we need to supply pki engine path.\n\nChange-Id: Idcb12085514166e87ebb174376b5e197ea451d3b\nSigned-off-by: Denys Mischenko \u003cdenys.mishchenko@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/5b6656c1ba59b9972a3f5d0dd6088287038d6ec1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/5b6656c1ba59b9972a3f5d0dd6088287038d6ec1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
