)]}'
{"id":"openstack%2Fopenstack-ansible~1006741","triplet_id":"openstack%2Fopenstack-ansible~master~I51966c00443cbbcc27449032a64cecc93ce10354","project":"openstack/openstack-ansible","branch":"master","topic":"osa/molecule","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-24 09:41:11.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"I51966c00443cbbcc27449032a64cecc93ce10354","subject":"Address ansible-galaxy fetching issues.","status":"NEW","created":"2026-09-22 14:31:01.000000000","updated":"2026-09-24 12:52:19.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"74e36782df6cc9061d752cf8fadfd96d1ffc2f2a","_number":1006741,"virtual_id_number":1006741,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-24 12:52:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-22 22:55:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8eee4704746c1f45c30c039485c0abcfac9500a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 14:31:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67ff810c716dc4f37ff3e1c7b3bf061e50cdca4f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 15:26:28.000000000","message":"Topic set to osa/molecule","accounts_in_message":[],"_revision_number":1},{"id":"c531bb7a3c6e5ac57195bea583389061f11bdcbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 17:18:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eec3b4f23a58a344e12589a8d61386e2a4297155","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 17:22:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"32d4e94c3452be377cd3f3a89f2183a370cad61c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 18:21:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4a539cfa41b3e54eb9cebe554143bab25a3b582c","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 19:12:02.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"35087f296c64b3632f36dbba1c5743801ba2e46d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 19:57:43.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"269749f2cca82ea6bacbfbb6e9b043c0442a0ad1","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-22 19:57:48.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"f6e56e2a76d6a4c809a0a7708c3e4b072c1fba0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-22 22:55:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38da3066fab2415cb268cc8fa63277a9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad2fcee8ad734fc59a5a51dfa6145667 : SUCCESS in 19m 40s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/9a368c10df1f4959bc368ae3c3dc8300 : SUCCESS in 18m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff05e6c2b9584e0e84d6364b3cf887ec : SUCCESS in 9m 19s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/2553eda6d5b74198b559fe2dcca3c28f : FAILURE in 44m 21s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/520f6c4e254d47a3b894fbc27ab84fa2 : SUCCESS in 2h 13m 49s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ef43c6e514a040cc9b16fd15617cfd4d : SUCCESS in 2h 42m 39s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64e30337bab940539b03b589d7501e1d : SUCCESS in 2h 25m 22s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/82e27d01566142d08cebaeb7cad396f6 : SUCCESS in 2h 24m 18s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/1cba4139407745dc937a349176413bd9 : SUCCESS in 1h 08m 10s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/e38a98f2c940466e8c67a2080e855a37 : SUCCESS in 1h 25m 41s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/844d3d44bed04baebafd55d8c4abae03 : SUCCESS in 1h 39m 18s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1a6d90315ff44919917c0db0eb016a4b : SUCCESS in 1h 20m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e4c42356f614002a25ea324c769988b : SUCCESS in 1h 33m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/bb8e0a907bea4f8f9066cd71bb3dba7d : SUCCESS in 1h 29m 24s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2741f39050634419a3d82bbc4c55d3a9 : SUCCESS in 1h 45m 25s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/373c63f110b64f7e98adb9a41e81db2f : FAILURE in 1h 21m 17s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d77b6215429341e39afc5fc570af8fff : SUCCESS in 2h 17m 24s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46cdaed8bf704565bc12b72aa629ee5f : SUCCESS in 2h 20m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b55b5acf86eb4369b897a2bf776074c2 : SUCCESS in 1h 25m 52s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/538b533e805c40149d8b6b864c969da0 : SUCCESS in 1h 20m 36s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/981543db2a7041378025663cc92b96e4 : SUCCESS in 2h 48m 22s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bd428bbdd354e149a4ce58e9053e808 : SUCCESS in 2h 29m 13s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e12af6dff8e3486aaff29c7e2c48aa63 : SUCCESS in 2h 32m 36s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9411004a93144d5d8c378ad841cf4029 : SUCCESS in 1h 32m 38s\n- openstack-ansible-deploy-aio_metal_openbao-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69f02abf93d54d9db23802f0e0ac1ec6 : SUCCESS in 1h 37m 03s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/eebe18ee97fe4aed87618ebb54c0e825 : SUCCESS in 4m 29s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a00abb0548ec497c8d486169e8c53970 : SUCCESS in 3m 24s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ecf8ba9a2a7d489e9538fb7c454b5cca : SUCCESS in 4m 54s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/a0771716a77749d1baae3b991e765c64 : SUCCESS in 4m 26s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/55af524f38e54bfe979ea32dc0d9aa0f : SUCCESS in 5m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2de1c3cd4bc4d8ea858dc77dd38fd55 : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":5},{"id":"7ebe1907a2d0dbb56a5f490c19bdc993e4839f0e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-23 08:59:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"77ec4fc7d9b54561d05b76611b892ad538103322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 12:56:36.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/48066c9eb81a4427b702357381062df1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bf542b84ece499ab1eade8aca3e98a1 : SUCCESS in 20m 18s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/bef7d1a675f7488187d940314c600f20 : SUCCESS in 21m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/259ce22a36304f179730588b150347c8 : SUCCESS in 11m 21s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/71a8419823e74269a702e879b561264d : SUCCESS in 1h 57m 51s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/f72c58ba8805418ba14ec2e332d25c0a : SUCCESS in 2h 17m 44s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c954f74d9b744f3ebd3ee246f8774708 : SUCCESS in 2h 08m 58s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/218357ef10b14866a404960b3f9fa20c : SUCCESS in 2h 19m 38s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d4a71a6554444dbdbc803430b9462201 : SUCCESS in 2h 08m 51s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/744d6f87657d43dbb9bc1287da9a81a8 : FAILURE in 1h 19m 24s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a31c01d348334b4bad48674a02a50e69 : SUCCESS in 1h 24m 59s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4f1f9e7077f94e5a920cd3b0a758d3b5 : SUCCESS in 1h 18m 16s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/aab0a8694df2471eaa8a62b5bcdc9208 : SUCCESS in 1h 39m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0766bd540794b4584a99512d4646ae3 : SUCCESS in 1h 34m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/2e7fce4472b34adb9c33d2fa83311780 : SUCCESS in 1h 34m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25f634e56df643089c9b251f2d839d41 : SUCCESS in 1h 30m 51s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/91d005e473f14038a014fd18cef3c414 : SUCCESS in 1h 58m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3181e9d7099344459920957673dd912c : SUCCESS in 2h 20m 21s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d71be90f2c964abca5b29e26669b3b24 : SUCCESS in 2h 43m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c10ab93f49c0402fb4c370ce078ca5b4 : SUCCESS in 1h 24m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5c831d56e2ad43218eb1fa8fb0978a08 : SUCCESS in 1h 23m 16s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6759dc3ff3ee49a8b48eddf7385b007e : SUCCESS in 2h 41m 08s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39e0b5f934c641638e1a52d555fb3873 : SUCCESS in 2h 29m 25s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e2a97eb59be41368e361c0ccd47475b : RETRY_LIMIT in 15m 04s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c397e8047e4449f9bdf6a2ae0915d283 : SUCCESS in 1h 26m 30s\n- openstack-ansible-deploy-aio_metal_openbao-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80207896a7b2490d95a4b7d37f965854 : SUCCESS in 1h 27m 23s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/e9a4380fcc954bc288727b6cdd2dc091 : SUCCESS in 5m 02s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3880acafeb224dcfa2c3d18b8559af04 : SUCCESS in 5m 28s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/1ebf46fcdf3e4023bbae840fb466988a : SUCCESS in 6m 48s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/2ef9760b9a4143809931a8469616fb5d : SUCCESS in 4m 47s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e70bb74ecd704971be70fe69dba47dc9 : SUCCESS in 6m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0058caa7da5c49248fbc3ded934d1842 : SUCCESS in 13m 07s","accounts_in_message":[],"_revision_number":5},{"id":"1f8093c0b5b9e0aecd12023bd4be511d06315c74","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-24 09:41:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"74e36782df6cc9061d752cf8fadfd96d1ffc2f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 12:52:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a294fab5116144868274b64419c3de36\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10f934ae1b6e4bc99ec82150d7aad49a : SUCCESS in 23m 01s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/67f6a8a8d1954e7b84c610301957bd41 : SUCCESS in 16m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f20432572e8489eb491235c0dc28cd7 : SUCCESS in 7m 18s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/c361e5aa667f4ec0ac90a127c42a8d6c : SUCCESS in 1h 19m 15s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/813eb9fc6b9143d0b6ecc684454b19c3 : SUCCESS in 1h 33m 41s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/30d1445fc9ba419996bfd26926e56cd6 : SUCCESS in 1h 40m 00s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4eab4c91cdef466d8b9de58507a030f5 : SUCCESS in 2h 27m 33s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a224d4bc33474809880d4706356bc7f2 : SUCCESS in 1h 55m 51s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/ff1c5f679eb14d52963fb4c0df4ab4a1 : SUCCESS in 1h 31m 49s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/b3b12abf7bbb4195956fe7cd66540081 : SUCCESS in 57m 19s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/088b672cdfd645e490f14246a24c4880 : SUCCESS in 1h 38m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ae0e20cecf8b492984dbe836273766ff : SUCCESS in 1h 02m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f27eef889c74832a317cfde9aa1aa21 : SUCCESS in 1h 34m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/6ddc8b02a23a468cb27f8220a621b173 : SUCCESS in 1h 35m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a42e109296346688e8eefb2c365022f : SUCCESS in 1h 27m 59s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/75ec5daf98e34ca7baf9e7ebedd8492d : SUCCESS in 1h 21m 36s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4e4ba066381461fa6fe6c88cd455764 : SUCCESS in 2h 15m 54s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e75388d8b34843e6955f1ee04c21b9c3 : SUCCESS in 2h 32m 31s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef7195ec1f7b44418b9cb72cb173d939 : SUCCESS in 1h 23m 32s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/3ae1137386d9412d80f32327ad0e8e57 : SUCCESS in 1h 24m 51s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/165db74d27ee4561ab61b7677b3df65d : SUCCESS in 2h 56m 42s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6768b5f300be40b0b847b8d2c23a00dc : SUCCESS in 1h 55m 55s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6ef94f5fbd348f786dbd9616eaec54e : SUCCESS in 2h 36m 26s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e198fc94cd594cbfac9f04f0d1441e9f : SUCCESS in 1h 40m 10s\n- openstack-ansible-deploy-aio_metal_openbao-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f692f22d39774c08a3d8970e395fc116 : SUCCESS in 1h 10m 12s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/6419d0e7437f42189b10c4c683e93f36 : SUCCESS in 3m 06s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99b3d83d1da94f0cb42f82a37390c8eb : SUCCESS in 5m 45s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/8b7cc36ee3224cc2bf22ffb93b5a3962 : SUCCESS in 3m 31s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/c1314e6660484354b03c7c5bfe6a816d : SUCCESS in 5m 33s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e27ef36e46b44eb49e4dc4d16d6157d8 : SUCCESS in 5m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01ebfb00955e460283b0c7db9b860b7e : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ca28e9b2f04b3d2984a6293ab5aaaf410003fee0","revisions":{"2dbc3ced922d37d71030c8ef1bd41449bd627fd4":{"kind":"REWORK","_number":1,"created":"2026-09-22 14:31:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/1006741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/41/1006741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/1"}}},"commit":{"parents":[{"commit":"18fb4c37511e08fec329900bc1a573739bf3c4a9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/18fb4c37511e08fec329900bc1a573739bf3c4a9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 14:30:49.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 14:30:49.000000000","tz":120},"subject":"Address molecule intermittent issues","message":"Address molecule intermittent issues\n\nThis patch addresses two different issues that are present in our\nmolecule testing:\n- Rocky Linux can intermittently fail on facts gathering due to\nrestricted permissions to the /root, where ansible creates a tmp stroage\nof 0550 while systemd-userdbd takes a while to activate to fix it.\nThus, setting remote_tmp under /tmp fixes the issue.\n- Intermittent issues reaching Ansible\u0027s galaxy for molecule\ndependencies. This can be worked around by creating\\\ntests/requirements.yml, which is parsed by default with ansible-compat,\nand it does proceed with installation first.\nIf all dependencies are already present, molecule won\u0027t attempt\nreaching galaxy. We see a lot of failures recently on CentOS 10 Stream.\n\nChange-Id: I51966c00443cbbcc27449032a64cecc93ce10354\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/2dbc3ced922d37d71030c8ef1bd41449bd627fd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/2dbc3ced922d37d71030c8ef1bd41449bd627fd4"}]},"branch":"refs/heads/master"},"c23e9837df05e7c19faaf22d5f9f34eeb4650197":{"kind":"REWORK","_number":2,"created":"2026-09-22 17:18:03.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/1006741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/41/1006741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/2"}}},"commit":{"parents":[{"commit":"18fb4c37511e08fec329900bc1a573739bf3c4a9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/18fb4c37511e08fec329900bc1a573739bf3c4a9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 14:30:49.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 17:17:50.000000000","tz":120},"subject":"Address molecule intermittent issues","message":"Address molecule intermittent issues\n\nThis patch addresses two different issues that are present in our\nmolecule testing:\n- Rocky Linux can intermittently fail on facts gathering due to\nrestricted permissions to the /root, where ansible creates a tmp stroage\nof 0550 while systemd-userdbd takes a while to activate to fix it.\nThus, setting remote_tmp under /tmp fixes the issue.\n- Intermittent issues reaching Ansible\u0027s galaxy for molecule\ndependencies. This can be worked around by creating\\\ntests/requirements.yml, which is parsed by default with ansible-compat,\nand it does proceed with installation first.\nIf all dependencies are already present, molecule won\u0027t attempt\nreaching galaxy. We see a lot of failures recently on CentOS 10 Stream.\n\nChange-Id: I51966c00443cbbcc27449032a64cecc93ce10354\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/c23e9837df05e7c19faaf22d5f9f34eeb4650197"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/c23e9837df05e7c19faaf22d5f9f34eeb4650197"}]},"branch":"refs/heads/master"},"6d8a24ada3fad3ce2baf7d5fc41294de5533bc00":{"kind":"REWORK","_number":3,"created":"2026-09-22 17:22:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/1006741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/41/1006741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/3"}}},"commit":{"parents":[{"commit":"18fb4c37511e08fec329900bc1a573739bf3c4a9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/18fb4c37511e08fec329900bc1a573739bf3c4a9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 14:30:49.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 17:21:54.000000000","tz":120},"subject":"Address molecule intermittent issues","message":"Address molecule intermittent issues\n\nThis patch addresses two different issues that are present in our\nmolecule testing:\n- Rocky Linux can intermittently fail on facts gathering due to\nrestricted permissions to the /root, where ansible creates a tmp stroage\nof 0550 while systemd-userdbd takes a while to activate to fix it.\nThus, setting remote_tmp under /tmp fixes the issue.\n- Intermittent issues reaching Ansible\u0027s galaxy for molecule\ndependencies. This can be worked around by creating\\\ntests/requirements.yml, which is parsed by default with ansible-compat,\nand it does proceed with installation first.\nIf all dependencies are already present, molecule won\u0027t attempt\nreaching galaxy. We see a lot of failures recently on CentOS 10 Stream.\n\nChange-Id: I51966c00443cbbcc27449032a64cecc93ce10354\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/6d8a24ada3fad3ce2baf7d5fc41294de5533bc00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/6d8a24ada3fad3ce2baf7d5fc41294de5533bc00"}]},"branch":"refs/heads/master"},"edd881bbfbd4359e51b92e81d50830772ba20758":{"kind":"REWORK","_number":4,"created":"2026-09-22 18:21:34.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/1006741/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/41/1006741/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/4"}}},"commit":{"parents":[{"commit":"18fb4c37511e08fec329900bc1a573739bf3c4a9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/18fb4c37511e08fec329900bc1a573739bf3c4a9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 14:30:49.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 18:21:22.000000000","tz":120},"subject":"Address molecule intermittent issues","message":"Address molecule intermittent issues\n\nThis patch addresses two different issues that are present in our\nmolecule testing:\n- Rocky Linux can intermittently fail on facts gathering due to\nrestricted permissions to the /root, where ansible creates a tmp stroage\nof 0550 while systemd-userdbd takes a while to activate to fix it.\nThus, setting remote_tmp under /tmp fixes the issue.\n- Intermittent issues reaching Ansible\u0027s galaxy for molecule\ndependencies. This can be worked around by creating\\\ntests/requirements.yml, which is parsed by default with ansible-compat,\nand it does proceed with installation first.\nIf all dependencies are already present, molecule won\u0027t attempt\nreaching galaxy. We see a lot of failures recently on CentOS 10 Stream.\n\nChange-Id: I51966c00443cbbcc27449032a64cecc93ce10354\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/edd881bbfbd4359e51b92e81d50830772ba20758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/edd881bbfbd4359e51b92e81d50830772ba20758"}]},"branch":"refs/heads/master"},"ca28e9b2f04b3d2984a6293ab5aaaf410003fee0":{"kind":"REWORK","_number":5,"created":"2026-09-22 19:57:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/1006741/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/41/1006741/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/41/1006741/5"}}},"commit":{"parents":[{"commit":"18fb4c37511e08fec329900bc1a573739bf3c4a9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/18fb4c37511e08fec329900bc1a573739bf3c4a9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 14:30:49.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-22 19:57:42.000000000","tz":0},"subject":"Address ansible-galaxy fetching issues.","message":"Address ansible-galaxy fetching issues.\n\nIntermittent issues reaching Ansible\u0027s galaxy for molecule\ndependencies. This can be worked around by creating\ntests/requirements.yml, which is parsed by default with \nansible-compat, and it does proceed with installation first.\nIf all dependencies are already present, molecule won\u0027t attempt\nreaching galaxy. We see a lot of failures recently on CentOS 10\nStream.\n\nChange-Id: I51966c00443cbbcc27449032a64cecc93ce10354\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/ca28e9b2f04b3d2984a6293ab5aaaf410003fee0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/ca28e9b2f04b3d2984a6293ab5aaaf410003fee0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
