)]}'
{"id":"openstack%2Fopenstack-ansible~689274","triplet_id":"openstack%2Fopenstack-ansible~master~I954f56fb380e230b36516d6cb44fd372b892e8a9","project":"openstack/openstack-ansible","branch":"master","hashtags":[],"change_id":"I954f56fb380e230b36516d6cb44fd372b892e8a9","subject":"Mark opensuse-15 jobs as non voting.","status":"ABANDONED","created":"2019-10-17 21:36:09.000000000","updated":"2020-01-21 13:19:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"32e096809fca26d3bfd5fd7b021fc4e07f9c78a6","_number":689274,"virtual_id_number":689274,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-18 00:33:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-10-18 07:47:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 00:33:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8cfa91f0132c1dcc03bb99ceb4800d83db724473","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-17 21:36:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de405f696b5d6136f6704add8bd1bb4385d287b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 00:33:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f2e1772a9b6e435e8bc00f60688a93c3 : SUCCESS in 8m 10s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/f0e7443574a244a2985fdcfd3bfd05da : SUCCESS in 15m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23606f07eb294e589851a3a89bbba968 : SUCCESS in 9m 08s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/18bbe8c57a6e4e3bbcab5b16162c20ea : SUCCESS in 2h 53m 28s\n- openstack-ansible-deploy-aio_lxc-debian-stable https://zuul.opendev.org/t/openstack/build/b04dffabc72e44b4a5a49ffd8f64a071 : SUCCESS in 1h 51m 10s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77ca755d55754505865a4818bcb393b7 : SUCCESS in 1h 58m 29s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/cf5a3af8840a4a47a640849b4a9edab7 : SUCCESS in 1h 15m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b0396149b0e04293840d73e001ce8e25 : SUCCESS in 1h 36m 59s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7494692fe387474180fc1a9e2be41d85 : FAILURE in 51m 03s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d23e425c53524252a534dfc714895cf3 : SUCCESS in 1h 22m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e2e94b4439674bf38118e7824f688274 : SUCCESS in 1h 20m 46s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cc9583bae21b4215ba2ce2634208274e : FAILURE in 45m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87deb288ebf34f279d1194d3630d26a0 : SUCCESS in 1h 03m 41s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/924db9a57f604eddbca01c483e20bd60 : SUCCESS in 2h 44m 50s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c99f360458a34161967c6f92a6580661 : SUCCESS in 1h 33m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9d7f80ad6c5216f5b39e8871eedb37ecb5844871","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-18 07:47:35.000000000","message":"Patch Set 1: Workflow-1\n\nLooks like dirk had a fix for this","accounts_in_message":[],"_revision_number":1},{"id":"32e096809fca26d3bfd5fd7b021fc4e07f9c78a6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-01-21 13:19:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"49d04235eab65d8897e0a3065ba1582357118887","revisions":{"49d04235eab65d8897e0a3065ba1582357118887":{"kind":"REWORK","_number":1,"created":"2019-10-17 21:36:09.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/74/689274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/74/689274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/74/689274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/74/689274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/74/689274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/74/689274/1"}}},"commit":{"parents":[{"commit":"61a6c177a2ed851e9cf3e06bd838e98300870e0b","subject":"Merge \"Add unified linters check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/61a6c177a2ed851e9cf3e06bd838e98300870e0b"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-10-17 21:33:11.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-10-17 21:35:30.000000000","tz":60},"subject":"Mark opensuse-15 jobs as non voting.","message":"Mark opensuse-15 jobs as non voting.\n\nRepos for opensuse-15 jobs seem to be missing packages we need:\n\nhttp://paste.openstack.org/show/784551/\n\nChange-Id: I954f56fb380e230b36516d6cb44fd372b892e8a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/49d04235eab65d8897e0a3065ba1582357118887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/49d04235eab65d8897e0a3065ba1582357118887"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
