)]}'
{"id":"openstack%2Fopenstack-ansible~697981","triplet_id":"openstack%2Fopenstack-ansible~stable%2Ftrain~I2a5f88f9047559b31ddce2e28d82c5e3f9eee7a5","project":"openstack/openstack-ansible","branch":"stable/train","topic":"osa-train","hashtags":[],"change_id":"I2a5f88f9047559b31ddce2e28d82c5e3f9eee7a5","subject":"Set neutron_service_region in group_vars","status":"MERGED","created":"2019-12-09 11:13:22.000000000","updated":"2019-12-10 22:29:45.000000000","submitted":"2019-12-10 22:27:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697981-1576016843420-656d7ee0","meta_rev_id":"21f0880455bbeb2c46cf6a1091526dab4e9927bd","_number":697981,"virtual_id_number":697981,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-09 22:50:09.000000000","_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"date":"2019-12-10 14:31:58.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-10 22:27:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":2,"date":"2019-12-09 17:52:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-09 16:55:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":1,"date":"2019-12-09 17:52:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-09 16:55:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-12-09 17:52:38.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2019-12-09 22:50:09.000000000","updated_by":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"reviewer":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"state":"REVIEWER"},{"updated":"2019-12-10 22:27:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c404a752259a170d27f8a646358a78b0443746c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-12-09 11:13:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8f6adb45e38e4e1ea75cdd3e4d5536aae8c8a821","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-12-09 11:34:44.000000000","message":"Topic changed from osa-train-stable/train to osa-train","accounts_in_message":[],"_revision_number":1},{"id":"d090ff083a7ecd3ba174cde4dbee1cbb288720c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 13:54:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dc3e60ee30964e77b4c23f0bbd441921 : SUCCESS in 8m 07s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/5a6e6465b739443f802d85c018a91942 : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a9ae8443e8248abbbd9175e0a7b5397 : SUCCESS in 9m 56s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/3bad820defc849c0b0f7fe758c1852a1 : SUCCESS in 2h 31m 55s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/822b2a8b43ca46b691b66c89f31e67c2 : SUCCESS in 1h 49m 30s\n- openstack-ansible-deploy-aio_lxc-debian-stable https://zuul.opendev.org/t/openstack/build/4275967a7d9d4df5837ffe22d91a0336 : SUCCESS in 2h 16m 25s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e71a5bdaab1545ee9fa2505c2fbc0b36 : SUCCESS in 2h 02m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/864542a7fa974cee80706842223c5160 : SUCCESS in 1h 04m 23s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/7f7210f976834b8695cba46e2097c60c : FAILURE in 1h 19m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/99682ad873824f61839dc734a4d90196 : SUCCESS in 1h 38m 07s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ba951fd94f234e4d96fb5a037dd05ca1 : SUCCESS in 1h 22m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/861458cd22554ff1b1da16df9a7022b4 : SUCCESS in 1h 21m 58s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/830c33c3648a4e33a99d45efab132b5c : SUCCESS in 2h 34m 02s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c7f6206f1734ce08f15b79f8ceeda55 : SUCCESS in 1h 43m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4fd59d52149142c3bd7331261f56fd7e : SUCCESS in 1h 13m 12s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/dc22a739b1c24f0f8501388dad160fce : SUCCESS in 1h 01m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2398919b94e64c869a7c16117b076127 : SUCCESS in 1h 02m 57s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d06ba74b77914a4eac4c4509f1678030 : SUCCESS in 2h 17m 32s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8088643bed14500b75fec448dfd3d5b : SUCCESS in 1h 53m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d6e41cfede5a4cb37966a1493396afa4446c7de7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-12-09 14:04:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6832567bc208491067457eb7b19e442e860fb5cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 16:52:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fdf74c8d3fb944929f9b94da8d35ad9d : SUCCESS in 10m 21s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/b37a16cf555e4254a1725f5e0ea0dd01 : SUCCESS in 11m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e9c7d45a2c46d4878d8b97857701a6 : SUCCESS in 10m 16s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/42be1516d93f4fa6bd747ca190f6d017 : SUCCESS in 2h 36m 54s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/97e90e73c90e483f8a0439d34492a136 : SUCCESS in 1h 44m 10s\n- openstack-ansible-deploy-aio_lxc-debian-stable https://zuul.opendev.org/t/openstack/build/4c29a40c18444cdb9b88022ea9871366 : SUCCESS in 2h 25m 58s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39e8ea7c1b2145c59e11d9d2d31ec692 : SUCCESS in 1h 59m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2cedcb0468044894bc57e6a3fbe741e4 : SUCCESS in 1h 28m 19s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/4d4ffaeef6c945ab8099965d6c653f7a : SUCCESS in 1h 48m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c42ad28b8c034ebb8eaa5708c7a7bd3f : SUCCESS in 2h 01m 28s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a02e7e3f89564b44a1329870afd4b90b : SUCCESS in 1h 44m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/358703d56b5b4a44b3fc61d011119bbf : SUCCESS in 1h 05m 35s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bdb35793191f4b4c915f333cc2421ece : SUCCESS in 2h 28m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/278410d1de5d49d2a598a8d7c7f47bed : SUCCESS in 1h 42m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c87e23f9d62c431bb5c80116de4efdd4 : SUCCESS in 1h 09m 10s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/eaf6617a14564dc882fd905d32778541 : SUCCESS in 1h 00m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbff1ed05ebd4e358a3efcc188d5ad31 : SUCCESS in 47m 47s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a38158b445aa4aa1bf008d2260b5806a : SUCCESS in 2h 43m 53s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c6d3e8bf9d04d338823fbd2e08f311b : SUCCESS in 1h 48m 55s","accounts_in_message":[],"_revision_number":1},{"id":"0a576fef72295b13122b87c3e7a62037da3cdc3a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-12-09 16:55:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c6583883f0a6074fce390eaecf4ac7b15d123995","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2019-12-09 17:52:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cedd36e97cd5f18394772f1fd1b540faf4d8eed2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 17:53:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5bccf6ff1db62d5e05d41c05649f5475bf1da7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 19:13:48.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"453f4addfe41a3316ed117b3ea6829848f26ece1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 22:34:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1a647eff99bb4a9fa783285c8fadf013 : SUCCESS in 9m 07s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/5a1a8f6123e34d32a32915ee501eb750 : SUCCESS in 11m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc62332584594a55ac3dfacb51fa6642 : SUCCESS in 11m 28s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/731926f0f9644f448560e04d3ad44b9f : TIMED_OUT in 3h 13m 00s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/9487b883a2ef4898b719d06997d47a3f : SUCCESS in 1h 36m 47s\n- openstack-ansible-deploy-aio_lxc-debian-stable https://zuul.opendev.org/t/openstack/build/2609febf8aaf4b089bb1f95544d91349 : SUCCESS in 2h 15m 26s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14f8f5882b274ef097e59cdb7e91d1c2 : SUCCESS in 1h 45m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4efa2f2de4db4be6b1ff2691238f3381 : SUCCESS in 1h 25m 13s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/af0c5f5371f9438bab0fbd551e4eab2e : SUCCESS in 1h 31m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/20fea542e49f49a0886c38c0756556bf : SUCCESS in 1h 47m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5945e39f7b1544418f063a760e7596dd : SUCCESS in 1h 22m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26387e4fb31f45f0a375169e6be1edb3 : SUCCESS in 1h 07m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a515fcb61cbc4bb3b4c6c5d5618e285d : SUCCESS in 1h 54m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5cac5cc5ddf14c049c311657804fb538 : SUCCESS in 1h 16m 52s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7953f0c0ad824ff0a505032f311cc597 : SUCCESS in 52m 42s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17987e88bce6406b809697c0442a83f2 : SUCCESS in 2h 07m 05s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc4315f10f504200b959f17d14817589 : SUCCESS in 2h 05m 10s","accounts_in_message":[],"_revision_number":1},{"id":"8bd2e3d4b8d2654f93de6feefede44e47e382a2b","author":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"date":"2019-12-09 22:50:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"43b2a590650eadcf752c02aea3cc7eaeee920b57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 01:28:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bfe9adc079c1439385d6fa138b18a8cb : SUCCESS in 8m 26s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/3d18d51fa9a24188a9bd945f13a5e008 : SUCCESS in 10m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff3f00d4551d48428e98c8f3353cf88f : SUCCESS in 11m 46s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/d8f6454328784ce297231b7665692c73 : SUCCESS in 2h 25m 59s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/7a6b81eca5ba477c8d177f0c3225d266 : SUCCESS in 1h 42m 35s\n- openstack-ansible-deploy-aio_lxc-debian-stable https://zuul.opendev.org/t/openstack/build/f6ef7af82cbe4829a2a3fb3520cae426 : SUCCESS in 2h 33m 22s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af7d9254bcb84d6a9b0127f85466b328 : SUCCESS in 1h 50m 37s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5bf8daeeeb364fbc854e331b9e164a3f : SUCCESS in 1h 07m 06s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/86ca5f94aa6f483785f062ab95fb281c : SUCCESS in 1h 10m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ae5fb6ef247a4b53b09f2767749b782f : SUCCESS in 1h 39m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7effc7b4272c443d8b946840dfbe6035 : FAILURE in 1h 31m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0dba86b6bc444b27a6202967d3032919 : SUCCESS in 1h 18m 24s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0830ed6d862e41e9881abad239a44dec : SUCCESS in 2h 16m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/143b2fbb5881405e9617cfdc8d0a784a : SUCCESS in 1h 38m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7c373cc354be421da98c8874e44d1b6e : SUCCESS in 1h 07m 29s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/18d881090fa941f6aee7865ad61d4545 : SUCCESS in 1h 00m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5840962541d44f139fabea4fa909303b : SUCCESS in 49m 22s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec7ef77a27014d14824f385d74c02359 : SUCCESS in 2h 14m 27s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c245e75a138b48448b690a221d7bf0b8 : SUCCESS in 1h 45m 59s","accounts_in_message":[],"_revision_number":1},{"id":"7a532b8623d3bba558185ddaa091f658e5209efc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-12-10 06:03:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7a02f8e2df31caa48995c93180495f5883cc0f1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 09:24:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ce09f02c11704d43ab004710b6b4e74b : SUCCESS in 10m 32s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/33b79908ca0f4404a086cae5d7f3a997 : SUCCESS in 10m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77ee619d29d341648862c3142d674785 : SUCCESS in 12m 14s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/d1317dc548db449da77dc730de59e35d : SUCCESS in 2h 49m 25s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/58e73b71b24e46bea0af127a33c7c296 : SUCCESS in 1h 46m 59s\n- openstack-ansible-deploy-aio_lxc-debian-stable https://zuul.opendev.org/t/openstack/build/dc744eb8e1164169aab88eb934bb8ffe : SUCCESS in 2h 06m 23s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5287c8268c64997aa845a3e813a22ed : TIMED_OUT in 3h 09m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f9e37c22b6b343759b3827d41f497aec : SUCCESS in 1h 31m 15s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/d283754e4af54460a17dbcdeedafad29 : SUCCESS in 1h 15m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/88f9d9fe0b414c41bd536cc4370264ae : SUCCESS in 1h 34m 18s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9ed7a8a2acaf4701962943f02f64a00c : SUCCESS in 1h 20m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/074796f548c44220a02ec5a6068c0ec5 : SUCCESS in 1h 09m 00s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/76b4cb3e923b422ca864d47c97204e9e : SUCCESS in 2h 26m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99ebc41a6aca40feb89bcacadcfbe92c : SUCCESS in 1h 41m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/25f5ebd4ca2d4acb83c938a5e425d8bb : SUCCESS in 1h 13m 32s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/24592d13601c49e9a8e841b19858dbfe : SUCCESS in 58m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08e541f4d2d64dd6b4c430eb3c43d493 : SUCCESS in 51m 30s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3eaa002901b84462b16679db1d765c98 : SUCCESS in 2h 05m 50s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/551f5fd47e48446fad25fe393969f461 : SUCCESS in 2h 52m 47s","accounts_in_message":[],"_revision_number":1},{"id":"42f0dfe90f7b370dda01da952ae8e8cb8d9b7901","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-12-10 10:52:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8fee7eb84e536c7f7270775fb85fcbc20fcf0e5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 13:51:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/70e600bfae3c42baab0225c29cc0542a : SUCCESS in 9m 56s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/cacac80fa0834b06ae48c15cc1846f1b : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a588a7b7c3a243ea9e164537cb07a949 : SUCCESS in 11m 59s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/29fff630a1974589a4bf533b1d61a587 : FAILURE in 39m 43s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e59a7060685846d8b2714bcd7decd6b9 : SUCCESS in 1h 57m 38s\n- openstack-ansible-deploy-aio_lxc-debian-stable https://zuul.opendev.org/t/openstack/build/e20f4621df914229a125a49c29a4b0ad : SUCCESS in 2h 35m 08s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3443ce44080d491bb9180c6d280ae277 : SUCCESS in 2h 05m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/61b4ed038a6942468be1547904c41b30 : SUCCESS in 1h 05m 45s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/7ea557d04f774b36a9f37a003dd392f9 : SUCCESS in 1h 09m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4bd396d789374fafb4b6cdf372cf8dce : SUCCESS in 1h 50m 18s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a97de46056b649d78b644d2470983c55 : SUCCESS in 1h 25m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8077374a17474e19adbc59316b208cc7 : SUCCESS in 1h 04m 41s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8fc7d7cc794d41bbb856913100915f71 : SUCCESS in 2h 28m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/642abcc446da4abfa380c188b8b563bc : SUCCESS in 1h 40m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bfa5ef1a7dca41a6a33c09e7a3f98f70 : SUCCESS in 1h 03m 33s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fd96e6b6e0b64e6c9a04a2526451511d : SUCCESS in 1h 02m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/365b3b29a06840a4980ebfd785063685 : SUCCESS in 54m 30s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebc87f2925d14749990da20abc49bc91 : SUCCESS in 2h 20m 46s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f97954bfa2484729bdcadfe960cd3f86 : SUCCESS in 1h 53m 32s","accounts_in_message":[],"_revision_number":1},{"id":"081f317b809db9efca8b30bffd3a1b7ec03f3164","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-12-10 14:31:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"94a13d71730f19acc62e2a4dba131b8e0e2e3f72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 17:40:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0dddc435d3684897894cc937bd569056 : SUCCESS in 12m 21s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/0d4f6d3298a945d4bc78b1a101e74fac : SUCCESS in 13m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52d313234a7847759b7b0b72c75ff567 : SUCCESS in 12m 31s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/5fc3898747dd4e59b6c7d8f6bc900c67 : SUCCESS in 2h 59m 06s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/3fbb7eb0774b444fbf77bafbe8de01f8 : SUCCESS in 1h 39m 46s\n- openstack-ansible-deploy-aio_lxc-debian-stable https://zuul.opendev.org/t/openstack/build/4e85d1b848444187a74906ab47669639 : SUCCESS in 2h 36m 32s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a726727dc7aa401aaa0f7f6396676dee : SUCCESS in 2h 00m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9d90edb8dbc54325a48764ee42279b48 : SUCCESS in 1h 03m 57s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e9fe4ef9c2b84fdaa121dd3a8d14586f : SUCCESS in 1h 22m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7f99e3447cf24426a945275521f35c09 : SUCCESS in 1h 37m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b9b1c8c70fa941bb81100eea2e49bbae : SUCCESS in 1h 22m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d467f6203d84d9ba89dc45eff7dce68 : SUCCESS in 1h 22m 22s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c40d0c487bd64849855ac9de9c208382 : SUCCESS in 2h 23m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5567d09422940eab9b8a4de28c5f348 : SUCCESS in 1h 42m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7f4fdb956c7d4128a87eb3d4a9438d1e : SUCCESS in 1h 07m 19s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ea8ad09816204910b3612a1ec6dd0af2 : SUCCESS in 1h 05m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06634881d32e4047921265f0b86630e3 : SUCCESS in 52m 51s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0405b4bee01e4dad9972baa4e6904cfa : SUCCESS in 2h 06m 39s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2fca181e108f4c6fa71a5ed838c856a8 : SUCCESS in 1h 49m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9eb7c408bb9d61fb32882e4c3e1e3c270d501103","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 17:40:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aecfcfd33c0b85648b5848d33cd8b424e0c6547a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 18:42:18.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db443701f0c4a845f1a92dca5e3d933b6d091a5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 19:34:34.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ea524c654d06d255cb01eab31c872adb868a468","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 22:27:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"272eb56a76ff35917af1e99e2aa2304a9dd1d221","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 22:27:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b7b6e6d1498149398bb044e8614df2ff : SUCCESS in 11m 07s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/b83dcc916cc246a2baeb9ed7eca1ae19 : SUCCESS in 11m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c57ee2c27f4a4759bd064f5d973d2809 : SUCCESS in 11m 06s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/3ee587b8479d45aeb9f6c4cc571212c9 : SUCCESS in 2h 37m 58s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c7c7f447bff04363b278b2aab9dece79 : SUCCESS in 1h 44m 35s\n- openstack-ansible-deploy-aio_lxc-debian-stable https://zuul.opendev.org/t/openstack/build/3dd8515a3f8f423b9093afbb136c6e16 : SUCCESS in 1h 55m 03s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c080e5b3a29449d9fc4b9315cc05350 : SUCCESS in 1h 59m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5a0c8b5198ef466d9a00e496ca73cd4d : SUCCESS in 1h 10m 43s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/2bdbbaee48164426affe90635bcaecee : SUCCESS in 1h 35m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f721272380ad4b9dbea064730f6fd24e : SUCCESS in 1h 55m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5d14c0c0c9294f3d8bf2a61af98c1195 : SUCCESS in 1h 34m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aab7d21e6e8b4de38b11970794b7c7b8 : SUCCESS in 1h 35m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d16d299767614b1b8db5ad1ab1db9264 : SUCCESS in 2h 14m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ce1a3949a4794e33a319671537609327 : SUCCESS in 1h 26m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a80906d85c814b0f8387912967fff684 : SUCCESS in 1h 08m 45s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3654f17cf9d4e2c83b5deb9a3ff6691 : SUCCESS in 2h 50m 01s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bce0c85761142a7b37acf473e389779 : SUCCESS in 2h 14m 37s","accounts_in_message":[],"_revision_number":1},{"id":"21f0880455bbeb2c46cf6a1091526dab4e9927bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 22:29:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17999739f367413db3735588011851bf : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79426881e68614a11b663589878ac95c24ee5bfe","revisions":{"79426881e68614a11b663589878ac95c24ee5bfe":{"kind":"REWORK","_number":1,"created":"2019-12-09 11:13:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/697981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/81/697981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/81/697981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/81/697981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/81/697981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/81/697981/1"}}},"commit":{"parents":[{"commit":"8679849119a0d72859aa5dda7d92cc37bcd3279c","subject":"Merge \"Remove murano from os-infra_containers\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/8679849119a0d72859aa5dda7d92cc37bcd3279c"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-12-07 19:26:01.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-12-09 11:13:22.000000000","tz":0},"subject":"Set neutron_service_region in group_vars","message":"Set neutron_service_region in group_vars\n\nThe other roles wire \u0027service_region\u0027 to the role specific service\nregion variable. Do the same for neutron.\n\nChange-Id: I2a5f88f9047559b31ddce2e28d82c5e3f9eee7a5\n(cherry picked from commit 2e8474efbb5b43f4e54404e3974eb5365c25e8ec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/79426881e68614a11b663589878ac95c24ee5bfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/79426881e68614a11b663589878ac95c24ee5bfe"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
