)]}'
{"id":"openstack%2Fopenstack-ansible~709986","triplet_id":"openstack%2Fopenstack-ansible~master~I4f701ca5e7caa4d5ef182ca3eba523702bdae36a","project":"openstack/openstack-ansible","branch":"master","hashtags":[],"change_id":"I4f701ca5e7caa4d5ef182ca3eba523702bdae36a","subject":"Re-order gate cleanup pip package removal","status":"MERGED","created":"2020-02-26 08:47:19.000000000","updated":"2020-03-06 05:37:47.000000000","submitted":"2020-03-06 05:34:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709986-1583472862703-4545ba28","meta_rev_id":"a912f1dcb7c430739ba96402de6c14e0a3e4a045","_number":709986,"virtual_id_number":709986,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-06 05:34:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-04 19:52:48.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2020-03-06 02:53:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2020-03-04 20:36:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":1,"date":"2020-03-06 02:53:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-04 20:36:28.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-03-06 02:53:52.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2020-03-06 05:34:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57d71ac793530725a45442fe0d38fc3a0bdfdcdc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-26 08:47:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31de3dbd2232bc2f4b2be1803543cb3b1c36526f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 11:33:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e4cea22cf7af47599f4145cfaef0cfba : SUCCESS in 11m 26s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/f7317f053a9e4ca5ab8c94e0aa80b4ed : SUCCESS in 12m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7348e24674ee464dac7322c80817a23a : SUCCESS in 13m 28s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/b9f2ecd3636d43888b8f2e6668c24dc2 : SUCCESS in 2h 43m 56s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/832300b555f04cdda5f151c0b7398580 : SUCCESS in 1h 52m 08s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f6f591e197742e7b58268e968c0c53a : SUCCESS in 1h 54m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/eea3457a905b489eadcb24e73a2d2e6f : SUCCESS in 1h 08m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/adf43919c8174e639375f8900b62790a : SUCCESS in 1h 34m 27s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3fff9d51e430445dadb204d861876938 : SUCCESS in 1h 34m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9ef7453b4b94d459e42442e3fca9778 : SUCCESS in 1h 07m 26s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/090fb30b8e8d44e49b74f95939583606 : FAILURE in 1h 15m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/038b14d45fcf48548d59befa69acd664 : FAILURE in 1h 00m 22s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/40539401fbc74e899519938aa1f9b94e : SUCCESS in 1h 09m 43s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ef4a4d9c877a483dae2f66a65ab832b6 : FAILURE in 21m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/decfe3a9b7c042d39e0d9c4f335fbe51 : SUCCESS in 54m 56s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/366661eb674842138e53cf25a03c3eb6 : SUCCESS in 2h 22m 18s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3427169a59954572b21002e0dff64aa6 : SUCCESS in 1h 44m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a029b742ce27c2e30f288016468886a6d7787b58","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-04 12:09:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"75c7a29c1ca24ab642870131acd3e1eb946bea28","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-04 15:55:15.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1680e6266a80e427353bb51ac3ec1c5a7b5bb365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 18:53:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d92995c438034f929b58142967a40ff9 : FAILURE in 6m 56s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/fdc07deb118c4fe1b2885e4284a52b73 : SUCCESS in 17m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/682cb0bf46d24c7cb512e2abf8b1db85 : SUCCESS in 12m 47s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/afd3f41423864b319824b3340aa860db : SUCCESS in 2h 40m 14s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e96df46617bf45dfbbdfbc9ac59de868 : SUCCESS in 1h 31m 49s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cd648246f09b4a2aa57b2d69b14aabcc : SUCCESS in 1h 51m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fb2e1f0cd8e041fc8e16e92001576023 : SUCCESS in 1h 20m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bbadf7537db24a87966a84fe42f6f3fc : SUCCESS in 1h 42m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3367f3fe9d4845b98030e7c63752005c : SUCCESS in 1h 31m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5ecf98b21204657bb92cf64093a46f4 : SUCCESS in 1h 16m 25s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e64d1d9ec2ab45889d120848a0991333 : FAILURE in 1h 31m 00s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f02385bb9524aa6b466bc0d9ba7e2c3 : FAILURE in 1h 02m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/eea9e09eb36a4d6ab330ec257cbda6e7 : SUCCESS in 1h 26m 14s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e25a3eae16724273a86bec992ef2f61a : FAILURE in 34m 25s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0b1d36410144a75bfc4f2ec040d7ca3 : SUCCESS in 1h 06m 41s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15da3d1f83e04beeaa52a0cb32b1d788 : SUCCESS in 2h 40m 03s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4386deb34c1e4c1eb2ab5a9e4bde8b1f : SUCCESS in 2h 09m 39s","accounts_in_message":[],"_revision_number":2},{"id":"1a6a964c17dc0be197cca3e793680a2327c8f5d6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-04 19:52:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a68007e01f321cffc8f1900619ed0f43701abe71","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-04 20:36:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4c891ee9d7df21297750170ff0d0a828a75c3877","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 22:46:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/599d87dd0f39428588a107eecc1f3f73 : SUCCESS in 11m 01s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/682202b3cf464ef99be91ade3604bb39 : SUCCESS in 16m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c6884c746504123be6e516167752af2 : SUCCESS in 12m 41s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/12143d9ccbd8400d8512c053c3e952c0 : SUCCESS in 2h 34m 21s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/3fb4c4f09d4b482598da7de03cc025d7 : SUCCESS in 1h 40m 28s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecf3c67638c34fdea8b5033bbd94ca47 : SUCCESS in 1h 53m 45s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a1573f9db28344ce8442151c6befc542 : SUCCESS in 1h 11m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/54eaa64bf27b45b89525490d0e8d9b6b : SUCCESS in 1h 32m 31s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4be635564d814769b04ddbf6a1a30ab8 : SUCCESS in 1h 31m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/912e2a36a98c4bc0a196d53a8cca17e1 : SUCCESS in 1h 30m 38s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/df58ec05b70942f6b47d1948863690e2 : FAILURE in 1h 26m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bc69b20ecef45a08f37dcafade69b40 : FAILURE in 1h 11m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9c52f87fe65f4d00993eb09e5962f9b7 : SUCCESS in 1h 06m 24s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ef3186c77b6741f193b54269b9fa97a5 : FAILURE in 42m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc79363438044e41bc11ec710f71814e : SUCCESS in 59m 17s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/482b311ff9dc47b887d71f6b92e83025 : SUCCESS in 2h 42m 36s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b2a7dfa446e4852b42ec2f329932f48 : SUCCESS in 2h 25m 10s","accounts_in_message":[],"_revision_number":2},{"id":"b72237136059d1defeb1d9ebb965076205ff6c78","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-03-06 02:53:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1e5d67cae12fe8249badaa4d4f0e8d62482e7fe5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 02:54:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"556d6df41a4719f144f0de6fb8f4dd18316fb83c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 05:34:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7790c8ff012c09e51bf11e7a56e4211912d74327","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 05:34:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/006e5b5c86a24ddb91cf2cbc8d9bac17 : SUCCESS in 13m 23s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/ba9460805e0f4d7ba3a9da9f7f7921d0 : SUCCESS in 13m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a8f89fe9de9431cb0a6961acd349f48 : SUCCESS in 12m 24s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/08c5115ac65a46ad9b92bb26c67b5cb8 : SUCCESS in 2h 18m 45s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/043993a19c6e4052837433d757a95d8c : SUCCESS in 1h 54m 41s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fb898928fc744dab23685438447347c : SUCCESS in 2h 00m 10s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0c9b81df6fbb42a49747d9570383bc37 : SUCCESS in 1h 15m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/31fa55bdcee0401ead15fe31ce756bb5 : SUCCESS in 1h 30m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4bb3b85dee6145f19b1733959fbf76d3 : SUCCESS in 1h 23m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9aa88491981b439ebc536419832a311b : SUCCESS in 1h 08m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bc01135bff4e41d5b28dc319046ec3e9 : SUCCESS in 1h 07m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d09e0465e34b4037ba4d35ff3d5b251f : SUCCESS in 1h 02m 21s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95db366bf4304258ada98a666232101f : SUCCESS in 2h 35m 39s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f04bdc45177e4b1398d8ec5f0cade1d3 : SUCCESS in 1h 55m 20s","accounts_in_message":[],"_revision_number":2},{"id":"a912f1dcb7c430739ba96402de6c14e0a3e4a045","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 05:37:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1822be3fbe0c4130953010ae3daf96ba : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b04481c82cd595fe775f43b3cb5ae566bd145ca3","revisions":{"b296e3c409ab318cc080a1c4413a9a5d9fbcb6ce":{"kind":"REWORK","_number":1,"created":"2020-02-26 08:47:19.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/86/709986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/86/709986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/86/709986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/86/709986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/86/709986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/86/709986/1"}}},"commit":{"parents":[{"commit":"78adec035761454806982424bced3dad094cc3fb","subject":"manila: refactor playbooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/78adec035761454806982424bced3dad094cc3fb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-26 08:43:46.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-26 08:43:46.000000000","tz":0},"subject":"Re-order gate cleanup pip package removal","message":"Re-order gate cleanup pip package removal\n\nThe packages are now removed in reverse dependency order. The previous\norder would remove packages until a dependancy conflict would arise,\nthen would stop and the pip task would return \u0027changed\u0027 rather than\n\u0027failed\u0027. See [1].\n\nIn addition, the pip installation of tox is removed.\n\n[1] http://paste.openstack.org/show/789996/\n\nChange-Id: I4f701ca5e7caa4d5ef182ca3eba523702bdae36a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/b296e3c409ab318cc080a1c4413a9a5d9fbcb6ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/b296e3c409ab318cc080a1c4413a9a5d9fbcb6ce"}]},"branch":"refs/heads/master"},"b04481c82cd595fe775f43b3cb5ae566bd145ca3":{"kind":"REWORK","_number":2,"created":"2020-03-04 15:55:15.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/86/709986/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/86/709986/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/86/709986/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/86/709986/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/86/709986/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/86/709986/2"}}},"commit":{"parents":[{"commit":"78adec035761454806982424bced3dad094cc3fb","subject":"manila: refactor playbooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/78adec035761454806982424bced3dad094cc3fb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-02-26 08:43:46.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-03-04 15:55:12.000000000","tz":0},"subject":"Re-order gate cleanup pip package removal","message":"Re-order gate cleanup pip package removal\n\nThe packages are now removed in reverse dependency order. The previous\norder would remove packages until a dependancy conflict would arise,\nthen would stop and the pip task would return \u0027changed\u0027 rather than\n\u0027failed\u0027. See [1].\n\nIn addition, the pip installation of tox is removed.\n\n[1] http://paste.openstack.org/show/789996/\n\nChange-Id: I4f701ca5e7caa4d5ef182ca3eba523702bdae36a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/b04481c82cd595fe775f43b3cb5ae566bd145ca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/b04481c82cd595fe775f43b3cb5ae566bd145ca3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
