)]}'
{"id":"openstack%2Fopenstack-ansible~771146","triplet_id":"openstack%2Fopenstack-ansible~master~I5fe5df97a189a21015fd79c3789d2b13fa7f950a","project":"openstack/openstack-ansible","branch":"master","topic":"bump_osa","hashtags":[],"change_id":"I5fe5df97a189a21015fd79c3789d2b13fa7f950a","subject":"Bump SHAs for master","status":"MERGED","created":"2021-01-17 17:25:31.000000000","updated":"2021-01-18 19:38:37.000000000","submitted":"2021-01-18 19:33:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771146-bump_osa","meta_rev_id":"d393d5aec9626067fdc9a96326d3ee208e70a5fb","_number":771146,"virtual_id_number":771146,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-18 19:33:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-01-18 08:25:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-18 08:51:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-18 08:51:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-17 19:53:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-18 08:25:29.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2021-01-18 08:51:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"d3a5d86901c64eee503933112ee7bf58233f3832","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-17 17:25:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91a0c02b5da95dd868a919fe44064294ae2cdf1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-17 19:53:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2ed16070c7e84702a70d11bd10550447 : SUCCESS in 9m 04s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/de1a65f0e3194ae0b486eaf1f530a2c2 : SUCCESS in 11m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f895c19b143d431e8b9d913cfd327249 : SUCCESS in 9m 47s\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/d78df9dc1d424e99a983e724f49d9af3 : SUCCESS in 2h 16m 16s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/a7a14774bb2f4984a8fb0e490d472749 : SUCCESS in 1h 30m 35s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57edd50fc5e5478996463b1c5a5a9357 : SUCCESS in 1h 33m 01s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/209c6cf87a68417383af18e05c9e7f53 : SUCCESS in 1h 28m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7619a6f83d654bb99714b0905149d3e1 : SUCCESS in 1h 05m 30s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ce81677e37af4c0595e0073b90bf2c2b : SUCCESS in 1h 21m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63865cf817dc46f0a5d073eb510e46e9 : SUCCESS in 1h 10m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be911e4e07364ba486a08dafa82b7061 : SUCCESS in 1h 09m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e0b7799d917f4ccca32d24780f380249 : SUCCESS in 1h 57m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/747a831136844ebf86e7b049b57e9f2e : SUCCESS in 1h 38m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f9372ea1602a4368b389d58af6715a43 : SUCCESS in 1h 03m 15s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c3dfcc1e2274525a858c1bda975d18d : SUCCESS in 56m 01s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5403ac75af54afb9e84d3318bcd90c5 : SUCCESS in 2h 02m 19s\n- openstack-ansible-deploy-aio_proxy-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fff1ed17932e4ba9a18ae561e4cf3620 : SUCCESS in 1h 38m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6111b69a296641ddba6056af4a1048fc : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e08ddc53badab26d44c3841e30954f92034b2d63","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-18 08:25:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fcdbfc99be3ee5417104fe7456673419d2a63929","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-18 08:51:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3300649ad661b61a468822e0bf06d56689ca4a68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 08:51:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea5ce45367e10af7afc91ee466b5711b10702074","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 10:26:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/81a1660ae507470fafcdc9641c5c3a3b : SUCCESS in 11m 13s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/0a75b0dc333e49c4b5f353f67a9cc477 : SUCCESS in 15m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef82e3204cf249a297eb6406cf34f0b4 : SUCCESS in 12m 15s\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/d728af48aff64c6e8b8dd4730207a47e : FAILURE in 1h 22m 44s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/53508aeb0fcf4ed6a8cf91937446fa2e : FAILURE in 1h 25m 28s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7de6f2abcc824e2e8936cda275cc2db0 : FAILURE in 1h 20m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c3b70da31f804711baaf051f85e6322d : SUCCESS in 1h 16m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ac58416fda4d4bb397296e57c62b687e : SUCCESS in 1h 19m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4eeb7846e664b4ba68b97fa2d3e9095 : FAILURE in 1h 17m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c590bf9d3264cfd9626476ebbc04325 : SUCCESS in 1h 05m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a16ee9614a564b56aa4af05ef5c4659e : SUCCESS in 1h 01m 24s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/461b4552d6944064a6c1b688f11f3443 : SUCCESS in 55m 26s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dda4ee02051941caa739d21f2fca9907 : FAILURE in 1h 24m 11s\n- openstack-ansible-deploy-aio_proxy-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cac980bd63f44ea094e6ff0d5eba766b : FAILURE in 1h 20m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5624bd2f32d4fd7a4ef1dd35e7b01c4 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"924b89315abb6d0c3af15b6cb916e36d011a8a9f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-18 14:36:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"110b44e07aa3776c66b0fbc311ebc9d257615589","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 17:18:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/981127c19ea240989329d1ecb806250f : SUCCESS in 10m 47s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/87a2677a6bfa4f12862e2cd38a792dd7 : SUCCESS in 20m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d97957a021f44f5abeaea50c54902b8 : SUCCESS in 12m 33s\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/3f5487268c5c4c33a365c24e61f8389a : SUCCESS in 2h 00m 08s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/6b5ce008652a4c3dbac97218cf20f85f : SUCCESS in 1h 38m 06s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c27ec2e72b184b7abe34c317aa74c43c : SUCCESS in 1h 40m 25s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1baae08a235c4766a33ef9d1904559c5 : SUCCESS in 1h 27m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b329fb11fd2c49baa40dcb69f058af38 : SUCCESS in 1h 11m 08s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f8c3f5e6e66b4e60ad444982bc74fe17 : SUCCESS in 1h 11m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ffe4ad6d1dea46af9b30bea11bc0edc4 : SUCCESS in 1h 13m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea75492f58ea4bb399f3a6d953f9082d : SUCCESS in 1h 13m 11s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e5d2a6c684a841c383559c3a18960503 : SUCCESS in 2h 18m 42s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a9669b7ade449f18251c1ec8e57ad86 : SUCCESS in 1h 41m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0787373d0e9e4d48a32560dd63b9ca5d : SUCCESS in 1h 08m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9b93191f80640399bc79729a54a39dc : SUCCESS in 1h 01m 44s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d73154abe0c45d3a274a4de5c8e3e01 : SUCCESS in 2h 27m 54s\n- openstack-ansible-deploy-aio_proxy-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72c6e22a94934587ba6fdc8a8e367bb5 : SUCCESS in 1h 34m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25f49dc9b2324b95ace7e5ab167ae8f2 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"893dffe008d43ad1418e0e275b782a4990906475","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 17:19:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e606044058b857992ad72241f8b5cf779aa2f88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 19:33:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/250026c4b6a64fcfb491c5fd43adeaa4 : SUCCESS in 13m 08s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/60b9c33e4e0049f6a0d86bfe127b0763 : SUCCESS in 15m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4163fc9f95794b16a161a194b1b0433b : SUCCESS in 10m 01s\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/434a4d065d434908a5d1f65d7f7542b8 : SUCCESS in 1h 55m 26s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/054a74f6dd034913a945bfd6ef8eeaf0 : SUCCESS in 1h 44m 38s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcd1ec3ada59494c927a02adad65d33b : SUCCESS in 1h 32m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/84d60f32a9334b2982a080ff9aaac4cd : SUCCESS in 1h 09m 42s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5d7f2e0a5b7246dfab04338d661a4fdb : SUCCESS in 1h 20m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7800eebdda1645b1bb1f4ed6711b12e6 : SUCCESS in 1h 16m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed217a86bf9d43dfbe9eb658c6341325 : SUCCESS in 1h 11m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7cbf3ab8b1d24c73ab4eeb4edb7e67b4 : SUCCESS in 1h 06m 10s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd6e4dbd7cf74800af5bc58d2384a66e : SUCCESS in 56m 23s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4cb83f45d034fd3aae65eaabeee605a : SUCCESS in 1h 56m 59s\n- openstack-ansible-deploy-aio_proxy-ubuntu-focal https://zuul.opendev.org/t/openstack/build/871b55f064b542fc9968f7bc0e1902de : SUCCESS in 1h 34m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/484d6b3cd44c48a785d110bc7c2e2966 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"412d445275363dd760be95fedd18b8a166ec390a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 19:33:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d393d5aec9626067fdc9a96326d3ee208e70a5fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 19:38:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a63ff0ee56d44daaacd67b4ca6c117f3 : SUCCESS in 1m 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d757167d85764eecbca73f2b69647e2c : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"35c826e32ba6ba9885e72c00495734ed6cb778c5","revisions":{"35c826e32ba6ba9885e72c00495734ed6cb778c5":{"kind":"REWORK","_number":1,"created":"2021-01-17 17:25:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/771146/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/46/771146/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/46/771146/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/46/771146/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/46/771146/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/46/771146/1"}}},"commit":{"parents":[{"commit":"ab8ed8b7c9ca3d5618f8c820746e02521cdbeaaa","subject":"Gather minimal facts to use ansible_distribution","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/ab8ed8b7c9ca3d5618f8c820746e02521cdbeaaa"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-01-17 17:25:26.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-01-17 17:25:30.000000000","tz":0},"subject":"Bump SHAs for master","message":"Bump SHAs for master\n\nChange-Id: I5fe5df97a189a21015fd79c3789d2b13fa7f950a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/35c826e32ba6ba9885e72c00495734ed6cb778c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/35c826e32ba6ba9885e72c00495734ed6cb778c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
