)]}'
{"id":"openstack%2Fopenstack-ansible~771147","triplet_id":"openstack%2Fopenstack-ansible~stable%2Ftrain~I42998a65f71019c2c1fbc0f4d250a862ed4ad847","project":"openstack/openstack-ansible","branch":"stable/train","topic":"bump_osa","hashtags":[],"change_id":"I42998a65f71019c2c1fbc0f4d250a862ed4ad847","subject":"Bump SHAs for stable/train","status":"MERGED","created":"2021-01-17 17:25:47.000000000","updated":"2021-01-19 01:11:07.000000000","submitted":"2021-01-19 01:05:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771147-bump_osa","meta_rev_id":"183de6f76d6cc9b38e671e4564c6102f8bc81551","_number":771147,"virtual_id_number":771147,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-19 01:05:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-01-18 08:26:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-18 08:54:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-18 08:54:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-17 19:57:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-18 08:26:30.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2021-01-18 08:54:24.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"405fde5ab7beea391fe634f41559e4767a50c260","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-17 17:25:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f64193159fe9daef311ea7af092d551df20b174f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-17 19:57:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0c1ed90980a04c21a2a8227dcc55d733 : SUCCESS in 10m 02s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/71d318046a754cc8b7d94ccf243fb4f1 : SUCCESS in 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2f45fcbcc764f748d529cb2141da229 : SUCCESS in 10m 48s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/abb406b7344d49a0800fa7d1de1f1652 : SUCCESS in 2h 22m 42s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/29eac1a242ae413b9c5108dedf247c56 : SUCCESS in 1h 09m 05s\n- openstack-ansible-deploy-aio_lxc-debian-stable https://zuul.opendev.org/t/openstack/build/938d573b48c24c109abb2ff1ae279eeb : SUCCESS in 2h 10m 39s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae1360eaab91408eb950def2f90b23c3 : SUCCESS in 1h 39m 42s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/185a3f00e61c42b0bb0343098b6a90b2 : SUCCESS in 44m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/fe00b58d323841fba9998a7353193c55 : SUCCESS in 1h 16m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/96ac58368d404817a4e9a7548fec1357 : SUCCESS in 1h 26m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4c10eb76754f4862b15641680e4c1dbc : RETRY_LIMIT in 15m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2db4728b2f8a4e98a931e050b90840a4 : SUCCESS in 1h 06m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5a84e0de0cc14031a35b093d8b1cba57 : SUCCESS in 1h 33m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea9963b4417d48f3b35e017b73545743 : SUCCESS in 1h 13m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cb10ffcb8f954abe860df792b6b0e9f3 : SUCCESS in 55m 27s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d81bfc4ced274c23b9349d6c261e0726 : RETRY_LIMIT in 12m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcc614ff35494dfead81abc4ea074e1f : SUCCESS in 50m 54s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e37955ab4dc48599e54b5c3845d837c : SUCCESS in 1h 56m 05s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/285dd9de4f0d4863af0b6753c68396b9 : SUCCESS in 1h 25m 08s","accounts_in_message":[],"_revision_number":1},{"id":"cc06ec9134c1cf4ad5c60624cf5d461f05fc5e79","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-18 08:26:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ec7bfc2517197bb912dd85319754b0ea9a868480","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-18 08:54:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"71fc4d4913f9d475e81bff89e8e766e2eefd3551","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 08:54:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ee39454431fef0e95f2e3fdb74e820e5ab9eae2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 14:09:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7203b005e79c4b46b7b865a7151d4492 : RETRY_LIMIT in 4m 10s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/0a4828285f1346bb8dcbbff52c0cd8d4 : SUCCESS in 12m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/351a3edffb264bb38c8fd43965a23a99 : RETRY_LIMIT in 1m 56s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/8018ee3472bf457d8bfdea1b29fc1fd5 : SUCCESS in 2h 47m 28s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/6cf8ebda30314c77afb5d0786cb30aa1 : SUCCESS in 1h 36m 06s\n- openstack-ansible-deploy-aio_lxc-debian-stable https://zuul.opendev.org/t/openstack/build/4f113ba8dc664af18fb33edc330370be : SUCCESS in 2h 12m 15s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5453f6ed85b44239d9debedfc2a9d37 : SUCCESS in 1h 33m 21s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/31bb31ebf921474586cb35e39c4023c9 : SUCCESS in 1h 20m 48s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/5f977077fedf4d19873b0f25a823c2fd : SUCCESS in 1h 19m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7dae0481da354000b200ee44210d3686 : SUCCESS in 1h 17m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/73a8ed41f3864e2291f1c3391e6660ff : RETRY_LIMIT in 17m 35s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54b1e385f59342de9b1d3b15e0155ba9 : SUCCESS in 1h 07m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f551526943a648a7b14196adc62781a7 : FAILURE in 10m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0d8885a47b80477ebab2c3571e735d0f : SUCCESS in 1h 01m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bd28962691146bdafa7ae49369edd8b : SUCCESS in 56m 32s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/729c598e041e4738872fd9789f3e2137 : SUCCESS in 1h 59m 37s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20c3cfa4254e419dae8b327cf1ab987e : SUCCESS in 1h 47m 58s","accounts_in_message":[],"_revision_number":1},{"id":"5a9d2cc46c32f2974b85fdac24148ba55347fe74","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-18 18:11:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b276cb3246e61c73aa7fd7fbc9b5825976b20aa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 22:14:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1dd967869bb94cc28f0925b3793393b2 : SUCCESS in 9m 48s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/740a714fe7534faa84b8a65ede59bc3c : SUCCESS in 14m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/140a8892d16e48228e5a67ef3eaab873 : SUCCESS in 11m 51s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/7315f9d4a66e4eb99bcb583ff563a34b : SUCCESS in 2h 22m 21s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/fefac145e7a84de3a907a06cefdb87bb : SUCCESS in 1h 31m 57s\n- openstack-ansible-deploy-aio_lxc-debian-stable https://zuul.opendev.org/t/openstack/build/1191f4982cc94c428fcfdc51b8e71511 : SUCCESS in 2h 17m 18s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97f8bd857989452c964401e2826efc58 : SUCCESS in 1h 36m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1d92741b315b40bca4244ee8932ad6e0 : SUCCESS in 1h 02m 28s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/caddb5dac1b14d628bc4db9fdf8013dc : SUCCESS in 1h 12m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c7f03d04e4e54ff69b8889af60f9e3b6 : SUCCESS in 1h 41m 14s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6087897201b841cfa31a45dc923f5166 : RETRY_LIMIT in 21m 43s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57aa1438b05b4cd4a516037cf0f17935 : SUCCESS in 1h 08m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/90e15ca91bfc436b813bb713a0f39cf0 : SUCCESS in 2h 26m 42s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/797343c9dec343259601868272650d67 : SUCCESS in 1h 26m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f5eda808185749e6a79738f188a34c52 : SUCCESS in 58m 23s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5b3a7eee942845fd9f70f2d0cac4a5c6 : RETRY_LIMIT in 16m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1becaed294024d548324ec3c7534c957 : SUCCESS in 53m 08s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5be5a7746444317bc2a199a20aec69c : SUCCESS in 2h 18m 46s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33eb3363c4ac4ff6ad318bbe2b56ade1 : SUCCESS in 1h 38m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4d3cac13d196a8354dd3150f81dd4c4d0c65198a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 22:15:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28ff6bd7295d57468270a6a46129dfaf85e17a20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 01:05:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/aac8e4f6e37d432eabcd542b7268e9eb : SUCCESS in 13m 22s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/a8518897dda543ed93620a0d25f588e9 : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2ff06c8986449e880c46807e4472bfa : SUCCESS in 9m 17s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/ca67981c2e16408e831d655ebf675fd3 : SUCCESS in 2h 41m 16s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/883483a95d08470dae6425cf58e3e455 : SUCCESS in 1h 54m 39s\n- openstack-ansible-deploy-aio_lxc-debian-stable https://zuul.opendev.org/t/openstack/build/d4db72ccf1a9426e903604fba73d5517 : SUCCESS in 1h 34m 26s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e3996689e994c928de22b7244bf7952 : SUCCESS in 2h 02m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/de2f4e72c57e4061bfb4999e4dca571b : SUCCESS in 57m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/88f3585857744021b146ca5851a303e6 : SUCCESS in 1h 10m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/397cee371b454a0cb72695fb7dd2f202 : SUCCESS in 1h 33m 05s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0dd55d2404df4aa5a7822cdb4a62194e : RETRY_LIMIT in 15m 51s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/752aa6e8bdd2475b8084211985d1a59c : SUCCESS in 1h 07m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4f4c4fb2dc647a1bea52f73b738fd5b : SUCCESS in 1h 22m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e1dc593cd9e846908b88305331770464 : SUCCESS in 1h 16m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9b62759275a41c3a267d176c5dc4cf7 : SUCCESS in 57m 14s\n- openstack-ansible-deploy-aio_ceph-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/833166cd867a4daea5349ec517a64a59 : SUCCESS in 1h 55m 56s\n- openstack-ansible-deploy-aio_proxy-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/226381d1ad704579861629881b4c1312 : SUCCESS in 1h 40m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5507ce03ecf70efd1c8ffdcc73f282950152707e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 01:05:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"183de6f76d6cc9b38e671e4564c6102f8bc81551","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 01:11:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73fb4d42567c4feb993b662de9a41a5c : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12ff92c6f6621b1592f3061eb4264b7dfde0c8b6","revisions":{"12ff92c6f6621b1592f3061eb4264b7dfde0c8b6":{"kind":"REWORK","_number":1,"created":"2021-01-17 17:25:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/47/771147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/47/771147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/47/771147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/47/771147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/47/771147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/47/771147/1"}}},"commit":{"parents":[{"commit":"a2397d2946d81c14d05778f6687440d28ff004c7","subject":"Bump SHAs for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/a2397d2946d81c14d05778f6687440d28ff004c7"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-01-17 17:25:43.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-01-17 17:25:47.000000000","tz":0},"subject":"Bump SHAs for stable/train","message":"Bump SHAs for stable/train\n\nChange-Id: I42998a65f71019c2c1fbc0f4d250a862ed4ad847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/12ff92c6f6621b1592f3061eb4264b7dfde0c8b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/12ff92c6f6621b1592f3061eb4264b7dfde0c8b6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
