)]}'
{"id":"openstack%2Fopenstack-ansible~771148","triplet_id":"openstack%2Fopenstack-ansible~stable%2Fussuri~Ief0ee81a11b7385d9eb6bc498ecef9eab147d898","project":"openstack/openstack-ansible","branch":"stable/ussuri","topic":"bump_osa","hashtags":[],"change_id":"Ief0ee81a11b7385d9eb6bc498ecef9eab147d898","subject":"Bump SHAs for stable/ussuri","status":"MERGED","created":"2021-01-17 17:25:48.000000000","updated":"2021-01-20 23:16:42.000000000","submitted":"2021-01-20 23:14:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771148-bump_osa","meta_rev_id":"d181cee09e3d4397942d5c58fa0a972c9722e38f","_number":771148,"virtual_id_number":771148,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-20 23:14:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2021-01-20 18:32:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-20 07:28:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2021-01-20 18:32:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"CC":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-17 20:42:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-18 08:26:49.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"CC"},{"updated":"2021-01-18 08:55:13.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-01-20 18:32:53.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"209d38381b095d8359fe464b890218f46c4271e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-17 17:25:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"768255ebe4daeeb666efbeb10c4153ec12fc3d69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-17 20:42:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/96547a14a19d49018743228284f7b1b3 : SUCCESS in 10m 11s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/6e249cb511104cf5baa335046ffec098 : SUCCESS in 10m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87521455e5aa4db6bae967555033c8f4 : SUCCESS in 11m 38s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/31d964524cd645e9bb5e06e6a3156d1a : SUCCESS in 2h 07m 01s\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/e62d3b6667244e78b6bcfc2c30d0e963 : SUCCESS in 1h 52m 00s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/ce744cbf67e04754ae667d738d67ddfa : SUCCESS in 1h 21m 39s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce8b56c43c5f45b2a4774cbc342699f5 : SUCCESS in 1h 39m 31s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1d7fc3264e64e128b3857f461309524 : SUCCESS in 1h 45m 25s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f31dd1c31b32462e92aeffe7b9662960 : SUCCESS in 1h 18m 58s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bcc964e5d68c4d81b3be46914c90a1a7 : SUCCESS in 1h 01m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/07be5120519c4efa927e5aa2d00681d9 : SUCCESS in 1h 05m 26s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3557ea3fbb274a2abf7961432475939a : SUCCESS in 1h 13m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77813b23752b4f9fbb1c40d163ebd36d : SUCCESS in 1h 02m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b181565e7bac44558b942abce1391b61 : TIMED_OUT in 3h 01m 44s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/49e44f458e7f4c27b636eae7fa39b07d : SUCCESS in 2h 14m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37d3ff675cf74d47993d1d2985a5e47f : SUCCESS in 1h 39m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/36adacfc7be54b25ba6d23dd136cdcc6 : SUCCESS in 55m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e88444cb71d04632bec4fbe44280ba87 : SUCCESS in 39m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b73140d8c084346864c69ada30b03cf : SUCCESS in 52m 32s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2da038bb0a894d9894c3aebd9a27479a : SUCCESS in 41m 24s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4ee003ebe9347feb260752868f3b8b3 : SUCCESS in 1h 53m 40s\n- openstack-ansible-deploy-aio_proxy-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c76fe99c3f604be5b428c651edf45387 : TIMED_OUT in 3h 09m 42s","accounts_in_message":[],"_revision_number":1},{"id":"2f325d5a6831487ad10594536cf3c293ffa7b878","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-01-18 08:26:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2c5fe1a5081fe49fc0faf434f7532c6797d8d4aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 10:32:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5dacd34ce7ac4b7ea7445d95f6bd8f91 : SUCCESS in 8m 34s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/11a57690a1e746ab8c1f2adb50f7faba : SUCCESS in 10m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/733ce10b8f6a42718931bd089f856ee5 : SUCCESS in 15m 35s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/f0c4739d663f4172a8be2ecb26c63d02 : FAILURE in 1h 45m 24s\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/e113b7cd80f741dbb7642bc34a9af41a : SUCCESS in 1h 46m 50s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/6af2f24e53eb4aa7a5bf7f5b7211bc01 : SUCCESS in 1h 25m 58s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed435c330c3d462b808e01b48e6f830e : SUCCESS in 1h 46m 56s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d7bcbe996ab4a7fb5dbf7aac3b5efc9 : SUCCESS in 1h 43m 15s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d3d82c6baf04659bc6a3d2d7a10222b : SUCCESS in 1h 34m 08s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e8c13267ce2642d38a00439942b618e0 : SUCCESS in 1h 08m 26s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e56f5e3bdd154debbc2c9a1c20ec2e4b : SUCCESS in 1h 30m 24s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8258a5c609124748967caf76b9a10fb7 : SUCCESS in 1h 10m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5302ca9f4974d0f9b0b60ef2d23bf6c : SUCCESS in 1h 05m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2ad56dd362848b980008ba064c75b5f : SUCCESS in 1h 04m 02s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d99b4e9058de4c7aa390e001b6481465 : FAILURE in 1h 50m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a27ec80c6f8b4320ba29663ecb774650 : SUCCESS in 1h 43m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/461fcba0286d4e8a85ca8562b7f57ed5 : SUCCESS in 57m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/37ce0e5a405342de8e57c520d4ae8ace : SUCCESS in 53m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d07eb66d91514ccb94acfe96c426e2c3 : SUCCESS in 53m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e34460d673f34566bba6e45c6979bf04 : SUCCESS in 55m 32s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0654f43027a24fd1b143109e6efa9075 : FAILURE in 1h 58m 46s\n- openstack-ansible-deploy-aio_proxy-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2aaed63ff7f24550a522a78237308164 : SUCCESS in 1h 35m 51s","accounts_in_message":[],"_revision_number":1},{"id":"22ebb51053d4aee414d554114c10606a846ead5c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-18 18:00:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d7751d6a4c6512e548d087e20fb827d7928838eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 21:37:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4c57436c88b54943ae3afb04790378f9 : SUCCESS in 10m 38s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/c1e881240a0940a884c502e978da1a01 : SUCCESS in 17m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c34a9d36d54d71880af4b488ae8a99 : SUCCESS in 11m 37s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/064d891c0e524a3ea22a5f8634e8a8e3 : SUCCESS in 1h 51m 57s\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/cca59b96ed2b4967bda4c4279baec9c4 : FAILURE in 2h 07m 55s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/a689bdf2c61141a8a6cc3991e37e75b7 : SUCCESS in 1h 42m 37s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d667571ddf104289bf0f5658a77dac3b : SUCCESS in 2h 04m 10s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43d8062697494a25b6f4dba1d5795237 : SUCCESS in 1h 49m 56s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ae6f107d5354f13a0c2bffafbbb02da : SUCCESS in 1h 54m 42s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/47b67a02e9d14023b00de87ef9eb03f1 : SUCCESS in 1h 06m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e44bd26d355b45e1a0799a8f479264b2 : SUCCESS in 1h 27m 48s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2df421cfba244d23a0d95c52f67254dd : SUCCESS in 1h 12m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b9d040092194f17b19b83245c7cc876 : SUCCESS in 1h 11m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3255c91504e4eb69d129410f44a59a8 : SUCCESS in 1h 03m 02s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2d62bc0b3be4468ea41e877fdb5ecfb6 : SUCCESS in 2h 07m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd6f0af06ab44f62837cc904bcf61d52 : SUCCESS in 1h 39m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/094db48fc3574e388a957278ec38e850 : SUCCESS in 1h 03m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a15a51718b5949c68f4376d0d0b28d70 : SUCCESS in 53m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45a2a047df3e4d16b5ff48bda09ce004 : SUCCESS in 52m 56s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07b65d19038b4968bb20d8f10fc2a141 : SUCCESS in 48m 34s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cfde7797cad40c69178bc116d04eb71 : SUCCESS in 1h 54m 57s\n- openstack-ansible-deploy-aio_proxy-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f96aa3aeaf44a83808e2bd1f59d2a0e : SUCCESS in 1h 31m 53s","accounts_in_message":[],"_revision_number":1},{"id":"da8213c0defefc0cac369c30857759607634d5f2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-19 15:36:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"043e77a7767d15bc52aa77b2f23789f2adfe26b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 19:39:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a1e6409827dd444785dd2d23c2f7e2e9 : SUCCESS in 12m 23s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/7b53f9c067a74fdfae51d6c436c46b8a : SUCCESS in 14m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/562c21efd41f42d4bfb33c79f059e2b0 : SUCCESS in 10m 44s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/1c1bbb1ca2a34c538c33267597ad98a9 : SUCCESS in 2h 07m 02s\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/03e7667ef96b467195f9cb0dc8984036 : SUCCESS in 1h 56m 11s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/63eed149d13d41999e0c33b2ba4a6b2f : SUCCESS in 1h 41m 50s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/890e0fb1142c4f52907a1cc0c1f33a63 : SUCCESS in 1h 31m 02s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64afbb1122034ade8d0796b0fa282b31 : SUCCESS in 1h 27m 27s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb5713c3443f4c918ddfc52f3cb9ffe0 : SUCCESS in 1h 36m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ac0cd322d0a34f2199842c689294a9ab : SUCCESS in 59m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a1f57769f3de45698a0b5f2bde1d2713 : SUCCESS in 1h 12m 42s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a36a90dd9a124bc097b9ad7dafe48044 : SUCCESS in 1h 12m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/156a68c828ed4215a8a3b07c41e8373d : SUCCESS in 1h 12m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68d51146e3cc41148b811c93594b2bd4 : SUCCESS in 59m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b08f635130a2493187301777aed1635a : TIMED_OUT in 3h 03m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/728b8b6b927b45519e664766d248532d : SUCCESS in 1h 48m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/729ada1391154bc5979dff31e0843443 : SUCCESS in 1h 08m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b8d5a28142d3490fa147f5ff6c20dc65 : SUCCESS in 49m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bfbb1440782946968f0a8e151fd7d5b9 : SUCCESS in 54m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1071987b2b1b484d864749637c5b4c49 : SUCCESS in 1h 11m 29s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d2815c1984f49d6bb311d079ef10ffa : SUCCESS in 2h 01m 04s\n- openstack-ansible-deploy-aio_proxy-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b9dced379a9439db93e0e552e342b5d : SUCCESS in 1h 28m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0d75b4350f9967bb417b785222bf582131912ac8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-20 07:28:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"361f3bf8a36fe3c029e7455f03effb608f313ec3","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2021-01-20 18:32:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"61a115caf73d8f8fb7dcdc316c27b954dea37ffd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 18:33:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ffcf6576327c5c5d2daa3c1dcb74082108fa7c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 23:14:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/02b99c6e18894dba8cf928649d56083a : SUCCESS in 11m 58s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/d0a1502a044545b6b075ca25e58523d5 : SUCCESS in 14m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1baf07194f2498183e43659f3bd8d53 : SUCCESS in 10m 12s\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/7bf489b9515a43b9a7b6554e5722cfb6 : SUCCESS in 2h 07m 49s\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/daf8a52f88cd43ca8da31e2da27e3296 : SUCCESS in 1h 47m 00s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/3f05a14502f342eba846380a3bcfc1bb : SUCCESS in 1h 32m 31s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b22c5e83699f48e5a05585b316e504df : SUCCESS in 1h 48m 42s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7376f5d04dd34df3b4ff336b7d6fa757 : SUCCESS in 1h 34m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2a06a2243ff44c34bc706b6719a4e70f : SUCCESS in 58m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0205dc4a4fca4776907eac72ffc2a551 : SUCCESS in 1h 18m 08s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4d03ac8b2ea143cf8e5c9b22ec0b6a8c : SUCCESS in 1h 06m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c51412b06874f419937e6e405e863da : SUCCESS in 1h 02m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/427a6c6458604211b873f933489f901d : SUCCESS in 57m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bafb642a06554687a46ac1ff98e9280e : SUCCESS in 1h 04m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/866d5e86a7ca4b80a94f280580aad2db : SUCCESS in 51m 45s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a928d8c1f87742989ed6a9d5b6009e75 : SUCCESS in 51m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0184b8b3ec2475eafcdffa19c56a152 : SUCCESS in 56m 00s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc572fe921d242559e52f32bacc24fbb : SUCCESS in 2h 12m 56s\n- openstack-ansible-deploy-aio_proxy-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f717bf2755a44ba792bf5f813f321950 : SUCCESS in 1h 41m 11s","accounts_in_message":[],"_revision_number":1},{"id":"349cda6aae1bcf592925f9b610d0d37fa483f40d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 23:14:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d181cee09e3d4397942d5c58fa0a972c9722e38f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 23:16:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84da4e5ae0294366be987a45a841f26d : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e383e97aa398b128e26157c933ed9e32a663d018","revisions":{"e383e97aa398b128e26157c933ed9e32a663d018":{"kind":"REWORK","_number":1,"created":"2021-01-17 17:25:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/48/771148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/48/771148/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/48/771148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/48/771148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/48/771148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/48/771148/1"}}},"commit":{"parents":[{"commit":"4599e4fe80823c16926c3e33dd0e61265cbe5bd1","subject":"Merge \"Bump SHAs for stable/ussuri\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/4599e4fe80823c16926c3e33dd0e61265cbe5bd1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-01-17 17:25:44.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-01-17 17:25:47.000000000","tz":0},"subject":"Bump SHAs for stable/ussuri","message":"Bump SHAs for stable/ussuri\n\nChange-Id: Ief0ee81a11b7385d9eb6bc498ecef9eab147d898\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/e383e97aa398b128e26157c933ed9e32a663d018"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/e383e97aa398b128e26157c933ed9e32a663d018"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
