)]}'
{"id":"openstack%2Fopenstack-ansible~888517","triplet_id":"openstack%2Fopenstack-ansible~master~I337796c8dab4eeb1552971c270f0c83809241199","project":"openstack/openstack-ansible","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-08-21 20:15:08.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"34653":{"account":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"last_update":"2023-08-21 22:51:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I337796c8dab4eeb1552971c270f0c83809241199","subject":"Update SHAs for Gnoochi and plugins","status":"MERGED","created":"2023-07-14 05:41:54.000000000","updated":"2023-08-21 22:53:04.000000000","submitted":"2023-08-21 22:51:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888517","meta_rev_id":"fd4fca8eaf151f8b60bce0e277b412e35c7dc857","_number":888517,"virtual_id_number":888517,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-21 22:51:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-08-18 13:31:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2023-08-21 20:14:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":1,"date":"2023-08-21 20:15:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2023-08-21 20:14:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"}],"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-14 08:10:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-15 08:39:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2023-08-15 18:29:52.000000000","updated_by":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"reviewer":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"state":"CC"},{"updated":"2023-08-18 13:31:33.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2023-08-21 20:14:30.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"}],"messages":[{"id":"9ad48e07086cca6e7b9afe9f7e7ba726bf3c2e79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-14 05:41:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7299c54556252e7da6cf0c4135aa58f0e50eb8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 08:10:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a5c50512d5747499e2fbd98911a3bd8\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d99b8b6884d4c609c1b3144684eecec : SUCCESS in 14m 53s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/0438c0ade2f245678af3560f857f5d92 : SUCCESS in 17m 16s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/f987b8c88dab46f4b164305f51b89b9e : SUCCESS in 11m 33s\n- openstack-ansible-tox-inventory https://zuul.opendev.org/t/openstack/build/286918db05e54cf6ae76de5dd4e9edaa : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d1bce958da84bad813ae1f1da32ed00 : SUCCESS in 9m 02s\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/f30c00350ea141819b532c1e2d646e10 : FAILURE in 36m 12s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/95839ee0b46c4631b75de2df181fcd5c : SUCCESS in 2h 20m 08s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/fc6de1f64d53499392bde7edf23f9ff2 : SUCCESS in 1h 39m 27s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/071856921273440181854ca5dbe9b8e8 : SUCCESS in 2h 06m 32s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/411b8ccb6d4541ef883f641df4af168f : SUCCESS in 1h 55m 02s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b7e56dca8c174700a1dcce36ed979440 : SUCCESS in 1h 13m 01s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/964145455da4434985f5d9bcd2567f72 : SUCCESS in 1h 31m 25s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ba5c1411a5bc4f7b8499c958dde89440 : SUCCESS in 1h 44m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c69a24f58dae4733954647091f54cf47 : SUCCESS in 1h 41m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13a4aec218d547948b4745da175e78cb : SUCCESS in 2h 15m 21s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/909cae1a647241aea307e9b4e32cf95f : SUCCESS in 1h 25m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b43e4212f618479f9d78943bc3b1c81e : SUCCESS in 1h 31m 01s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/17e8d949fe7d46c8acfe266d88d45d72 : FAILURE in 2h 00m 37s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe743c3335b94f5b9d1115fc1a78d5e0 : SUCCESS in 1h 49m 14s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bf7d41a295da4425bcef6c5698d52a60 : SUCCESS in 1h 02m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4d7750549f74c3ca8ea78af696d8edb : SUCCESS in 1h 01m 59s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60fc3fae8aa54b6d98d1214f4eb5f4ae : SUCCESS in 2h 18m 27s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd7a14880d3b4bcf97f4e061e637af2a : SUCCESS in 2h 05m 10s\n- openstack-ansible-deploy-aio_proxy-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47bfad599e554f7db69071663bf7b504 : SUCCESS in 1h 38m 13s\n- openstack-ansible-deploy-aio_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/735753eb79284c06adc90a6ccab3faf3 : SUCCESS in 1h 59m 51s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a48bbad13e3442929f35fccd0c6e56bc : SUCCESS in 1h 14m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3ada92ae45bb892c65080c2cafb08a3438677a39","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-08-15 08:39:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"107c12ecddfea6e5720f1ec1a647a797a1085623","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 11:54:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2098cff378e49f5a34b9fb15f7002e9\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca7fcba71ccd4bf3a76abb3f75113f43 : SUCCESS in 16m 33s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/5d3682c4510d4d209d857800cbacab9c : SUCCESS in 18m 53s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/4b71e55923044cc68c7f26d29a2b2be4 : SUCCESS in 16m 00s\n- openstack-ansible-tox-inventory https://zuul.opendev.org/t/openstack/build/4745a99324be4a47aaab94d27f3802b1 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0947f7ea1f8a4f2dbcd4ac8f88c3e4e6 : SUCCESS in 7m 29s\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/e5220e609d054ccc85220c28bc4fe4e1 : SUCCESS in 1h 55m 06s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2af2c2fe553645b1befd5fb8f52562ad : SUCCESS in 1h 54m 31s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/414cc030eb424d479de311c178dc5444 : SUCCESS in 1h 30m 24s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1713977d9d4b4a8a98db005eb3a9c8dc : SUCCESS in 1h 47m 45s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/59c6c8de5bb847bdb163563c3ef3721b : SUCCESS in 1h 07m 45s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/f61de32428e2405a9ab0f9bbb2432e77 : SUCCESS in 1h 14m 05s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/09c6c4562ef84b8ca41e16d72c8a84c1 : SUCCESS in 1h 13m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/501cc9aff2884e3aa548b4341941dff3 : SUCCESS in 1h 05m 50s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0d372d128bc542acbb9214acac71d392 : SUCCESS in 1h 16m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/883c25f06f0744108f5e5436032e4a70 : SUCCESS in 1h 29m 30s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8afcea872f284aa69b7e4fca0fd4e811 : SUCCESS in 2h 33m 56s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46418d6512c34327814cf8c967f830d9 : SUCCESS in 2h 00m 30s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/67d4ba805ead4ca692f5aab3a2a71138 : SUCCESS in 1h 00m 37s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd95807bed5a4bbab45c51fb71548413 : SUCCESS in 1h 03m 08s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a98b321b02044fac8714a19c8e08a78d : SUCCESS in 2h 08m 31s\n- openstack-ansible-deploy-aio_proxy-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f492a2f5f20471da4fcfbee445dbb37 : SUCCESS in 2h 03m 54s\n- openstack-ansible-deploy-aio_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f45ce820faa0486eacfd89d7c755408c : SUCCESS in 1h 59m 41s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e44448a9a5b480e8715f24e2d93f39e : SUCCESS in 1h 11m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e370ce43825292dbeb1fd7b963243286e44989d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-15 14:55:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"74dd21f54a52acb3e158abd3bcd8a24ac40acef6","author":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"date":"2023-08-15 18:29:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b7fb066644590af057afd4c3a5249a56990fb95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 20:38:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8275f504067b44f7af5ed716868ab274\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/227c012df5ef4dd9bf47ff05188b8dbd : SUCCESS in 12m 41s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/c1cd62b6e15d438cbe12c1434aa324e5 : SUCCESS in 15m 13s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/00b0d3e787fe4ec5912aebf718580b0e : SUCCESS in 10m 57s\n- openstack-ansible-tox-inventory https://zuul.opendev.org/t/openstack/build/049bfa553f2f474d86ea760fce29593b : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5939870e93144e91bb06fa08cde44ced : SUCCESS in 9m 08s\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b83256a6e93464aabd54cc56b977293 : SUCCESS in 2h 16m 04s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f26dbb97ba8947da8c141be1e1384022 : SUCCESS in 2h 49m 20s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/49f8228f81b4454298918ed4666f47a4 : SUCCESS in 2h 03m 06s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9799314d267e4524b991f90bdd57c95c : SUCCESS in 2h 08m 23s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6efa10e08e474a8396aaaab26867e539 : SUCCESS in 1h 37m 15s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a62b20a722b4c138fe089fe24ae02f7 : SUCCESS in 1h 50m 43s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b4e5fc8fdc74edfa852cb7f5ca038b9 : SUCCESS in 1h 49m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70c112065e4546a58f0303a92197d2a2 : SUCCESS in 1h 44m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/85b611002dac44f594d135d0ef338b5f : SUCCESS in 1h 50m 56s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89dd204a8cad4c58ba8f3718d0b9d114 : SUCCESS in 1h 40m 55s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6c6846b0c1b64396a372ac1f6dd9bcf2 : SUCCESS in 2h 15m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a59ac13f744a4518bb2fc868bfd13aaa : FAILURE in 22m 53s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/397c7c7b8ced4482bbdf466186ff18ca : SUCCESS in 1h 03m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d22b58ee320d47678fade4d3a86d823c : SUCCESS in 1h 05m 10s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b3f95de1e3440229041dd1b11e8985d : FAILURE in 2h 04m 20s\n- openstack-ansible-deploy-aio_proxy-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79783d88e180405bb7e877d6f8bef872 : SUCCESS in 1h 50m 34s\n- openstack-ansible-deploy-aio_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/358bb2f63a124c0b806bd612494af7ae : SUCCESS in 2h 21m 01s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f76fce61e9d045db8c032b45b44c8050 : SUCCESS in 1h 42m 32s","accounts_in_message":[],"_revision_number":2},{"id":"49368425604d5d84908474c17cce28964b4b92cf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-16 09:41:10.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c22b97cc1b9bdaca6e66fd70f8c6efb4fcd462c4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-16 13:34:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb7c449ba8119997f9524d8e8c5136bcdae1a5b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 16:32:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98efaa0dd2864a4294994a0b678a871d\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48a8f29880ab4738b1ad66b6c320e2a5 : SUCCESS in 17m 35s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/1fa99f1a3d594cbba6ca935cbcfe1d15 : SUCCESS in 15m 32s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/6fb56b5fd5814056a1469391359d6bd8 : SUCCESS in 12m 24s\n- openstack-ansible-tox-inventory https://zuul.opendev.org/t/openstack/build/9f04b8fee31e4abfb9236fa2aab43483 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06045ebb0eee48228cfd57f20ab5af96 : SUCCESS in 9m 36s\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3ce19a1621740e494aa7bcef2ddd655 : SUCCESS in 2h 12m 14s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9fa9d408ad524f3fbf523be82a24e459 : SUCCESS in 2h 01m 28s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/0be1ae28a2c144b1905293d00b4319fa : SUCCESS in 2h 05m 29s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0384e261e5d4e74a2b9c3ee0dfa2a1c : SUCCESS in 2h 19m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bf08b61902cb4d6aadfd88b272a0cff5 : SUCCESS in 1h 19m 38s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/43945574016c4e5cb9a36568e89a7f71 : SUCCESS in 1h 34m 46s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/de4c303e9bbe449b8d9cd4a1c7758a5a : SUCCESS in 1h 39m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfa732cb68c9495bbb16578c9ea1ec71 : SUCCESS in 1h 42m 46s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3ef8dde61fbb4d559fcfc6c88e9dca81 : SUCCESS in 1h 39m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05cdd3eaf23249c5a660861e0e782b77 : SUCCESS in 1h 37m 32s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bf625cdfcb4b481c91013f1b75570b5c : SUCCESS in 2h 27m 10s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a291c40f793a49f6802e0afbadbfda29 : SUCCESS in 2h 18m 27s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/662705a46bb94e8790fcc048284f2fd1 : SUCCESS in 1h 07m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b6011613cca468ba901137aa94065e5 : SUCCESS in 1h 04m 46s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73482ad70f014f149120c16ab28b953a : FAILURE in 2h 39m 18s\n- openstack-ansible-deploy-aio_proxy-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a171d665c1fd407e9f36fcdd4932f0a9 : SUCCESS in 2h 14m 48s\n- openstack-ansible-deploy-aio_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39180ace9b594760bd24edd70b11d58a : SUCCESS in 2h 01m 49s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8153c6e38fa14be7b096c7f6d4005686 : SUCCESS in 1h 48m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a984d88ba2627c117326776ce712b1864bb65bbd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-17 15:21:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"29445a6c85556b6e142960f0861d342fb3e47716","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 21:15:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f977bf3326e84336a4a5285fbe46d105\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0380b02a8204e40b3f6a9c21022bc61 : SUCCESS in 18m 30s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/01c0b10c22d14e4e913c041f8e77b8ae : SUCCESS in 15m 07s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/7cf05a6c8ed449418f0fe659127df084 : SUCCESS in 11m 42s\n- openstack-ansible-tox-inventory https://zuul.opendev.org/t/openstack/build/5b95a2e8e2544110aaba1ad59f1eb77d : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3c75e51db6e4df8aacb647e0f81f822 : SUCCESS in 8m 39s\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/74d36a1dbfaf421d84940f78880625b1 : SUCCESS in 2h 33m 28s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/959b3bd55c564d50be225344cfbf1639 : SUCCESS in 2h 06m 55s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/cf9af5620ffa479493df615e0299bdc5 : SUCCESS in 1h 55m 19s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4cce38c1558b49f4bd55ab8f5e2868da : SUCCESS in 2h 08m 17s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/cbe44e2b33374f0ebc27ffb92b8e26f3 : SUCCESS in 1h 38m 54s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/da44a9e8b66946878a2fd5d4e8ac35a2 : SUCCESS in 1h 38m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/19bef954e26a4972b3d067784a5ca314 : SUCCESS in 1h 50m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8bedfad808940d7a801eb2434b4930a : SUCCESS in 1h 40m 45s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dcf8026b018b4910993be2b4ee55b7e4 : SUCCESS in 1h 47m 26s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/141be087716c406db605c01fab3c3b4c : SUCCESS in 1h 30m 56s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3c513036dbff4fd48b4ed945b427a9a3 : SUCCESS in 2h 28m 23s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/405aaedaa6164ab6a5113ab5e2ccfdde : SUCCESS in 2h 16m 20s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9957786bbc3c4dcdac7cacfecf421fcd : SUCCESS in 1h 02m 11s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/404e47afe69f4fc1af523e7986fcf4f7 : SUCCESS in 1h 06m 46s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1645d1d5628944d5a7b096575acb6d17 : SUCCESS in 2h 23m 39s\n- openstack-ansible-deploy-aio_proxy-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/647c66414e7043c287001e0e7265008e : SUCCESS in 2h 08m 43s\n- openstack-ansible-deploy-aio_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/980b18806e5e49d9a70451285791d6fa : SUCCESS in 2h 18m 25s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3241a351fe34d6b929196e14a716b30 : SUCCESS in 1h 37m 32s","accounts_in_message":[],"_revision_number":2},{"id":"3e5f4262ee4cae2903445d43747322d8c99f90c3","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-08-18 13:31:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"032c1c3398589fd753d62fba9fd7343232ac0dc3","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2023-08-21 20:14:30.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"489561b7ea756ab4f64f10565cfeb9594366e814","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-21 20:15:08.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b00738f40e0c86e3b2690a1ba293b840db36657f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 20:17:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8f3cbacfa98df5f648958f67ce827dcf07380940","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 22:51:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cb23fc7dfc0460a9783325d7c79f5e5\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85ffe10cc03742e19080ddf6c1eb361a : SUCCESS in 14m 34s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/7334b1ec6aee4a6d9b3a95a485315e27 : SUCCESS in 14m 39s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/21dd4ab4f12d4805a2121dc452153800 : SUCCESS in 17m 11s\n- openstack-ansible-tox-inventory https://zuul.opendev.org/t/openstack/build/78c28e9b5de44ba0b480e3482f463266 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47dd05a9c1c44da395a8301abfdd3fea : SUCCESS in 9m 04s\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4f010b7f59994a3295a8c7a1577131c4 : SUCCESS in 1h 59m 50s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a46cecfec8674aa0b81eef8284feadd2 : SUCCESS in 1h 41m 43s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99e5963c4bfa4df5b7bed8f16974879e : SUCCESS in 1h 43m 31s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7636f857bb7645cb80e000eee2e9e642 : SUCCESS in 1h 24m 05s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/30f9ea70c0cb4391ad4c5fbc4d2ae375 : SUCCESS in 1h 37m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cece72c8f709410fa19aa5b905408d59 : SUCCESS in 1h 25m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d2c5bd83bcb14a37892c8ad1861209cb : SUCCESS in 1h 38m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2152b4947b944f99aeb3058bcf53c63 : SUCCESS in 1h 44m 40s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d1ef44ac5114002b16211cce8d9d98c : SUCCESS in 2h 09m 20s\n- openstack-ansible-deploy-aio_proxy-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b780681105f433bb948f2563e7e9bf3 : SUCCESS in 2h 07m 59s\n- openstack-ansible-deploy-aio_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bac3461518aa430c80e95c79b7cc1d15 : SUCCESS in 2h 09m 34s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9da7db4316fd48f9aa1bd750b90fb511 : SUCCESS in 1h 30m 48s","accounts_in_message":[],"_revision_number":2},{"id":"2983732660ae156f2decd6e963b0e7d7fa4abad0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 22:51:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fd4fca8eaf151f8b60bce0e277b412e35c7dc857","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 22:53:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03f42264b1a04af896cec0fb84768879\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e336306e79e24092af6a66ef01032d39 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"45afeaefbafdd393348d87bdd59d8905912805f4","revisions":{"b84d623af1c2df40e675a37f8d4be1103dddc0bd":{"kind":"REWORK","_number":1,"created":"2023-07-14 05:41:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/888517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/17/888517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/17/888517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/17/888517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/17/888517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/17/888517/1"}}},"commit":{"parents":[{"commit":"ea0bbd2c582b313fc7174cc714a52e17e27e6e14","subject":"Merge \"Use include_role in task to avoid lack of access to vars\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/ea0bbd2c582b313fc7174cc714a52e17e27e6e14"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-14 05:41:46.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-14 05:41:46.000000000","tz":120},"subject":"Update SHAs for Gnoochi and plugins","message":"Update SHAs for Gnoochi and plugins\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible-os_ceilometer/+/888476\nChange-Id: I337796c8dab4eeb1552971c270f0c83809241199\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/b84d623af1c2df40e675a37f8d4be1103dddc0bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/b84d623af1c2df40e675a37f8d4be1103dddc0bd"}]},"branch":"refs/heads/master"},"45afeaefbafdd393348d87bdd59d8905912805f4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-08-15 14:55:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/888517/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/17/888517/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/17/888517/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/17/888517/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/17/888517/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/17/888517/2"}}},"commit":{"parents":[{"commit":"d1e30257ae0c818780684fe77e1b34ba4dd0dc40","subject":"nova/haproxy: fix typo in detection of \u0027serialconsole\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/d1e30257ae0c818780684fe77e1b34ba4dd0dc40"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-14 05:41:46.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-08-15 14:55:12.000000000","tz":120},"subject":"Update SHAs for Gnoochi and plugins","message":"Update SHAs for Gnoochi and plugins\n\nNeeded-By: https://review.opendev.org/c/openstack/openstack-ansible-os_ceilometer/+/888476\nChange-Id: I337796c8dab4eeb1552971c270f0c83809241199\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/45afeaefbafdd393348d87bdd59d8905912805f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/45afeaefbafdd393348d87bdd59d8905912805f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
