)]}'
{"id":"openstack%2Fopenstack-ansible~894449","triplet_id":"openstack%2Fopenstack-ansible~master~I0f6da707fe9490b4209e9e4ae141f5914035cd3f","project":"openstack/openstack-ansible","branch":"master","hashtags":[],"change_id":"I0f6da707fe9490b4209e9e4ae141f5914035cd3f","subject":"DNM: Test ara 1.7.0rc2 with openstack-ansible","status":"NEW","created":"2023-09-09 21:48:15.000000000","updated":"2023-09-10 00:21:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b134ead28e8bec5076ff457dc580303f39188356","_number":894449,"virtual_id_number":894449,"owner":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-10 00:21:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-10 00:21:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff3ef1fcd3b6ba972f2c7b3b3ccc51fed80ba849","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"date":"2023-09-09 21:48:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b134ead28e8bec5076ff457dc580303f39188356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-10 00:21:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5edea93ea7b04e20a07805cea025d168\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef7812f6913e435ba7d8260b07510207 : SUCCESS in 10m 44s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/b0b02b4f5f3c41c791dc158cd1ca3c33 : SUCCESS in 14m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8ef26380a7e49ffa4f696fe11d7bc88 : SUCCESS in 8m 35s\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbe3c9ed6984498cb5227272e7c2023f : SUCCESS in 2h 14m 16s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a0a6441c8d8e4706999c361255503c30 : SUCCESS in 2h 31m 08s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/8494359b98224ff6a0a49758d72a05f8 : SUCCESS in 1h 54m 52s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1295016a66a54109a484b5bf98514bcb : SUCCESS in 1h 51m 26s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7dbd2e83f2644812a3354568f5e8a92d : SUCCESS in 1h 34m 04s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/47a268f6bdf843fb94c7068130597354 : SUCCESS in 1h 40m 50s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ef343f8356914905ba27942ad8940faf : SUCCESS in 1h 35m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce3427695222489c845e4f4452809a7a : SUCCESS in 1h 34m 10s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/628e645b8d424df0bb2c41ed11c073e0 : SUCCESS in 1h 47m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b39f671b8bb7480d9a534f4d094f072f : SUCCESS in 1h 09m 39s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/35817f933c5742089912c9dcb18a27d3 : SUCCESS in 2h 22m 31s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9da2d281340c40c1a73f2543b1f98861 : SUCCESS in 1h 41m 58s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/219433720dac4f4998e048f8d27b996e : SUCCESS in 1h 05m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1b76d909e1341308f6660b59fed4a4f : SUCCESS in 50m 51s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43819d840d5d42a5b016da159b299344 : SUCCESS in 1h 30m 46s\n- openstack-ansible-deploy-aio_proxy-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/928ff383e4ef4f409d124b3a8e37869c : SUCCESS in 1h 24m 43s\n- openstack-ansible-deploy-aio_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e311becd1d54408bc9be00d3dde7272 : SUCCESS in 1h 29m 06s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afe9528b80364492a42eb1d574375926 : SUCCESS in 1h 08m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2fef8eccfe4bd058413fa9db73c348a295af96e7","revisions":{"2fef8eccfe4bd058413fa9db73c348a295af96e7":{"kind":"REWORK","_number":1,"created":"2023-09-09 21:48:15.000000000","uploader":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"ref":"refs/changes/49/894449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/49/894449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/49/894449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/49/894449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/49/894449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/49/894449/1"}}},"commit":{"parents":[{"commit":"bf6ba53fef07aa32c51fe448280e9c3c5d601db9","subject":"Merge \"Bump SHA for openstack-ansible-plugins collection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/bf6ba53fef07aa32c51fe448280e9c3c5d601db9"}]}],"author":{"name":"David Moreau Simard","email":"moi@dmsimard.com","date":"2023-09-09 21:47:52.000000000","tz":-240},"committer":{"name":"David Moreau Simard","email":"moi@dmsimard.com","date":"2023-09-09 21:47:52.000000000","tz":-240},"subject":"DNM: Test ara 1.7.0rc2 with openstack-ansible","message":"DNM: Test ara 1.7.0rc2 with openstack-ansible\n\nChange-Id: I0f6da707fe9490b4209e9e4ae141f5914035cd3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/2fef8eccfe4bd058413fa9db73c348a295af96e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/2fef8eccfe4bd058413fa9db73c348a295af96e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
