)]}'
{"id":"openstack%2Fopenstack-ansible~913418","triplet_id":"openstack%2Fopenstack-ansible~stable%2Fvictoria~I91b8cf26d568d8b20d927e4723232de735407f21","project":"openstack/openstack-ansible","branch":"stable/victoria","topic":"osa-eom","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-04-11 13:40:27.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-04-11 13:40:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I91b8cf26d568d8b20d927e4723232de735407f21","subject":"Switch SHAs to EOM","status":"MERGED","created":"2024-03-15 15:36:35.000000000","updated":"2024-04-11 13:42:13.000000000","submitted":"2024-04-11 13:40:27.000000000","submitter":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913418-osa-eom","meta_rev_id":"d30bfc80bd4c23ea2752ff7a88d43c28bbeffaee","_number":913418,"virtual_id_number":913418,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2024-04-11 13:40:27.000000000","_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-04-11 13:20:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2024-04-11 13:40:27.000000000","_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-11 13:40:27.000000000","updated_by":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"reviewer":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"state":"REVIEWER"},{"updated":"2024-04-11 13:42:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"ed356c43d3201a2671ea8c758b1127e56f6b7325","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-15 15:36:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4662da5823c9396ed891ad84a8bc1f44fa9560ed","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-19 15:12:14.000000000","message":"Topic set to osa-eom","accounts_in_message":[],"_revision_number":1},{"id":"782e3c58348d8e7e7261745533fa5839a66c14ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-22 10:46:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f62f45153182a030b425198b45b79a7bb7206807","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-28 17:09:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a27d24404bad81187858f50c29616f992a687408","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-03-28 17:09:48.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8a588317de9ead130ecd2593d0ace6eab2198dd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-11 12:26:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0bb37dda0a6b37758950d9e5faf37bac9c2788dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-11 12:37:51.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5a1dab53458404001921fc9dc0a11064e4296e6d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-04-11 13:20:37.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"5c54a2cdb44e515ad2ffd125b735cf5da036c8bc","author":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"date":"2024-04-11 13:40:27.000000000","message":"Patch Set 6: Verified+2 Workflow+1\n\nforce merging to fix zuul config error issue","accounts_in_message":[],"_revision_number":6},{"id":"588711cf9d1a82fb3cf505255ee70305e8826665","tag":"autogenerated:gerrit:merged","author":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"date":"2024-04-11 13:40:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"d30bfc80bd4c23ea2752ff7a88d43c28bbeffaee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 13:42:13.000000000","message":"Patch Set 6:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5397829435154add891cc32a093ac0b4\n\n- promote-deploy-guide https://zuul.opendev.org/t/openstack/build/de4c8a403ce14b709ca833081e4cbad1 : FAILURE in 24s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdbe1eb7599e41749ed5c3400a1ccb97 : FAILURE in 31s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"dc728b77a888a61deb61bcb28d5a68fdd0163011","revisions":{"8fee94f8175a89481103f6b147032ff2c4a8ff3b":{"kind":"REWORK","_number":1,"created":"2024-03-15 15:36:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/913418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/18/913418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/1"}}},"commit":{"parents":[{"commit":"7df32250d301c5a3d5d996793e41c922ef35fb66","subject":"Merge \"Add mistra-extra repo\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/7df32250d301c5a3d5d996793e41c922ef35fb66"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-15 15:36:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-15 15:36:25.000000000","tz":60},"subject":"Switch SHAs to EOM","message":"Switch SHAs to EOM\n\nChange-Id: I91b8cf26d568d8b20d927e4723232de735407f21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/8fee94f8175a89481103f6b147032ff2c4a8ff3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/8fee94f8175a89481103f6b147032ff2c4a8ff3b"}]},"branch":"refs/heads/stable/victoria"},"521c97bbb7ebe6894a88d0aab0f006ee58559dd0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-22 10:46:56.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/913418/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/18/913418/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/2"}}},"commit":{"parents":[{"commit":"62082b510ff518064c10c50445f71282caecc7c1","subject":"Remove use of undefined ceph distro job zuul template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/62082b510ff518064c10c50445f71282caecc7c1"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-15 15:36:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-03-22 10:46:56.000000000","tz":0},"subject":"Switch SHAs to EOM","message":"Switch SHAs to EOM\n\nChange-Id: I91b8cf26d568d8b20d927e4723232de735407f21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/521c97bbb7ebe6894a88d0aab0f006ee58559dd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/521c97bbb7ebe6894a88d0aab0f006ee58559dd0"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"},"e9d49257191c60ad1a3aed6c9a147504117ea3a4":{"kind":"REWORK","_number":3,"created":"2024-03-28 17:09:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/913418/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/18/913418/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/3"}}},"commit":{"parents":[{"commit":"7df32250d301c5a3d5d996793e41c922ef35fb66","subject":"Merge \"Add mistra-extra repo\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/7df32250d301c5a3d5d996793e41c922ef35fb66"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-15 15:36:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-28 17:09:10.000000000","tz":60},"subject":"Switch SHAs to EOM","message":"Switch SHAs to EOM\n\nThis also removes undefined jobs as well as leaving bare\nminimal coverage to ensure patch is mergable.\n\nChange-Id: I91b8cf26d568d8b20d927e4723232de735407f21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/e9d49257191c60ad1a3aed6c9a147504117ea3a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/e9d49257191c60ad1a3aed6c9a147504117ea3a4"}]},"branch":"refs/heads/stable/victoria"},"8ecb59586d91af244e8ba0cfb5aea533ec7e848f":{"kind":"REWORK","_number":4,"created":"2024-03-28 17:09:48.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/913418/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/18/913418/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/4"}}},"commit":{"parents":[{"commit":"7df32250d301c5a3d5d996793e41c922ef35fb66","subject":"Merge \"Add mistra-extra repo\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/7df32250d301c5a3d5d996793e41c922ef35fb66"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-15 15:36:25.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-03-28 17:09:46.000000000","tz":0},"subject":"Switch SHAs to EOM","message":"Switch SHAs to EOM\n\nThis also removes undefined jobs as well as leaving bare\nminimal coverage to ensure patch is mergable.\n\nChange-Id: I91b8cf26d568d8b20d927e4723232de735407f21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/8ecb59586d91af244e8ba0cfb5aea533ec7e848f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/8ecb59586d91af244e8ba0cfb5aea533ec7e848f"}]},"branch":"refs/heads/stable/victoria"},"63489cca1893c52ef26eddf68d18f8ce656b01fa":{"kind":"REWORK","_number":5,"created":"2024-04-11 12:26:33.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/18/913418/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/18/913418/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/5"}}},"commit":{"parents":[{"commit":"7df32250d301c5a3d5d996793e41c922ef35fb66","subject":"Merge \"Add mistra-extra repo\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/7df32250d301c5a3d5d996793e41c922ef35fb66"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-15 15:36:25.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-04-11 12:26:31.000000000","tz":120},"subject":"Switch SHAs to EOM","message":"Switch SHAs to EOM\n\nThis also removes undefined jobs as well as leaving bare\nminimal coverage to ensure patch is mergable.\n\nIn addition drop jobs referencing debian-buster as that release is no\nlonger available in opendev CI.\n\nChange-Id: I91b8cf26d568d8b20d927e4723232de735407f21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/63489cca1893c52ef26eddf68d18f8ce656b01fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/63489cca1893c52ef26eddf68d18f8ce656b01fa"}]},"branch":"refs/heads/stable/victoria"},"dc728b77a888a61deb61bcb28d5a68fdd0163011":{"kind":"REWORK","_number":6,"created":"2024-04-11 12:37:51.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/18/913418/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/18/913418/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/18/913418/6"}}},"commit":{"parents":[{"commit":"7df32250d301c5a3d5d996793e41c922ef35fb66","subject":"Merge \"Add mistra-extra repo\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/7df32250d301c5a3d5d996793e41c922ef35fb66"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-15 15:36:25.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-04-11 12:37:38.000000000","tz":120},"subject":"Switch SHAs to EOM","message":"Switch SHAs to EOM\n\nThis also removes undefined jobs as well as leaving bare\nminimal coverage to ensure patch is mergable.\n\nIn addition drop jobs referencing debian-buster as that release is no\nlonger available in opendev CI.\n\nChange-Id: I91b8cf26d568d8b20d927e4723232de735407f21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/dc728b77a888a61deb61bcb28d5a68fdd0163011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/dc728b77a888a61deb61bcb28d5a68fdd0163011"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
