)]}'
{"id":"openstack%2Fopenstack-ansible~918979","triplet_id":"openstack%2Fopenstack-ansible~stable%2Fzed~I3cbf2f216817e80e62f8a61019f1d12f5b713d51","project":"openstack/openstack-ansible","branch":"stable/zed","topic":"bump_osa","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-06-06 17:44:58.000000000","reason":"Change was submitted"},"16011":{"account":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"last_update":"2024-06-06 12:34:22.000000000","reason":"\u003cGERRIT_ACCOUNT_16011\u003e replied on the change","reason_account":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2024-06-06 15:27:57.000000000","reason":"\u003cGERRIT_ACCOUNT_25023\u003e replied on the change","reason_account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}}},"hashtags":[],"change_id":"I3cbf2f216817e80e62f8a61019f1d12f5b713d51","subject":"Transition Zed to Unmaintained","status":"MERGED","created":"2024-05-10 09:09:59.000000000","updated":"2024-06-06 17:46:12.000000000","submitted":"2024-06-06 17:44:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918979-bump_osa","meta_rev_id":"c96c98511d8622a7f0bab45c76c2e3e2bb60642c","_number":918979,"virtual_id_number":918979,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-06 17:44:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"all":[{"value":2,"date":"2024-06-06 12:34:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-06 15:27:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-06 15:27:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-10 09:11:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-05 07:54:30.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2024-06-06 12:30:34.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"CC"},{"updated":"2024-06-06 12:34:22.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"}],"messages":[{"id":"fe7b40a0a9192f537def4f37819ea1e364cd6b49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-10 09:09:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f11f874aeec2d8e4fd7476fa7a2f08aded346775","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-10 09:10:28.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a357b57a299bbd67723a35549caaa00dd01d5759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 09:11:24.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/918978 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"da755e35e46be3530b80d04e90320f1a77336a49","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-10 09:19:00.000000000","message":"Topic set to release_osa","accounts_in_message":[],"_revision_number":2},{"id":"d195f486b927cdeb972ef2cef82bbbe3443ae5f9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-10 14:12:27.000000000","message":"Topic release_osa removed","accounts_in_message":[],"_revision_number":2},{"id":"df4aef54d792315f56c7d3f6cce4ba33d15e98fe","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-10 14:12:32.000000000","message":"Topic set to bump_osa","accounts_in_message":[],"_revision_number":2},{"id":"5e3a68bd25f4d198ceea5b8f0d6e247e7e54cc53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-11 10:02:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"76ecc4cf196cfa5c24648938d965155620527899","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 12:26:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/118567074562484b82adff1e6dd74d9e\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfabcbca2dbd49a59891d031c5c555b7 : SUCCESS in 11m 26s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/16cb0215c9d44718a5ba86564a2a0750 : FAILURE in 2m 47s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/c42804979c8d4a099f6d0ed55bbda18d : SUCCESS in 12m 55s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/043e9c31aef6400d8160c4b34c543c54 : FAILURE in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/898ba33c39574f0f9956741e8af510e2 : FAILURE in 3m 40s\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/439e830a6fc84379a95407248bc80705 : SUCCESS in 1h 57m 40s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2a046a0c94ff43268a8ea901db08e6fc : SUCCESS in 2h 06m 35s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d85e5d2bd16f4d8483303b1864f2bfcf : SUCCESS in 1h 27m 21s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ade09e52ab474f47ae7fc15a5d693d0c : SUCCESS in 1h 27m 18s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db1143c83acf42d7a6cf99cc445884f5 : SUCCESS in 1h 47m 28s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ff9e6a91ce304a26b0eb571fd4a7dcc1 : SUCCESS in 1h 00m 13s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/abdcc8fcce93416d9b47d5e0e244d6e6 : SUCCESS in 1h 09m 26s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dfe71f8fffca4721bb58d398f8d61591 : SUCCESS in 1h 25m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d35dd52449644e5c8fd1ff47d1686364 : SUCCESS in 1h 00m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99fe57b4e9a3473e86b882aa3fce9275 : SUCCESS in 1h 04m 52s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d25eb52b61464952bc41536879a97732 : SUCCESS in 1h 44m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89b513ef1ced4ee994e20b4eaed169c1 : SUCCESS in 1h 22m 08s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/91e18d196eb84cac95ed5c40500a2bf2 : FAILURE in 37m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66c467dd4e954976a4dc90c33e726c54 : FAILURE in 43m 18s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27cb9f8da3764be68b01d0121fcf9f60 : SUCCESS in 1h 36m 24s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/793fe74be8fa4a1aab591f030061803c : SUCCESS in 2h 05m 15s\n- openstack-ansible-deploy-aio_proxy-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4eaa2d1a117e44278b8a9c8cf8129ef7 : SUCCESS in 1h 30m 52s","accounts_in_message":[],"_revision_number":3},{"id":"6f534262539dcd063cf78696c68bb5fae701b3a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-20 18:17:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"472580fc0fb8583751a67717dbbe31d0e45753f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 18:32:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/17289db96f1b4eadb3fac3b02fac2ecd\n\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/1676dbd12da94e0b91d891010eb236e1 : FAILURE in 2m 54s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/beb29dbb733543e39f5f3feb5d36dd31 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e012e2d950734f1b84e71cc15b9bb8e2 : SUCCESS in 9m 33s","accounts_in_message":[],"_revision_number":4},{"id":"e23905064a76db5af5989750690d8bc8a0c2ed2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-03 09:34:08.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e87ec2db9a07128ad636db519077a024fa9b87ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-03 10:15:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"877f2d2c0a4f28c9feec7084169b9839224cc5c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-03 11:15:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"833fcf01b85c966075038c5accc91e300b06c128","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 13:25:09.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1990d43aec24effa2d6878dc2c8aed8\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cef262977f2a4582bc32a24f0e15b702 : SUCCESS in 10m 26s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/1ab42a5a70b04b54a00f19675d7c8af2 : SUCCESS in 3m 14s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/e69a774520f1453fb859d8454c15c9b7 : SUCCESS in 12m 57s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/cc35e591ce7a4d23957bcaba9f8398ee : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f705e493594fb2bd8801c179b88306 : SUCCESS in 7m 58s\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ced4be3099042b9a18bc63955540a67 : SUCCESS in 1h 43m 34s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/79039bd549804b19ab4604cfe9aab739 : SUCCESS in 2h 02m 55s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/150d5879ef6845ec97c4d13182dc0571 : SUCCESS in 1h 38m 56s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd6fe7bf8e27474cbe3d27c7457c0d19 : FAILURE in 47m 01s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/475216d9beea446da650c8764b876fd6 : FAILURE in 44m 35s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/62ba1730fba4489b8f49ba706935af57 : SUCCESS in 56m 42s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/21fd1dab59ca4f2cb6f102cafeaad563 : SUCCESS in 1h 11m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fa1a4fdbe9a3420882be01180d5fc4ed : SUCCESS in 1h 15m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1772d5d239394775bbd8c8d423bc67c1 : FAILURE in 32m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97f6949f5fe74aa9a8f87851f3cd8906 : FAILURE in 35m 18s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/83ce2515e0f14e9382862cbeee94f275 : SUCCESS in 1h 53m 59s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d094369c3f464a57aaffb723bfc951a0 : FAILURE in 1h 01m 39s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a8223c9f08b242b6aed4d7b44656cbfb : FAILURE in 35m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/034e45d5c0ff49178b1cc423aed7a4d8 : FAILURE in 50m 48s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44ab701e4e98456d8076c8b92c553b63 : FAILURE in 49m 31s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0f2bdb7e6694225bcc5f3b4f323f797 : FAILURE in 57m 23s\n- openstack-ansible-deploy-aio_proxy-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb02a391156e4f44b1a69cb69030f0a6 : FAILURE in 46m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b64da17c388b403c94b1360758f40609 : SUCCESS in 10m 19s","accounts_in_message":[],"_revision_number":7},{"id":"072b3742a0c5a4664a90a0a20fbe63562256c5e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-03 13:39:56.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"411f585fba0a111a8cde35be33f1e0e59dce5014","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-03 14:24:11.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"58bccbe73a6c4fb09e910310a8b59538cfbe367a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 16:56:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e712e932b4f9408dbea4fc1b098672dc\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67cd25b8800441c4a045bb8f92104c5c : SUCCESS in 12m 32s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/1e56e8fbba0c40cdb5efc75cb72adb63 : SUCCESS in 3m 06s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/eadbd91c1c734bd1a811e8490f7458f2 : SUCCESS in 14m 47s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/292eb1185baf4a3bbba1c77fd2399e8a : SUCCESS in 15m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7892686f5c3c46d189c92cfe5de3dda9 : SUCCESS in 9m 28s\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/7621ac3164ff4a7fa8a0f74600949e44 : SUCCESS in 2h 17m 32s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3ff64fb8753840f18f7a42d3edd8f1e2 : SUCCESS in 2h 16m 10s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/5669e663baa54db28696afdaf8bef221 : SUCCESS in 2h 18m 11s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/334422a7ebf046898ac4c19d5fff7ba5 : SUCCESS in 1h 40m 43s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1c4e0f88b164901939bb7b1c5dc43e5 : SUCCESS in 2h 02m 00s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ba650a89b8484630b97245130d9c8414 : SUCCESS in 1h 15m 45s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/eeedac64ec7040f8a6898d3a2732fb1a : SUCCESS in 1h 31m 27s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bc125d491e6e48dca4f948b453ebffb7 : SUCCESS in 1h 32m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/188fbe3b4eb54cddaf8c0b3dc703e61e : SUCCESS in 1h 07m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c722c1bb1de940dcaf1268239c99340a : SUCCESS in 1h 25m 47s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a22d2f3b15e14306b8d1830ecacdaaff : SUCCESS in 1h 58m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28f64d75e012441d9da98b8242f55e04 : SUCCESS in 1h 35m 32s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0bc27a6560434fa99817c8ae94d68cd3 : FAILURE in 42m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dae0bcc4ef16488db35c5209d5bdd939 : FAILURE in 46m 18s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d70bd461ba14431b8a70bae201e31490 : SUCCESS in 2h 09m 58s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66e3b2def98f401ab307bf6aeb212bfb : SUCCESS in 2h 11m 42s\n- openstack-ansible-deploy-aio_proxy-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10d6195ea3054ff1878cc21fb8137dc4 : SUCCESS in 1h 53m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/844f8270a6884e0792b380804b4886d1 : SUCCESS in 11m 26s","accounts_in_message":[],"_revision_number":9},{"id":"8a3bd6c0d171ee31be8a25856488b1aa8ee6bb9a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-06-05 07:54:30.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"a37762f608518996c356c3b016905e4dfa2333b1","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2024-06-06 12:30:34.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"60c48a11e5b6c40e42c3c1f85ca8ea3135066844","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-06 12:32:38.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"102cbab7a75cf4328be9e885c6a47e8016362a5e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-06 12:32:49.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6ccf47f46f4d9c553bb6a4261dfbd27f16f0b63c","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2024-06-06 12:34:22.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"3de08e44c606a8cd18fe1405ec3487e887d8be34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 14:54:16.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce89de4f63d14462b2b52fa1beb97169\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88837d66b7634adeb9f23b1bf319398d : SUCCESS in 10m 21s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/490c206506e749a3b6bf4919b1fe02b0 : SUCCESS in 3m 01s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/91e448afd2c14e31a45756ec023878a1 : SUCCESS in 11m 00s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/9addffda893f45d8a00f40739bcdbd8c : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb5a9a2bf1c4400a903683747a7050cd : SUCCESS in 10m 26s\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c8d30322a82404cb10bc9358db7ef3b : SUCCESS in 2h 01m 55s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dcb0c63caa9b4d948b59d5811abc0fa5 : SUCCESS in 2h 04m 47s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a4d1cd0f5cd3474a93c67c2a2314163f : SUCCESS in 1h 42m 05s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89336c7166bd49be867798b3350abf03 : SUCCESS in 1h 46m 27s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4758d29df2bc499da0d02cfdda5dacaf : SUCCESS in 1h 47m 46s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/98432a52726a48119d9179490ffcc9d5 : SUCCESS in 1h 03m 56s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff77feb40c0b4faeb162150ab7485e49 : SUCCESS in 1h 17m 16s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dcc9187f86db43599c274bfdc4cb6958 : SUCCESS in 1h 24m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8edc0ac70134a2f9129b40326ef3594 : SUCCESS in 1h 05m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/962d806f70c54dfb9afb8c37e1ad77f0 : SUCCESS in 1h 28m 42s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b717ac06e914f34993da17434dbcf35 : SUCCESS in 2h 08m 54s\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08a3e5f7db0e43d0802e1650ba85e2c3 : SUCCESS in 1h 52m 38s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/13d30b1ecf664e1581b5dba41d40521a : FAILURE in 34m 25s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfc0b6e66e6447e481019d90a57808b3 : FAILURE in 47m 15s (non-voting)\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3f99697c0a24913bf3e8dfeee345e3d : SUCCESS in 1h 57m 20s\n- openstack-ansible-deploy-aio_ceph-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39c16cc507374467bf46b5ba1c133a70 : SUCCESS in 1h 59m 07s\n- openstack-ansible-deploy-aio_proxy-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a83fd04a5f1e46258ea859f04c6b77ea : SUCCESS in 2h 00m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f601f19ba544e7fb4e2ab92bb8dc88c : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":10},{"id":"740e6ac84f967d4991380f9774583a7f9adf204f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-06-06 15:27:57.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"9448e2984f0dd076f14950544cd464eff5cae453","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 15:29:40.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"3ea1984cbed198b1b52b7542a868eac24167f76f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 17:44:57.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/681e7a72e6434f70bf6c25bca352699c\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1626e073f6d8436b8864300f27aa9de3 : SUCCESS in 9m 47s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/ba67097eead040219dafc33f8f0b7b89 : SUCCESS in 2m 59s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/4596d1ec249c4c5eb08562d434c135ce : SUCCESS in 12m 46s\n- openstack-ansible-shastest-metal https://zuul.opendev.org/t/openstack/build/cfd4eeeff62c4bb6b3d4ca21e09fd391 : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65b4233c6ede4c738c3b9c9b0d22b218 : SUCCESS in 9m 02s\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1dd9ca7819604349900f4225d0882cb3 : SUCCESS in 1h 53m 17s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/a7692aa155904a3198781ac8a360b3a4 : SUCCESS in 1h 43m 39s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34214e5d55aa4a13a90369e50e051a0c : SUCCESS in 1h 37m 08s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/32afe3c18e894728a5f8df57001cb700 : SUCCESS in 1h 21m 20s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b45d7c76e61f4956a950606485b2eab0 : SUCCESS in 1h 14m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04910a9388c34d7e905954d1ab823fd8 : SUCCESS in 1h 16m 24s\n- openstack-ansible-deploy-aio_ceph-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc63bd14e160427e955bb073f289a44e : SUCCESS in 1h 58m 19s\n- openstack-ansible-deploy-aio_proxy-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f70ed10447ae4f42a2f00eba8f88049a : SUCCESS in 2h 01m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f637c316edcd464b992d0a8b71c533e4 : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":10},{"id":"c9a735a752f4cffec35a2164eb32e65206b7a700","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 17:44:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"c96c98511d8622a7f0bab45c76c2e3e2bb60642c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 17:46:12.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86ef26f0f26d4d97ab4c021c6e0dbf6e\n\n- promote-deploy-guide https://zuul.opendev.org/t/openstack/build/236446fad3634f22abec99cb2ec592c5 : SUCCESS in 37s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d578d445f7664d0a965c3679af89e337 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef98b758fe03425b9aa476f5af19691b : SUCCESS in 49s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"eadd8d4667e449bc927f0e69279ef6430b0ffc32","revisions":{"a2075055d951aa2445f769556b22b183350310c4":{"kind":"REWORK","_number":1,"created":"2024-05-10 09:09:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/79/918979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/79/918979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/1"}}},"commit":{"parents":[{"commit":"dfc5a989b79f91b5503746a922857a462811e192","subject":"Bump SHAs for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/dfc5a989b79f91b5503746a922857a462811e192"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-10 09:09:52.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-10 09:09:52.000000000","tz":120},"subject":"Transition Zed to Unmaintained","message":"Transition Zed to Unmaintained\n\nChange-Id: I3cbf2f216817e80e62f8a61019f1d12f5b713d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/a2075055d951aa2445f769556b22b183350310c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/a2075055d951aa2445f769556b22b183350310c4"}]},"branch":"refs/heads/stable/zed"},"8ad7bb3e1e571cb72248869d8bdcc3cf26bfb4a8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-05-10 09:10:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/79/918979/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/79/918979/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/2"}}},"commit":{"parents":[{"commit":"dfc5a989b79f91b5503746a922857a462811e192","subject":"Bump SHAs for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/dfc5a989b79f91b5503746a922857a462811e192"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-10 09:09:52.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-05-10 09:10:28.000000000","tz":0},"subject":"Transition Zed to Unmaintained","message":"Transition Zed to Unmaintained\n\nDepends-On: https://review.opendev.org/c/openstack/releases/+/918978\nChange-Id: I3cbf2f216817e80e62f8a61019f1d12f5b713d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/8ad7bb3e1e571cb72248869d8bdcc3cf26bfb4a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/8ad7bb3e1e571cb72248869d8bdcc3cf26bfb4a8"}]},"branch":"refs/heads/stable/zed","description":"Edit commit message"},"79f2dc10c90edb655689685787b25a150494868f":{"kind":"REWORK","_number":3,"created":"2024-05-11 10:02:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/79/918979/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/79/918979/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/3"}}},"commit":{"parents":[{"commit":"dfc5a989b79f91b5503746a922857a462811e192","subject":"Bump SHAs for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/dfc5a989b79f91b5503746a922857a462811e192"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-10 09:09:52.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-11 10:02:22.000000000","tz":120},"subject":"Transition Zed to Unmaintained","message":"Transition Zed to Unmaintained\n\nDepends-On: https://review.opendev.org/c/openstack/releases/+/918978\nChange-Id: I3cbf2f216817e80e62f8a61019f1d12f5b713d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/79f2dc10c90edb655689685787b25a150494868f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/79f2dc10c90edb655689685787b25a150494868f"}]},"branch":"refs/heads/stable/zed"},"bee9db920a211be4f4445b585b4027866d7054bb":{"kind":"REWORK","_number":4,"created":"2024-05-20 18:17:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/79/918979/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/79/918979/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/4"}}},"commit":{"parents":[{"commit":"dfc5a989b79f91b5503746a922857a462811e192","subject":"Bump SHAs for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/dfc5a989b79f91b5503746a922857a462811e192"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-10 09:09:52.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-20 18:17:29.000000000","tz":120},"subject":"Transition Zed to Unmaintained","message":"Transition Zed to Unmaintained\n\nDepends-On: https://review.opendev.org/c/openstack/releases/+/918978\nChange-Id: I3cbf2f216817e80e62f8a61019f1d12f5b713d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/bee9db920a211be4f4445b585b4027866d7054bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/bee9db920a211be4f4445b585b4027866d7054bb"}]},"branch":"refs/heads/stable/zed"},"f9e816fcf5ade096777cebfd37493ae0a7b1b697":{"kind":"REWORK","_number":5,"created":"2024-06-03 09:34:08.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/79/918979/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/79/918979/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/5"}}},"commit":{"parents":[{"commit":"7c6e5ef7f32656d642ba4de1f87313bea63bad1f","subject":"Remove retired projects from zuul required-projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/7c6e5ef7f32656d642ba4de1f87313bea63bad1f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-10 09:09:52.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-06-03 09:33:54.000000000","tz":120},"subject":"Transition Zed to Unmaintained","message":"Transition Zed to Unmaintained\n\nChange-Id: I3cbf2f216817e80e62f8a61019f1d12f5b713d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/f9e816fcf5ade096777cebfd37493ae0a7b1b697"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/f9e816fcf5ade096777cebfd37493ae0a7b1b697"}]},"branch":"refs/heads/stable/zed"},"13bdf806c82ecb730ae513ca10e7eab5626c734c":{"kind":"REWORK","_number":6,"created":"2024-06-03 10:15:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/79/918979/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/79/918979/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/6"}}},"commit":{"parents":[{"commit":"7c6e5ef7f32656d642ba4de1f87313bea63bad1f","subject":"Remove retired projects from zuul required-projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/7c6e5ef7f32656d642ba4de1f87313bea63bad1f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-10 09:09:52.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-06-03 10:15:32.000000000","tz":120},"subject":"Transition Zed to Unmaintained","message":"Transition Zed to Unmaintained\n\nChange-Id: I3cbf2f216817e80e62f8a61019f1d12f5b713d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/13bdf806c82ecb730ae513ca10e7eab5626c734c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/13bdf806c82ecb730ae513ca10e7eab5626c734c"}]},"branch":"refs/heads/stable/zed"},"08cdd28acf68346376e33244676170aaf3ea77a2":{"kind":"REWORK","_number":7,"created":"2024-06-03 11:15:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/79/918979/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/79/918979/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/7"}}},"commit":{"parents":[{"commit":"7c6e5ef7f32656d642ba4de1f87313bea63bad1f","subject":"Remove retired projects from zuul required-projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/7c6e5ef7f32656d642ba4de1f87313bea63bad1f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-10 09:09:52.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-06-03 11:15:49.000000000","tz":120},"subject":"Transition Zed to Unmaintained","message":"Transition Zed to Unmaintained\n\nChange-Id: I3cbf2f216817e80e62f8a61019f1d12f5b713d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/08cdd28acf68346376e33244676170aaf3ea77a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/08cdd28acf68346376e33244676170aaf3ea77a2"}]},"branch":"refs/heads/stable/zed"},"0b56bcfbff45b20af4bb18ee484cec4cbdc286ef":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2024-06-03 13:39:56.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/79/918979/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/79/918979/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/8"}}},"commit":{"parents":[{"commit":"7c6e5ef7f32656d642ba4de1f87313bea63bad1f","subject":"Remove retired projects from zuul required-projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/7c6e5ef7f32656d642ba4de1f87313bea63bad1f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-10 09:09:52.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-06-03 13:39:55.000000000","tz":0},"subject":"Transition Zed to Unmaintained","message":"Transition Zed to Unmaintained\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-openstack_hosts/+/921028\nChange-Id: I3cbf2f216817e80e62f8a61019f1d12f5b713d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/0b56bcfbff45b20af4bb18ee484cec4cbdc286ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/0b56bcfbff45b20af4bb18ee484cec4cbdc286ef"}]},"branch":"refs/heads/stable/zed","description":"Edit commit message"},"58718d1d46ed4f7ea8921a8a2be24e4a5b5f4088":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2024-06-03 14:24:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/79/918979/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/79/918979/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/9"}}},"commit":{"parents":[{"commit":"7c6e5ef7f32656d642ba4de1f87313bea63bad1f","subject":"Remove retired projects from zuul required-projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/7c6e5ef7f32656d642ba4de1f87313bea63bad1f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-10 09:09:52.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-06-03 14:24:11.000000000","tz":0},"subject":"Transition Zed to Unmaintained","message":"Transition Zed to Unmaintained\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-openstack_hosts/+/921037\nChange-Id: I3cbf2f216817e80e62f8a61019f1d12f5b713d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/58718d1d46ed4f7ea8921a8a2be24e4a5b5f4088"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/58718d1d46ed4f7ea8921a8a2be24e4a5b5f4088"}]},"branch":"refs/heads/stable/zed","description":"Edit commit message"},"eadd8d4667e449bc927f0e69279ef6430b0ffc32":{"kind":"REWORK","_number":10,"created":"2024-06-06 12:32:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/79/918979/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/79/918979/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/79/918979/10"}}},"commit":{"parents":[{"commit":"7c6e5ef7f32656d642ba4de1f87313bea63bad1f","subject":"Remove retired projects from zuul required-projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/7c6e5ef7f32656d642ba4de1f87313bea63bad1f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-05-10 09:09:52.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-06-06 12:32:38.000000000","tz":0},"subject":"Transition Zed to Unmaintained","message":"Transition Zed to Unmaintained\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-openstack_hosts/+/921037\nChange-Id: I3cbf2f216817e80e62f8a61019f1d12f5b713d51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/eadd8d4667e449bc927f0e69279ef6430b0ffc32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/eadd8d4667e449bc927f0e69279ef6430b0ffc32"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
