)]}'
{"id":"openstack%2Fopenstack-ansible~991980","triplet_id":"openstack%2Fopenstack-ansible~master~I721300834d25b46286299c00cdf020bf5e20c5bf","project":"openstack/openstack-ansible","branch":"master","topic":"osa/policy_yaml","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-09-28 10:07:03.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-21 09:14:48.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-28 10:07:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I721300834d25b46286299c00cdf020bf5e20c5bf","subject":"docs: update details about overriding .yaml files","status":"MERGED","created":"2026-06-06 18:05:47.000000000","updated":"2026-09-28 10:09:02.000000000","submitted":"2026-09-28 10:07:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991980-osa/policy_yaml","meta_rev_id":"5264672a9bc19200d8b5f6ab6615e89f86342468","_number":991980,"virtual_id_number":991980,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-28 10:07:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-09-28 09:16:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2026-09-21 09:14:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2026-09-28 09:16:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-06 18:28:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-06 18:57:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-08-26 12:57:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-09-28 09:16:18.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"ae0de08f6b8205e6183eb6f2dfb692982daf5e20","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-06 18:05:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba21b3cf692984c9b393cc39400b6532961c83ff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-06 18:11:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"55b58fe7673bed66e328b717a75bf3f8200c16d7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-06 18:11:25.000000000","message":"Topic set to osa/policy_yaml","accounts_in_message":[],"_revision_number":2},{"id":"510b490d58fcdd3f54401a601080c1ab145657d1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-06 18:11:36.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d0b51866bb715a28b16411eb4d63c1389abf28e0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-06 18:12:26.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"e334471c3e976ba4cb3a879a5e4659e1f0d060ed","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-06 18:12:43.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"429ffbfb75811a58ac9ca46549ab01dc4168bc46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-06 18:12:49.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"e325412abc879fcb2f91efd5a18938f1fe422c57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 18:28:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c66c8f5d53ba4083a167f3e2e4f237bb\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a4fc9846b5b46348aabaa93dc31a903 : SUCCESS in 12m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffab038576c44f1d95953c3fa98fad2b : SUCCESS in 10m 31s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/8f7a865ac6a84e87856b2ff665438cce : SUCCESS in 5m 10s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/3f6d13c92300442da9ee94396d7d47e8 : POST_FAILURE in 6m 47s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8531c5ed9594c71bb8c2808fa536caa : SUCCESS in 5m 37s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/1935c5d4901a4c76a4226de8e0f79380 : SUCCESS in 6m 14s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/1cbff29ed096408b9820909488d098f6 : SUCCESS in 5m 09s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/230b60f386f541848e8e0eb14f9f8c62 : SUCCESS in 4m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dffee7500e684d2aa4c43fb04d2a116c : SUCCESS in 9m 30s","accounts_in_message":[],"_revision_number":5},{"id":"658ffbdfc7af9f737013ae3647d6241a0570f99a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-06 18:28:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cda55d0a07d008d665a3fc973a16b841f9408385","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-06 18:54:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bc51a5af53e42958e086963522ee58c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a48e3b92be44fd2a51598921278a438 : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d8a107513249cab0224d2bfb4fd5cd : SUCCESS in 11m 21s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/b96b006173154bd09cfbea5cb646191b : SUCCESS in 4m 58s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/d99f1d42dc4d437bb9e78f27fa96f2a8 : SUCCESS in 3m 47s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98788be8ca3e4957a3aae992afabbe2d : SUCCESS in 4m 43s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/bbdb53084690480ab7e4fa2a80e1b54c : SUCCESS in 4m 55s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/da3eba51683c42559a61a591eafa0e82 : SUCCESS in 5m 32s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4e0172dd21fc4db4be75b01aac1c042b : SUCCESS in 4m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e05c21a65c84223a14542cd18ead265 : SUCCESS in 11m 16s","accounts_in_message":[],"_revision_number":5},{"id":"8260b9e22d8c89e4f79f31698703eb8e44ed5baa","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-09 15:25:54.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b504563892da815724e5e4cf116f29e9e276cf6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 05:13:05.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"6eefbdd062359c8ec87a324314907b579eb22358","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 05:13:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c25f4ed0f3b9a97a205108dba0408b4047d0dff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-25 05:41:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c9c58f49c0f486fb55734ac876ba279\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6ad53d68a93415ba1a537719710a62f : SUCCESS in 16m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26101dc4cc1f46b78d5b47f85ae30441 : FAILURE in 8m 40s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/008d36d14c6c4435a5dab2b0d50184ab : SUCCESS in 5m 39s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/356d2f00ca424e2c93883b0a5a9d05e9 : SUCCESS in 5m 54s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5214b8192f784a84ac8d53f2d37561dc : SUCCESS in 5m 50s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/87a00764e4714c8cbef6e3186ae42122 : SUCCESS in 5m 23s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/aa519c97ec3e41ee9c6489e52be58358 : SUCCESS in 5m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e88a243876542ada16711175a5fc546 : SUCCESS in 12m 04s","accounts_in_message":[],"_revision_number":6},{"id":"7fd4e97076f274cfbceb50259cf00567a149b3f3","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 06:07:23.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":6},{"id":"7b89eb0f386896a967011109de4d663b348dc82a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 06:17:51.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"adaa84c0e91b49dd1331092716bce147878af1a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-25 06:17:57.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/openstack-ansible for 991980,7","accounts_in_message":[],"_revision_number":7},{"id":"a12cf043ad82be9dcb6d7165538e2b016ae8c656","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 06:18:08.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"02e307e40ea0c2a93c840c1bddd8d07f0b99bd77","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 06:18:11.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"e73009ed3b1612210917ffdf33553f0adf2ecb28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-25 06:39:04.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/653dde3e42f3478ebd530e6d007c4344\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15fdcf8ee95c43af99c4ce13096d4cc0 : SUCCESS in 10m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b9577f5c40c4954988f3a596a59de2d : FAILURE in 9m 44s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/ab7c267a253a4301868bc915dadffa95 : SUCCESS in 5m 28s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e0197c992b14ee9a23fbb3be07f8a88 : SUCCESS in 5m 15s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/2c2ae328eb33483b834d6f886651e152 : SUCCESS in 5m 25s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/5a2f11a45c304795bcd95a4eb3064079 : SUCCESS in 4m 32s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/83ebce763e964434bc25df328132df9b : SUCCESS in 4m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33bfe0d3c8dc42c4bb662db77d08f256 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":9},{"id":"1ce104426e26b361e4d1647f9a3a94d6bc88927c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 06:41:12.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"05fbcad3feee1da9acb8a999ac8f9a5eb1326fa0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 06:42:24.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"3ab645bdb870700c157cb4222953147a4806457f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-25 06:58:08.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce2aebcf47c24056b3c4f859cf2c8b57\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91c38e37448343cea0ff99db01508409 : SUCCESS in 7m 09s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/9e2880e2d2284ac49e8643154906ff61 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9db789f9b7347e88f2296c68d81b745 : FAILURE in 8m 26s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/0151bf344d064ac1b874b9a72dfa5e1c : SUCCESS in 5m 13s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97d230f0101b472581b13723f0872990 : SUCCESS in 5m 23s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/0d9b8f3a78ff4ee7ae61f95435d09629 : SUCCESS in 4m 08s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/a905c9cfa7c54db181c0a5d065971648 : SUCCESS in 4m 29s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8eb3e69ae60f4f0e825cd7ed7cea2f9b : SUCCESS in 3m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ccace7ba2f64f0bb85f14b232a2c76b : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":11},{"id":"b217794f56ddc7de23cd7042f6bf210d7fc504d8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 07:11:53.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"39e23f09843492c488ef2617a2da6fc297a658bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-25 07:23:41.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e9cf6aa1a674a6ea8cb8a8c5a986952\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3acc9f899fd2461bb42b31930426dd20 : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2faa7a1d65174e14956c841321650978 : SUCCESS in 4m 59s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/cc767e90c4344c68b715bc450f912e9e : SUCCESS in 3m 21s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/58d778ee76b24e11aaa4332e4bb652eb : SUCCESS in 3m 11s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/0dc54d33e34d402290515d69b6390849 : SUCCESS in 3m 28s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/d1a1c427a91e414a992ee4500646a594 : SUCCESS in 4m 30s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1b29f18a7a8045499b72a99657d1d1d3 : SUCCESS in 3m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0a02ef672954831882e4d1eaa8e3a16 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":12},{"id":"09ffa27cecde9308010c17d5d89cf83e720ca74b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-25 07:24:22.000000000","message":"Patch Set 12:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":12},{"id":"a1508bbd4e9906ac5e779bac0f0c5d0eaa6c749c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-26 06:34:40.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"3a16c9449d32775c690e8972a919052b7874fa55","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-26 06:35:35.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b9d62b061589a0c3ee1c4a0fa698f57c745dc9a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-26 09:38:35.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"c07ef96c35244978d2d14dda0c05b70c89d37fe8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-26 09:38:46.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"5e2d458943fef7eb71e2a63f7b0ea1be770ce1fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 10:07:58.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e740b681688c4a6eb5454d8e06df596a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8473dcf804a44b4ea50e2570619dbd23 : SUCCESS in 15m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eea730581a3b40bfa4daed5af6430c20 : FAILURE in 8m 37s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/5886687f50c2482b8a22ae172f8ff7f8 : SUCCESS in 5m 26s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ac67edb86184ac783fe18c8875d6ceb : SUCCESS in 4m 52s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/7eb982fd2f8648f49a746f41d57e8b8e : SUCCESS in 5m 33s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/f96fca389c614eca8759e2c188076618 : SUCCESS in 5m 32s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/54f2a919c53442938f2ca5e8496b031a : SUCCESS in 5m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4e439c3be8f4a1da74a92a29d866bda : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":13},{"id":"82b223f495d904d48e5839fd21094d273b7f672c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-26 12:56:59.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"904584a49ea2bd9563be8d82e253ada06c2948a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 13:13:06.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/589f4daea102416e86b3d90e5a23feef\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f43686ded6574bb2a1bfed73e50a9cc3 : SUCCESS in 10m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/981fbe932c0a4a0588a47c484eaaf985 : FAILURE in 8m 40s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/a7d0137fb51c4998bf8b671edad890cb : SUCCESS in 6m 33s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80430676ceb04b949c3bb5a8b13b42b7 : SUCCESS in 5m 54s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/cba8c5a2dff9424fa67dd94ef030c215 : SUCCESS in 6m 08s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/c407f97e18b048bdbab2841fa5d61d46 : SUCCESS in 5m 50s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/71c308225cba459fab2b5cc944a3ddb6 : SUCCESS in 6m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8075d9a900d4aa9ac4f3181f859c609 : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":13},{"id":"d74c5a81e3290119d33834bcc094065832a86aa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-26 13:18:14.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"362672ccb7d049813ffc499f31209d4d265442e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 13:46:25.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc653f7076d34c158515d72666b63310\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73e82f9fb46a4a25a8767cc6f20152e3 : SUCCESS in 13m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a0ff42b3c344599ab1f830648373527 : SUCCESS in 4m 47s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/82fd009de62b498db3084cabb33e4ab9 : SUCCESS in 5m 17s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/038d930c182f4c6eb795013fa6499b01 : SUCCESS in 5m 48s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b1d29fcbacff446ba70bb2a96cae5930 : SUCCESS in 7m 43s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/f4440031213741ddb015cb6cbbc9fa38 : SUCCESS in 4m 35s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5594dc7ac2604544852c979b8ecf4384 : SUCCESS in 6m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/110191ec13174f59b0e7abed74cf84a1 : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":14},{"id":"53a0c7c59197016eb0f197f9382f041c552685de","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-21 09:14:48.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"e514058a4880631276428ea66d94e75b462956f2","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-09-28 09:16:18.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"d66b3aa304011b8068235df57ebc08603bafca5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 09:21:32.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"468028bedd8e18beae7533d843a75d24f3069f2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 10:07:03.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06532f5c7d9a41bb9c83ede55c2bf967\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1beec033ddb4398bea68e12eb0d6926 : SUCCESS in 22m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d99b09a3ac034f8cae5c4f373268cc42 : SUCCESS in 8m 16s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/179aff4d4e5649ac90e958202d51173c : SUCCESS in 4m 19s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fae542db455c4bac8cbddb892b2ee683 : SUCCESS in 5m 47s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/757e3726d35d4531b4860ac80873a67d : SUCCESS in 6m 43s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/2dbbd761348541c4a4d8306e93adb08a : SUCCESS in 13m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/009a6344ed2840b0848b28a183225d14 : SUCCESS in 8m 31s","accounts_in_message":[],"_revision_number":14},{"id":"563a66605eb19300aead06b9abf2b2e717d2d873","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 10:07:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"5264672a9bc19200d8b5f6ab6615e89f86342468","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 10:09:02.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3dfc970b095407ab64cf81b5a9f3a8a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96ee78c3412e49a7b1db27f0b03b0000 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"a8927b6d9d7578ccd46535f5fb4d4d1980a7abb2","revisions":{"a1489c7e4f6bc347f9b315cea5932dac481416e8":{"kind":"REWORK","_number":1,"created":"2026-06-06 18:05:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/991980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/80/991980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/1"}}},"commit":{"parents":[{"commit":"eb37f5371b3e9a8b86fec2d6e91fc86bc2315281","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/eb37f5371b3e9a8b86fec2d6e91fc86bc2315281"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:05:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:05:47.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I721300834d25b46286299c00cdf020bf5e20c5bf\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/a1489c7e4f6bc347f9b315cea5932dac481416e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/a1489c7e4f6bc347f9b315cea5932dac481416e8"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"2b5a6c4f0347f763ea90a82fcdfc85ae38425026":{"kind":"REWORK","_number":2,"created":"2026-06-06 18:11:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/991980/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/80/991980/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/2"}}},"commit":{"parents":[{"commit":"eb37f5371b3e9a8b86fec2d6e91fc86bc2315281","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/eb37f5371b3e9a8b86fec2d6e91fc86bc2315281"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:05:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:11:18.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I721300834d25b46286299c00cdf020bf5e20c5bf\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/2b5a6c4f0347f763ea90a82fcdfc85ae38425026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/2b5a6c4f0347f763ea90a82fcdfc85ae38425026"}]},"branch":"refs/heads/master"},"232ed23992fcee7d49a149c2e9c9900d6a1a58f9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-06-06 18:11:36.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/991980/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/80/991980/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/3"}}},"commit":{"parents":[{"commit":"eb37f5371b3e9a8b86fec2d6e91fc86bc2315281","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/eb37f5371b3e9a8b86fec2d6e91fc86bc2315281"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:05:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:11:36.000000000","tz":0},"subject":"Update details about overriding .yaml files","message":"Update details about overriding .yaml files\n\nChange-Id: I721300834d25b46286299c00cdf020bf5e20c5bf\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/232ed23992fcee7d49a149c2e9c9900d6a1a58f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/232ed23992fcee7d49a149c2e9c9900d6a1a58f9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"22a9f1373894f1c01728ba168f4a49cbb3c82670":{"kind":"REWORK","_number":4,"created":"2026-06-06 18:12:26.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/991980/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/80/991980/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/4"}}},"commit":{"parents":[{"commit":"eb37f5371b3e9a8b86fec2d6e91fc86bc2315281","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/eb37f5371b3e9a8b86fec2d6e91fc86bc2315281"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:05:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:12:26.000000000","tz":0},"subject":"Update details about overriding .yaml files","message":"Update details about overriding .yaml files\n\nChange-Id: I721300834d25b46286299c00cdf020bf5e20c5bf\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/22a9f1373894f1c01728ba168f4a49cbb3c82670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/22a9f1373894f1c01728ba168f4a49cbb3c82670"}]},"branch":"refs/heads/master"},"f2dcfa64744beb9ab49ba133761066b33c63850a":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-06-06 18:12:49.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/991980/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/80/991980/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/5"}}},"commit":{"parents":[{"commit":"eb37f5371b3e9a8b86fec2d6e91fc86bc2315281","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/eb37f5371b3e9a8b86fec2d6e91fc86bc2315281"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:05:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:12:49.000000000","tz":0},"subject":"docs: update details about overriding .yaml files","message":"docs: update details about overriding .yaml files\n\nChange-Id: I721300834d25b46286299c00cdf020bf5e20c5bf\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/f2dcfa64744beb9ab49ba133761066b33c63850a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/f2dcfa64744beb9ab49ba133761066b33c63850a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9ea9485f4d61d0c441c030e9a0739d29b71026e3":{"kind":"REWORK","_number":6,"created":"2026-07-25 05:13:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/991980/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/80/991980/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/6"}}},"commit":{"parents":[{"commit":"eb37f5371b3e9a8b86fec2d6e91fc86bc2315281","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/eb37f5371b3e9a8b86fec2d6e91fc86bc2315281"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:05:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 05:13:05.000000000","tz":0},"subject":"docs: update details about overriding .yaml files","message":"docs: update details about overriding .yaml files\n\nChange-Id: I721300834d25b46286299c00cdf020bf5e20c5bf\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/9ea9485f4d61d0c441c030e9a0739d29b71026e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/9ea9485f4d61d0c441c030e9a0739d29b71026e3"}]},"branch":"refs/heads/master"},"00819f4b3cdccc53089eb9ea5fdee75eb38d7c35":{"kind":"REWORK","_number":7,"created":"2026-07-25 06:17:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/991980/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/80/991980/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/7"}}},"commit":{"parents":[{"commit":"eb37f5371b3e9a8b86fec2d6e91fc86bc2315281","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/eb37f5371b3e9a8b86fec2d6e91fc86bc2315281"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:05:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 06:17:51.000000000","tz":0},"subject":"docs: update details about overriding .yaml files","message":"docs: update details about overriding .yaml files\n\nChange-Id: I721300834d25b46286299c00cdf020bf5e20c5bf\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/00819f4b3cdccc53089eb9ea5fdee75eb38d7c35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/00819f4b3cdccc53089eb9ea5fdee75eb38d7c35"}]},"branch":"refs/heads/master"},"5da373597f8b2723b07854e38d66bd42e4b6211c":{"kind":"REWORK","_number":8,"created":"2026-07-25 06:18:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/991980/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/80/991980/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/8"}}},"commit":{"parents":[{"commit":"eb37f5371b3e9a8b86fec2d6e91fc86bc2315281","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/eb37f5371b3e9a8b86fec2d6e91fc86bc2315281"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:05:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 06:18:06.000000000","tz":0},"subject":"docs: update details about overriding .yaml files","message":"docs: update details about overriding .yaml files\n\nChange-Id: I721300834d25b46286299c00cdf020bf5e20c5bf\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/5da373597f8b2723b07854e38d66bd42e4b6211c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/5da373597f8b2723b07854e38d66bd42e4b6211c"}]},"branch":"refs/heads/master"},"be5fc1d22fd2f76d825ad9daaf1469c19ca11d2a":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-07-25 06:18:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/991980/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/80/991980/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/9"}}},"commit":{"parents":[{"commit":"ff0803744ab500e08e3f6b61a586dbc19cc4e821","subject":"HW_ARCH traits are reported automatically","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/ff0803744ab500e08e3f6b61a586dbc19cc4e821"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:05:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 06:18:11.000000000","tz":0},"subject":"docs: update details about overriding .yaml files","message":"docs: update details about overriding .yaml files\n\nChange-Id: I721300834d25b46286299c00cdf020bf5e20c5bf\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/be5fc1d22fd2f76d825ad9daaf1469c19ca11d2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/be5fc1d22fd2f76d825ad9daaf1469c19ca11d2a"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"eb37f5371b3e9a8b86fec2d6e91fc86bc2315281","ours":"5da373597f8b2723b07854e38d66bd42e4b6211c","theirs":"ff0803744ab500e08e3f6b61a586dbc19cc4e821","merge_strategy":"recursive","contains_conflicts":false}},"3f727ce7d9c90621226c22a6e09a6a696cc4c85f":{"kind":"REWORK","_number":10,"created":"2026-07-25 06:41:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/991980/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/80/991980/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/10"}}},"commit":{"parents":[{"commit":"ff0803744ab500e08e3f6b61a586dbc19cc4e821","subject":"HW_ARCH traits are reported automatically","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/ff0803744ab500e08e3f6b61a586dbc19cc4e821"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:05:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 06:41:12.000000000","tz":0},"subject":"docs: update details about overriding .yaml files","message":"docs: update details about overriding .yaml files\n\nChange-Id: I721300834d25b46286299c00cdf020bf5e20c5bf\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/3f727ce7d9c90621226c22a6e09a6a696cc4c85f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/3f727ce7d9c90621226c22a6e09a6a696cc4c85f"}]},"branch":"refs/heads/master"},"aa7c41707f706529a7378286126fb4838f6bcfa1":{"kind":"REWORK","_number":11,"created":"2026-07-25 06:42:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/991980/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/80/991980/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/11"}}},"commit":{"parents":[{"commit":"ff0803744ab500e08e3f6b61a586dbc19cc4e821","subject":"HW_ARCH traits are reported automatically","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/ff0803744ab500e08e3f6b61a586dbc19cc4e821"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:05:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 06:42:24.000000000","tz":0},"subject":"docs: update details about overriding .yaml files","message":"docs: update details about overriding .yaml files\n\nChange-Id: I721300834d25b46286299c00cdf020bf5e20c5bf\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/aa7c41707f706529a7378286126fb4838f6bcfa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/aa7c41707f706529a7378286126fb4838f6bcfa1"}]},"branch":"refs/heads/master"},"07da1d662c72e270bc07c2ccdcc12d7b9ddd0a12":{"kind":"REWORK","_number":12,"created":"2026-07-25 07:11:53.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/991980/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/80/991980/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/12"}}},"commit":{"parents":[{"commit":"ff0803744ab500e08e3f6b61a586dbc19cc4e821","subject":"HW_ARCH traits are reported automatically","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/ff0803744ab500e08e3f6b61a586dbc19cc4e821"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:05:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-25 07:11:53.000000000","tz":0},"subject":"docs: update details about overriding .yaml files","message":"docs: update details about overriding .yaml files\n\nChange-Id: I721300834d25b46286299c00cdf020bf5e20c5bf\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/07da1d662c72e270bc07c2ccdcc12d7b9ddd0a12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/07da1d662c72e270bc07c2ccdcc12d7b9ddd0a12"}]},"branch":"refs/heads/master"},"387b5c25a2150dee45e4e46ef53417465413c362":{"kind":"REWORK","_number":13,"created":"2026-08-26 09:38:35.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/991980/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/80/991980/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/13"}}},"commit":{"parents":[{"commit":"ff0803744ab500e08e3f6b61a586dbc19cc4e821","subject":"HW_ARCH traits are reported automatically","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/ff0803744ab500e08e3f6b61a586dbc19cc4e821"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:05:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-26 09:38:31.000000000","tz":0},"subject":"docs: update details about overriding .yaml files","message":"docs: update details about overriding .yaml files\n\nChange-Id: I721300834d25b46286299c00cdf020bf5e20c5bf\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/387b5c25a2150dee45e4e46ef53417465413c362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/387b5c25a2150dee45e4e46ef53417465413c362"}]},"branch":"refs/heads/master"},"a8927b6d9d7578ccd46535f5fb4d4d1980a7abb2":{"kind":"REWORK","_number":14,"created":"2026-08-26 13:18:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/80/991980/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/80/991980/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/80/991980/14"}}},"commit":{"parents":[{"commit":"ff0803744ab500e08e3f6b61a586dbc19cc4e821","subject":"HW_ARCH traits are reported automatically","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/ff0803744ab500e08e3f6b61a586dbc19cc4e821"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-06 18:05:47.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-08-26 13:18:14.000000000","tz":0},"subject":"docs: update details about overriding .yaml files","message":"docs: update details about overriding .yaml files\n\nChange-Id: I721300834d25b46286299c00cdf020bf5e20c5bf\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/a8927b6d9d7578ccd46535f5fb4d4d1980a7abb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/a8927b6d9d7578ccd46535f5fb4d4d1980a7abb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
