)]}'
{"id":"openstack%2Fopenstack-ansible~995339","triplet_id":"openstack%2Fopenstack-ansible~master~I648cb5d561bc7b34fc8d85fe12dd8f1c4a1e4271","project":"openstack/openstack-ansible","branch":"master","topic":"osa/molecule-plugins-systemd","attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-06-29 17:10:14.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I648cb5d561bc7b34fc8d85fe12dd8f1c4a1e4271","subject":"Bump molecule and ansible-lint requirements","status":"NEW","created":"2026-06-29 14:18:40.000000000","updated":"2026-08-19 11:22:12.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"4ebdf0587d15ce41537987fd5d3716e4cf1360a1","_number":995339,"virtual_id_number":995339,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-19 11:22:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-29 14:33:54.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-06-29 17:10:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf8a65ec2da9f26f6af6d062025209a8abed7830","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-29 14:18:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ece0b0d2fbf735643b09e6ab7841afb2ccfa2c26","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-29 14:33:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"303249693869adb387704931f791a16de452455a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 17:10:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/544d52cfcc014d1f92440b4eb8e3449a\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e4e2ae7bde54d7ab40a00b21e7406b9 : SUCCESS in 12m 58s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/f7cc12df38db4f6d91992071ed37cba0 : SUCCESS in 7m 12s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/2d46538197d74f7ca2b8161dab98190e : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66a189464dd9431f89d211b9c9d4389a : SUCCESS in 9m 25s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/d6af8f0ca602413cb102a8e1d65eb99d : SUCCESS in 1h 53m 47s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c7d5ca9b1d82490188aeedf0250b846a : SUCCESS in 1h 27m 03s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/036a0fe5a9f24551913774967c38494b : SUCCESS in 1h 18m 19s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16cac6f8d4ad4f2395d58eb100f4fde2 : SUCCESS in 2h 27m 11s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5a734bb384074adebb0c9fc40105b9a4 : SUCCESS in 2h 22m 55s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/174d7241e84a464e9087b4c11e173e1b : SUCCESS in 1h 14m 19s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/4d78edbb65a04e76a9cc447249ba92e2 : SUCCESS in 1h 24m 25s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f293b4d1f56642288169fb4e279fbd0b : SUCCESS in 1h 35m 15s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e43609ad054d436eae0f77f0fd79b0c2 : SUCCESS in 1h 15m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9fe5f1f28134423aa66e11136503994 : SUCCESS in 1h 30m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/4e179ab5b86045f88a2dfcd846ed923b : SUCCESS in 1h 26m 23s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ca6eabb8038457ab51181a6d7adf13c : SUCCESS in 51m 19s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/5535706b14c34d27a872c9de4dd9451c : SUCCESS in 2h 04m 49s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2120b771ca94deca8c046f099699691 : SUCCESS in 1h 20m 22s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0d739c852c3409a8e3ddc4f5fc9dbd1 : SUCCESS in 2h 29m 20s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23f5eb5da17b49f8909055208a6db4a5 : SUCCESS in 1h 20m 38s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5d013282e64c46438344b9c955db2fb7 : SUCCESS in 1h 29m 57s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a883dcbdfb7643cfb6ded6e39d755d81 : SUCCESS in 2h 42m 50s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a22c8962b064f8f83b336a4820004e1 : SUCCESS in 1h 57m 04s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bf0ebc080354919b0221ecc1300b637 : SUCCESS in 2h 21m 43s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3309c6cb4e9944c2bed8287435fcb207 : SUCCESS in 1h 28m 24s\n- openstack-ansible-deploy-aio_metal_openbao-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad80169f39324a03927be5a0fd7998ba : SUCCESS in 52m 21s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/5f553176048c44e7961d4ef4001c9aa7 : FAILURE in 3m 28s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6f908348564497ba78ba72e61dd9a7a : FAILURE in 2m 58s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ee1d3cce343146448e193ab65bbae6be : FAILURE in 6m 10s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/0a4f7cd2ec9a4eb08992bc249be4beb4 : FAILURE in 4m 04s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a9c044fa90a34d08a2745471d7134327 : FAILURE in 4m 53s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/16eb564bfb83431c880b38a8b3142246 : SUCCESS in 3m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7780a7e0aae54a09aa45565f398cd9d5 : SUCCESS in 13m 13s","accounts_in_message":[],"_revision_number":1},{"id":"985c279195357b07056f8907aa3a77a4c78d23a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-18 12:49:33.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7be24eb644053aad54dbec12e2aad61b3b7bad25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-18 13:11:17.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d7d189bdc16853a5829e91fb9120befecbf27f45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-18 14:25:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d39cdef10888ec4a18e01203d95b4486387de800","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-18 14:28:20.000000000","message":"Topic set to osa/molecule-plugins-systemd","accounts_in_message":[],"_revision_number":4},{"id":"2d015193e1318ab41a77f843fd9c0681551aedbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 17:20:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51c48904e79145429407d580589876d9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ede435fbbc50491181349ea0ba0cea25 : SUCCESS in 16m 32s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/f69fd2f837174391bdc61c942940b538 : SUCCESS in 8m 11s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/c294c0b52c9d4a6b9115017b2f01d9f9 : SUCCESS in 15m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d4c388ff19432caed89284e20c53fb : SUCCESS in 9m 38s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/e8a6baa3db4742d3923c0c9982eac013 : SUCCESS in 1h 59m 28s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/c6a66b2cbebb42b692c17815eac4353d : SUCCESS in 2h 23m 58s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f416ea7a11ba4739862192a28b6cf20c : FAILURE in 29m 51s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed2a7482ff7a4bd88136c027ae1f6881 : SUCCESS in 2h 03m 45s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/76d548c0b6524f12b48f3d42d60d4a72 : SUCCESS in 1h 20m 30s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/fe7bcbc3920f4b11891f8fb42740fdc7 : SUCCESS in 1h 14m 06s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f565e1519c8c426c805eec232385505e : SUCCESS in 1h 43m 37s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f969facfaf224c26b3477582503a6e34 : SUCCESS in 1h 30m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7f81d74a8a8a4bdaa41cfb76d2cbfc8d : SUCCESS in 1h 16m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5ea69e765a845088f7bb79cf3690679 : SUCCESS in 1h 45m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/7f3147093bea4c05b4cbd0d2730491ce : SUCCESS in 1h 08m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf67fb30ea824c26a3edf3cc1358facf : SUCCESS in 1h 37m 25s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/10d1253d9e0d4a1d8e6218a1d494eb33 : SUCCESS in 2h 11m 57s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9ed6fcffd984218b7b20629b8e3dfe4 : SUCCESS in 2h 37m 13s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/191d482fff914a4ab15087dfb8d8561c : SUCCESS in 2h 32m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e17773585234bf9b2b0e297a8b96a29 : SUCCESS in 1h 28m 42s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5e2e1c47715a4011bc324f390625a625 : SUCCESS in 1h 41m 21s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e83d84d651a84916b982653a39cb8985 : FAILURE in 50m 27s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3c86dfc60d64f6092a71070ac07ffcf : SUCCESS in 2h 24m 29s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a118a0cafc33489e937f8fb338dd95ad : SUCCESS in 2h 27m 12s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bff856701b2944afb33cc35b55e9d285 : SUCCESS in 1h 35m 26s\n- openstack-ansible-deploy-aio_metal_openbao-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff592d1c403c4fd4b69641b290ca0190 : SUCCESS in 1h 39m 19s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/0268c38d637f40739f804e3de742bb53 : SUCCESS in 7m 17s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc93920bfb2c47e0b33f5048627a5446 : SUCCESS in 6m 52s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/80383d6bbca54d8e9b32058ae6675c92 : SUCCESS in 6m 32s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/e1c110194fc24ba0b0fa8c1808b88aa7 : SUCCESS in 7m 09s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6cbe239a1108453db5755f8c749916c9 : SUCCESS in 7m 05s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/5a2dd37ae98346d19a3d213299bea1c5 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8194dffd815c436198b2b65336c6ff9d : SUCCESS in 13m 30s","accounts_in_message":[],"_revision_number":4},{"id":"b3036751ceffdccfbe0fbeaefc342c15e3cd1d41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-19 08:25:51.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"4ebdf0587d15ce41537987fd5d3716e4cf1360a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 11:22:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60b063f1bce84449b54da68f5b7509e8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/839725ed0ba84ebb94a22d7d38fea6a0 : SUCCESS in 10m 04s\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/b921e57a4af449f1850c81192f8c605f : SUCCESS in 6m 45s\n- openstack-ansible-varstest-aio https://zuul.opendev.org/t/openstack/build/7f9a1c18241f4744b47d1fa17544fbf8 : SUCCESS in 15m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/588087a7ad3c46bba5abd6ccc1087203 : SUCCESS in 10m 08s\n- openstack-ansible-deploy-aio_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/a49c6686a1ae4ab69e447094cd7394e4 : SUCCESS in 2h 24m 29s\n- openstack-ansible-deploy-aio_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/8d0bcbcfa7f9427192883df22616fb53 : SUCCESS in 1h 22m 56s\n- openstack-ansible-deploy-aio_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3c9cc933738a4d46bb5c88d6d39c8506 : SUCCESS in 2h 40m 03s\n- openstack-ansible-deploy-aio_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/329a460ff2a74b8b820ae592fcf59890 : SUCCESS in 2h 16m 17s\n- openstack-ansible-deploy-aio_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/ef1464ff26714669b7cf40117601b4b7 : SUCCESS in 2h 21m 05s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/b28c07c144204868a891633195bcdcf6 : SUCCESS in 1h 13m 45s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/9c7dadbfd8b045cbacfe8e771b988547 : SUCCESS in 1h 29m 13s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/63e0a58927414e6e8f19652b2f4dd1ec : SUCCESS in 1h 26m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b6b39f2341fb46a892b935d532b50061 : SUCCESS in 1h 17m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2902fff5e8734110b22b5c02c042dd44 : SUCCESS in 1h 43m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/9f80f1f76edd432695c21003c37fa335 : SUCCESS in 1h 31m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ac9bbfb44f940be886fb050ca89dfc5 : SUCCESS in 1h 40m 00s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3bcb12f5b43145808f561bd997394d57 : SUCCESS in 2h 13m 19s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93aac51ce93a49499fb4ab2158ecc3a3 : SUCCESS in 2h 28m 05s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70cc35f2bb69485abc170bb463de548f : SUCCESS in 2h 45m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19ea13a78251485c9ebbcbca39db961e : SUCCESS in 1h 18m 20s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a834bcd70f7c43268a6bd6027368c01e : SUCCESS in 57m 46s\n- openstack-ansible-deploy-aio_ceph-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97321a3f17884c0c8c0d5948f88d1e2b : SUCCESS in 2h 47m 36s\n- openstack-ansible-deploy-aio_proxy-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84f5115aee1c482787aedde375bd05b6 : SUCCESS in 2h 24m 35s\n- openstack-ansible-deploy-aio_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4dff5783469499ebc744dcd885ef238 : SUCCESS in 1h 22m 48s\n- openstack-ansible-deploy-aio_metal_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e6ff54863014c608d5f6224aec2c60f : SUCCESS in 1h 35m 06s\n- openstack-ansible-deploy-aio_metal_openbao-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0432977652414e1b82beea80bdc627f2 : SUCCESS in 1h 15m 01s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/b092f40d3bc249788aee7e0bbfc96334 : SUCCESS in 6m 18s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c6b3395f65b44c4b874985ddabe0ac3 : SUCCESS in 7m 17s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/109e6562494d49bbba4585d7808ecfb7 : SUCCESS in 7m 23s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/e450aaf5de094b94a02010cdf767d375 : SUCCESS in 6m 21s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cd7bd5dddf6f48a8809be02c26fbebb0 : SUCCESS in 6m 31s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/ca80c410b7a444e99e6168a5d5e5df61 : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eaddbaf8e89940ff90a2a65e8377650d : SUCCESS in 11m 35s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9db8f713a872f54ae203b5547fee41a6fd8f494b","revisions":{"a91b45e345f500cb556b3a29473cd5030cdfb0e2":{"kind":"REWORK","_number":1,"created":"2026-06-29 14:18:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/39/995339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/39/995339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/1"}}},"commit":{"parents":[{"commit":"947791ee1757937bb6ce072e843c8000851b3de2","subject":"Merge \"Switch development to track master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/947791ee1757937bb6ce072e843c8000851b3de2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-29 14:18:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-29 14:18:30.000000000","tz":120},"subject":"Bump molecule and ansible-lint requirements","message":"Bump molecule and ansible-lint requirements\n\nChange-Id: I648cb5d561bc7b34fc8d85fe12dd8f1c4a1e4271\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/a91b45e345f500cb556b3a29473cd5030cdfb0e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/a91b45e345f500cb556b3a29473cd5030cdfb0e2"}]},"branch":"refs/heads/master"},"241fe5acd489c61abefa1bf2acbfa58feac7f644":{"kind":"REWORK","_number":2,"created":"2026-08-18 12:49:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/39/995339/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/39/995339/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/2"}}},"commit":{"parents":[{"commit":"947791ee1757937bb6ce072e843c8000851b3de2","subject":"Merge \"Switch development to track master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/947791ee1757937bb6ce072e843c8000851b3de2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-29 14:18:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-18 12:49:33.000000000","tz":0},"subject":"Bump molecule and ansible-lint requirements","message":"Bump molecule and ansible-lint requirements\n\nChange-Id: I648cb5d561bc7b34fc8d85fe12dd8f1c4a1e4271\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/241fe5acd489c61abefa1bf2acbfa58feac7f644"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/241fe5acd489c61abefa1bf2acbfa58feac7f644"}]},"branch":"refs/heads/master"},"ac3a46213898cc55ef6e5bc73703d2767219eaca":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-08-18 13:11:17.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/39/995339/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/39/995339/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/3"}}},"commit":{"parents":[{"commit":"30dcaf0d452697b9d921b3618daea5dc4a7fc8ce","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/30dcaf0d452697b9d921b3618daea5dc4a7fc8ce"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-29 14:18:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-18 13:05:46.000000000","tz":120},"subject":"Bump molecule and ansible-lint requirements","message":"Bump molecule and ansible-lint requirements\n\nChange-Id: I648cb5d561bc7b34fc8d85fe12dd8f1c4a1e4271\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/ac3a46213898cc55ef6e5bc73703d2767219eaca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/ac3a46213898cc55ef6e5bc73703d2767219eaca"}]},"branch":"refs/heads/master"},"afedf26113304ec8170f89132ac2ceddfea4153c":{"kind":"REWORK","_number":4,"created":"2026-08-18 14:25:38.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/39/995339/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/39/995339/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/4"}}},"commit":{"parents":[{"commit":"30dcaf0d452697b9d921b3618daea5dc4a7fc8ce","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/30dcaf0d452697b9d921b3618daea5dc4a7fc8ce"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-29 14:18:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-18 14:06:16.000000000","tz":120},"subject":"Bump molecule and ansible-lint requirements","message":"Bump molecule and ansible-lint requirements\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-python_venv_build/+/1001201\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/1001348\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-ops/+/1001350\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/1001351\nDepends-On: https://review.opendev.org/c/openstack/ansible-config_template/+/1001354\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-httpd/+/1001355\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_mount/+/1001358\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_networkd/+/1001359\nDeepnds-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/1001360\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-uwsgi/+/1001361\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-apt_package_pinning/+/1001362\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-haproxy_server/+/1001363\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-memcached_server/+/1001365\nChange-Id: I648cb5d561bc7b34fc8d85fe12dd8f1c4a1e4271\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/afedf26113304ec8170f89132ac2ceddfea4153c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/afedf26113304ec8170f89132ac2ceddfea4153c"}]},"branch":"refs/heads/master"},"9db8f713a872f54ae203b5547fee41a6fd8f494b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-08-19 08:25:51.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/39/995339/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/39/995339/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/39/995339/5"}}},"commit":{"parents":[{"commit":"5d0a17d2aa365baa680405c7f26e8bf971dc0883","subject":"Switch ceph release to Umbrella","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/5d0a17d2aa365baa680405c7f26e8bf971dc0883"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-29 14:18:30.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-08-19 08:25:51.000000000","tz":0},"subject":"Bump molecule and ansible-lint requirements","message":"Bump molecule and ansible-lint requirements\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-python_venv_build/+/1001201\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-pki/+/1001348\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-ops/+/1001350\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-plugins/+/1001351\nDepends-On: https://review.opendev.org/c/openstack/ansible-config_template/+/1001354\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-httpd/+/1001355\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_mount/+/1001358\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-systemd_networkd/+/1001359\nDeepnds-On: https://review.opendev.org/c/openstack/ansible-role-systemd_service/+/1001360\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-uwsgi/+/1001361\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-apt_package_pinning/+/1001362\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-haproxy_server/+/1001363\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-memcached_server/+/1001365\nChange-Id: I648cb5d561bc7b34fc8d85fe12dd8f1c4a1e4271\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/9db8f713a872f54ae203b5547fee41a6fd8f494b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/9db8f713a872f54ae203b5547fee41a6fd8f494b"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"30dcaf0d452697b9d921b3618daea5dc4a7fc8ce","ours":"afedf26113304ec8170f89132ac2ceddfea4153c","theirs":"5d0a17d2aa365baa680405c7f26e8bf971dc0883","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
