)]}'
{"id":"openstack%2Fopenstack-ansible~996735","triplet_id":"openstack%2Fopenstack-ansible~master~I13abeb74c378d7e7c27e92c4cd4f3629ed3bed66","project":"openstack/openstack-ansible","branch":"master","topic":"zanata/translations","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-08-17 09:04:30.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-10 09:21:15.000000000","reason":"removed on reply"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-08-17 09:23:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I13abeb74c378d7e7c27e92c4cd4f3629ed3bed66","subject":"Imported Translations from Zanata","status":"MERGED","created":"2026-07-10 04:16:38.000000000","updated":"2026-08-17 09:24:33.000000000","submitted":"2026-08-17 09:23:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996735-zanata/translations","meta_rev_id":"5a07a50e2d90b179321f74100e8c7196c1c40e9d","_number":996735,"virtual_id_number":996735,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-17 09:23:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-08-10 08:03:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-08-17 09:04:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-10 09:21:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-08-17 09:04:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-10 04:30:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-10 17:26:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-07-10 17:26:22.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-08-10 08:03:16.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-08-10 09:21:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"0f53e9e22729b7e18195d9d74a4d1eafb4cf07ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-10 04:16:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53483a2f9d1346eed8527c0cbbcf8d87777a52b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 04:30:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74e0afd1b1584087a45f4664353b76d8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e28e9946d45436fb41a25e4be62bee6 : SUCCESS in 9m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ded337097bb949718130021f5f97e205 : SUCCESS in 6m 06s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/d6ff76756e684a60a0d0c7a7218aa9b9 : SUCCESS in 3m 48s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df82d817395a459d8b9691ab229be8cc : SUCCESS in 3m 51s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e4b0726b37b0444d902be34269ef0ba7 : SUCCESS in 5m 34s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/a212991d92db48d49ef10ca1fe958e96 : SUCCESS in 5m 21s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/95712de8768c42faae4161503ba84c50 : SUCCESS in 5m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4a1f135add0444daa9192d357f52811 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":1},{"id":"bd7100a1f724b9c0e45a98533f3dda55258fad64","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-10 17:26:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"adce6c966e05860d3abf195608acd3c921815bdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-07-23 04:16:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2404be79fc60c56face641a49247f7a522a99a08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 04:29:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/905b679b69f74c16b582e157be4c8316\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c83b87668946468781299881fe94f672 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/405c875df48b4357abe5887ef34fd802 : SUCCESS in 5m 15s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/784a2e7a5fb3432b80f1b6878e289d29 : SUCCESS in 4m 47s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69b9402220c94b6e8518166ae1a462e0 : SUCCESS in 3m 19s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/5161500bd97f482588abdaf9b9b98a31 : SUCCESS in 4m 52s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/3a9b8637249d4261b6a07ba663de9b80 : SUCCESS in 4m 27s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8d140f602b3b472687bc1e45e99b23c5 : SUCCESS in 3m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d981ad31d6884eac8e22d3e471d32b09 : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":2},{"id":"1b21333a65a440b87d0e3c7c928bbdf004be51c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-08-10 04:11:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f209d145ec0e34637a7e590fd046f53c824eaf9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 04:30:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da10e559444848dda6330140d6ceb778\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3015f3a9b45e4f958dbac79f73c5ea05 : SUCCESS in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b736342b65e49dc975043a108bbd427 : SUCCESS in 8m 57s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/cb0b953896084c0186f71fea365af03e : SUCCESS in 5m 06s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9931b220fed94251b5a620852d4735a6 : SUCCESS in 4m 44s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/fa96546789e044a4a350dd5f7d05dcd9 : SUCCESS in 3m 25s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/aa521f3cba83403d8bc2c708d5f9c162 : SUCCESS in 5m 23s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6c4276e8e6b343e69f19d3d6da6e26ab : SUCCESS in 4m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2d22bc934bf4473978211d397cd3196 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":3},{"id":"7722565418df66fabe66c3e0d7435b0f9da42452","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-08-10 08:03:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f819f29891526dbf3b335569958dd7ede9d4dd5a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-08-10 09:21:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c15317a5f343a54aa9c315b1e3b4700d7c909b2e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-17 09:04:30.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e4acbd2274c1a17d708a99d3cde58435c5ffb8ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 09:07:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"64b3829bae176ebda5d765717781047aff018ff4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 09:23:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60957f141e9a4e93a8962216135733e5\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae77ada2d6be404483d29013c48f076c : SUCCESS in 14m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2801b3e8d6f641c0ae4f294c25e165bd : SUCCESS in 10m 08s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/88e8b9b9bba8466a91447ec0cfcac802 : SUCCESS in 3m 24s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47de7646b18f456b9f7f87e1d9cd069c : SUCCESS in 5m 53s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/3a61e4fbe2eb4df3831b324115a51383 : SUCCESS in 4m 48s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/8f9bceef815b47d3a786c4f53c9e6f47 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0215bbca8184335980e1b38539f1c14 : SUCCESS in 11m 41s","accounts_in_message":[],"_revision_number":3},{"id":"098159b092eaa9e446dd01700027e13e68dfd44b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 09:23:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5a07a50e2d90b179321f74100e8c7196c1c40e9d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 09:24:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f294ba034db248429a1d3c873fcf4021\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b94f0ed7d3b43328ba6982ebd8ad837 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"30dcaf0d452697b9d921b3618daea5dc4a7fc8ce","revisions":{"45f2f0d5621500e891d02da0ea115a749718218c":{"kind":"REWORK","_number":1,"created":"2026-07-10 04:16:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/996735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/35/996735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/35/996735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/35/996735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/35/996735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/35/996735/1"}}},"commit":{"parents":[{"commit":"b7c10cbcc04a06207c2344a312d4f612a0810939","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/b7c10cbcc04a06207c2344a312d4f612a0810939"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-10 04:16:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-10 04:16:34.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I13abeb74c378d7e7c27e92c4cd4f3629ed3bed66\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/openstack-zuul-jobs:roles/prepare-zanata-client/files/common_translation_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/45f2f0d5621500e891d02da0ea115a749718218c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/45f2f0d5621500e891d02da0ea115a749718218c"}]},"branch":"refs/heads/master"},"1c5ff3e86267888470ad48154efadcfaf1d2857a":{"kind":"REWORK","_number":2,"created":"2026-07-23 04:16:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/996735/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/35/996735/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/35/996735/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/35/996735/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/35/996735/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/35/996735/2"}}},"commit":{"parents":[{"commit":"ff0803744ab500e08e3f6b61a586dbc19cc4e821","subject":"HW_ARCH traits are reported automatically","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/ff0803744ab500e08e3f6b61a586dbc19cc4e821"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-23 04:16:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-07-23 04:16:57.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I13abeb74c378d7e7c27e92c4cd4f3629ed3bed66\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/openstack-zuul-jobs:roles/prepare-zanata-client/files/common_translation_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/1c5ff3e86267888470ad48154efadcfaf1d2857a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/1c5ff3e86267888470ad48154efadcfaf1d2857a"}]},"branch":"refs/heads/master"},"30dcaf0d452697b9d921b3618daea5dc4a7fc8ce":{"kind":"REWORK","_number":3,"created":"2026-08-10 04:11:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/996735/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible","ref":"refs/changes/35/996735/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/35/996735/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/35/996735/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible refs/changes/35/996735/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible refs/changes/35/996735/3"}}},"commit":{"parents":[{"commit":"ff0803744ab500e08e3f6b61a586dbc19cc4e821","subject":"HW_ARCH traits are reported automatically","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/ff0803744ab500e08e3f6b61a586dbc19cc4e821"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-10 04:11:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-08-10 04:11:52.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I13abeb74c378d7e7c27e92c4cd4f3629ed3bed66\nSigned-off-by: OpenStack Proposal Bot \u003copenstack-infra@lists.openstack.org\u003e\nGenerated-By: openstack/openstack-zuul-jobs:roles/prepare-zanata-client/files/common_translation_update.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/30dcaf0d452697b9d921b3618daea5dc4a7fc8ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible/commit/30dcaf0d452697b9d921b3618daea5dc4a7fc8ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
