)]}'
{"id":"openstack%2Fopenstack-chef-repo~114413","triplet_id":"openstack%2Fopenstack-chef-repo~master~I398c40f274cbb12caa228297b0bcc2c49ece0fcc","project":"openstack/openstack-chef-repo","branch":"master","topic":"chef-zero-mode","hashtags":[],"change_id":"I398c40f274cbb12caa228297b0bcc2c49ece0fcc","subject":"Add Chef Zero(local mode) deploy guide","status":"MERGED","created":"2014-08-15 01:31:39.000000000","updated":"2014-10-06 14:59:12.000000000","submitted":"2014-10-06 14:59:12.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2c3c064618a01c3eb298bc4a45ff1c2a1556c429","_number":114413,"virtual_id_number":114413,"owner":{"_account_id":9488,"name":"chenzhiwei","email":"zhiweik@gmail.com","username":"chenzhiwei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2014-10-06 14:59:12.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12323,"name":"JJ Asghar","email":"jjasghar@gmail.com","username":"jj"},{"value":0,"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},{"value":0,"_account_id":1032,"name":"Justin Shepherd","email":"jshepher@rackspace.com","username":"jshepher"},{"value":0,"_account_id":9884,"name":"Wherenow.org CI","email":"jenkins@wherenow.org","username":"wherenowjenkins","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12323,"name":"JJ Asghar","email":"jjasghar@gmail.com","username":"jj"},{"value":2,"date":"2014-09-29 16:43:50.000000000","_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},{"value":2,"date":"2014-10-06 14:53:28.000000000","_account_id":1032,"name":"Justin Shepherd","email":"jshepher@rackspace.com","username":"jshepher"},{"value":0,"_account_id":9884,"name":"Wherenow.org CI","email":"jenkins@wherenow.org","username":"wherenowjenkins","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1032,"name":"Justin Shepherd","email":"jshepher@rackspace.com","username":"jshepher"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":12323,"name":"JJ Asghar","email":"jjasghar@gmail.com","username":"jj"},{"value":0,"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},{"value":1,"date":"2014-10-06 14:53:28.000000000","_account_id":1032,"name":"Justin Shepherd","email":"jshepher@rackspace.com","username":"jshepher"},{"value":0,"_account_id":9884,"name":"Wherenow.org CI","email":"jenkins@wherenow.org","username":"wherenowjenkins","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1032,"name":"Justin Shepherd","email":"jshepher@rackspace.com","username":"jshepher"},{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},{"_account_id":9884,"name":"Wherenow.org CI","email":"jenkins@wherenow.org","username":"wherenowjenkins","tags":["SERVICE_USER"]},{"_account_id":12323,"name":"JJ Asghar","email":"jjasghar@gmail.com","username":"jj"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-08-18 13:50:53.000000000","updated_by":{"_account_id":9884,"name":"Wherenow.org CI","email":"jenkins@wherenow.org","username":"wherenowjenkins","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9884,"name":"Wherenow.org CI","email":"jenkins@wherenow.org","username":"wherenowjenkins","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-08-29 22:08:53.000000000","updated_by":{"_account_id":12323,"name":"JJ Asghar","email":"jjasghar@gmail.com","username":"jj"},"reviewer":{"_account_id":12323,"name":"JJ Asghar","email":"jjasghar@gmail.com","username":"jj"},"state":"REVIEWER"},{"updated":"2014-09-29 16:43:50.000000000","updated_by":{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},"reviewer":{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},"state":"REVIEWER"},{"updated":"2014-10-06 14:53:28.000000000","updated_by":{"_account_id":1032,"name":"Justin Shepherd","email":"jshepher@rackspace.com","username":"jshepher"},"reviewer":{"_account_id":1032,"name":"Justin Shepherd","email":"jshepher@rackspace.com","username":"jshepher"},"state":"REVIEWER"},{"updated":"2014-10-06 14:59:12.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"2024ccc1b26735bcf487d1d663a6477b8743fa5f","author":{"_account_id":9488,"name":"chenzhiwei","email":"zhiweik@gmail.com","username":"chenzhiwei"},"date":"2014-08-15 01:31:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92494826a670391c8c3c4b9446a1971c8a9c47d6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-15 01:31:40.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"ab1dbabddf073f53eb2de64ac32d4b4419ad47ae","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-15 01:56:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- gate-openstack-chef-repo http://logs.openstack.org/13/114413/1/check/gate-openstack-chef-repo/f7ad2cc : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a8d978d509df53c824914c2339d5c00617870aa9","author":{"_account_id":9884,"name":"Wherenow.org CI","email":"jenkins@wherenow.org","username":"wherenowjenkins","tags":["SERVICE_USER"]},"date":"2014-08-18 13:50:53.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://jenkins.uk.rs.wherenow.org/job/Stackforge_Cookbooks_AIO_Matrix/2237/ : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"63e712e41e1a58dfa1e561fc83d3a69ee8732baa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-18 13:50:54.000000000","message":"Patch Set 1: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"e560da465dd453096485004deb02357b74f1eb25","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-08-18 14:02:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- gate-openstack-chef-repo http://logs.openstack.org/13/114413/1/check/gate-openstack-chef-repo/ddd2eed : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":1},{"id":"81a4747e198a8a1d0440ba69bfa5500cc64522d1","author":{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},"date":"2014-08-18 14:24:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)\n\nNice addition, added a few inline editorial suggestions","accounts_in_message":[],"_revision_number":1},{"id":"2069323b275a83c8c65b60a4f564d8e182c9c508","author":{"_account_id":12323,"name":"JJ Asghar","email":"jjasghar@gmail.com","username":"jj"},"date":"2014-08-29 22:08:53.000000000","message":"Patch Set 1: Code-Review-1\n\nGreat addition, but please follow through with Mark\u0027s comments.","accounts_in_message":[],"_revision_number":1},{"id":"35b6c25525f1bea270fc09d8785935108ce5a4dc","author":{"_account_id":9488,"name":"chenzhiwei","email":"zhiweik@gmail.com","username":"chenzhiwei"},"date":"2014-09-16 05:43:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"659b950b56da2d3de93f478952a3be6c6dec4d69","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-16 05:51:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-openstack-chef-repo http://logs.openstack.org/13/114413/2/check/gate-openstack-chef-repo/23c9bff : FAILURE in 3m 52s","accounts_in_message":[],"_revision_number":2},{"id":"d8e1f8dd28229681f8fed55fbbbd8d38e422dff7","author":{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},"date":"2014-09-16 16:10:36.000000000","message":"Patch Set 2:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":2},{"id":"72b2043e0790e4b1f11872c031b98c84d29e2599","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-16 18:11:22.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-openstack-chef-repo http://logs.openstack.org/13/114413/2/check/gate-openstack-chef-repo/863616e : FAILURE in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"6d15f52927f264f9c8c35a2b3945dceb35d6de71","author":{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},"date":"2014-09-16 19:13:36.000000000","message":"Patch Set 2: Workflow-1\n\nThe issue is with the gate jobs that I changed, sorry.  I created bug https://bugs.launchpad.net/openstack-chef/+bug/1370210\n\nI will post a patch for the gates to fix that.","accounts_in_message":[],"_revision_number":2},{"id":"1dfe0d933acd270174bec37de1c6a8a70577526b","author":{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},"date":"2014-09-16 19:20:34.000000000","message":"Patch Set 2:\n\nPatch for infra to fix this: https://review.openstack.org/#/c/121958/","accounts_in_message":[],"_revision_number":2},{"id":"02584a01ad83286dba31ada8827a60722e1fa26b","author":{"_account_id":9488,"name":"chenzhiwei","email":"zhiweik@gmail.com","username":"chenzhiwei"},"date":"2014-09-26 08:16:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d4f1b478f00f185a9318cbaa771a0746518297ee","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-26 09:40:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-openstack-chef-repo http://logs.openstack.org/13/114413/3/check/gate-openstack-chef-repo/21d934b : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":3},{"id":"911cfaf96d97939c61b90ad5934cc82bac19d131","author":{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},"date":"2014-09-29 16:43:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"015770fc2fc86d184ba2c3b8553f0726478173a1","author":{"_account_id":1032,"name":"Justin Shepherd","email":"jshepher@rackspace.com","username":"jshepher"},"date":"2014-10-06 14:53:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n++","accounts_in_message":[],"_revision_number":3},{"id":"a73c96e25dacbff9ec1801cf03170d29ec9824b2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-06 14:53:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"252e3eba1711f55390a8af33e90734269d598e9a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-06 14:59:12.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":3},{"id":"218aa6b7ac3c96000b436217c10944e574b1e0c8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-06 14:59:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded.\n\n- gate-openstack-chef-repo http://logs.openstack.org/13/114413/3/gate/gate-openstack-chef-repo/09c8d16 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"63e35f9d4f6b21f3a8abc8c18bdab41c4fd9cf8a","revisions":{"d21de82d642fc144da8e34b847b68cfb2b74de76":{"kind":"REWORK","_number":1,"created":"2014-08-15 01:31:39.000000000","uploader":{"_account_id":9488,"name":"chenzhiwei","email":"zhiweik@gmail.com","username":"chenzhiwei"},"ref":"refs/changes/13/114413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-chef-repo","ref":"refs/changes/13/114413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/13/114413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/13/114413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/13/114413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-chef-repo refs/changes/13/114413/1"}}},"commit":{"parents":[{"commit":"569f4af706e55802fa23a9fb93b134fe4ad1b216","subject":"Adding roles from cookbook-ceph","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/569f4af706e55802fa23a9fb93b134fe4ad1b216"}]}],"author":{"name":"Chen Zhiwei","email":"zhiwchen@cn.ibm.com","date":"2014-08-15 01:30:50.000000000","tz":0},"committer":{"name":"Chen Zhiwei","email":"zhiwchen@cn.ibm.com","date":"2014-08-15 01:30:50.000000000","tz":0},"subject":"Add Chef Zero(local mode) deploy guide.","message":"Add Chef Zero(local mode) deploy guide.\n\nChange-Id: I398c40f274cbb12caa228297b0bcc2c49ece0fcc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/d21de82d642fc144da8e34b847b68cfb2b74de76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/d21de82d642fc144da8e34b847b68cfb2b74de76"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"569f4af706e55802fa23a9fb93b134fe4ad1b216","is_merged_in_target_branch":true,"change_id":"I9f132efefdf5b78ea1a8d1ee66185c89e13713de","change_number":109387,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"5ccfd01de440cd6d1345e96d5a2953e71b37e857":{"kind":"REWORK","_number":2,"created":"2014-09-16 05:43:21.000000000","uploader":{"_account_id":9488,"name":"chenzhiwei","email":"zhiweik@gmail.com","username":"chenzhiwei"},"ref":"refs/changes/13/114413/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-chef-repo","ref":"refs/changes/13/114413/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/13/114413/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/13/114413/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/13/114413/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-chef-repo refs/changes/13/114413/2"}}},"commit":{"parents":[{"commit":"ddbbdf82c6fb54252a42c481f24c191247464771","subject":"Added explicit loops for data bag creation.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/ddbbdf82c6fb54252a42c481f24c191247464771"}]}],"author":{"name":"Chen Zhiwei","email":"zhiwchen@cn.ibm.com","date":"2014-08-15 01:30:50.000000000","tz":0},"committer":{"name":"Chen Zhiwei","email":"zhiwchen@cn.ibm.com","date":"2014-09-16 05:39:16.000000000","tz":0},"subject":"Add Chef Zero(local mode) deploy guide.","message":"Add Chef Zero(local mode) deploy guide.\n\nChange-Id: I398c40f274cbb12caa228297b0bcc2c49ece0fcc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/5ccfd01de440cd6d1345e96d5a2953e71b37e857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/5ccfd01de440cd6d1345e96d5a2953e71b37e857"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ddbbdf82c6fb54252a42c481f24c191247464771","is_merged_in_target_branch":true,"change_id":"I22c7c87cae850ff0bbd40fb51cd62fcd239aba26","change_number":114613,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"63e35f9d4f6b21f3a8abc8c18bdab41c4fd9cf8a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2014-09-26 08:16:13.000000000","uploader":{"_account_id":9488,"name":"chenzhiwei","email":"zhiweik@gmail.com","username":"chenzhiwei"},"ref":"refs/changes/13/114413/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-chef-repo","ref":"refs/changes/13/114413/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/13/114413/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/13/114413/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/13/114413/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-chef-repo refs/changes/13/114413/3"}}},"commit":{"parents":[{"commit":"ddbbdf82c6fb54252a42c481f24c191247464771","subject":"Added explicit loops for data bag creation.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/ddbbdf82c6fb54252a42c481f24c191247464771"}]}],"author":{"name":"Chen Zhiwei","email":"zhiwchen@cn.ibm.com","date":"2014-08-15 01:30:50.000000000","tz":0},"committer":{"name":"Chen Zhiwei","email":"zhiwchen@cn.ibm.com","date":"2014-09-26 08:15:46.000000000","tz":0},"subject":"Add Chef Zero(local mode) deploy guide","message":"Add Chef Zero(local mode) deploy guide\n\nChange-Id: I398c40f274cbb12caa228297b0bcc2c49ece0fcc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/63e35f9d4f6b21f3a8abc8c18bdab41c4fd9cf8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/63e35f9d4f6b21f3a8abc8c18bdab41c4fd9cf8a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ddbbdf82c6fb54252a42c481f24c191247464771","is_merged_in_target_branch":true,"change_id":"I22c7c87cae850ff0bbd40fb51cd62fcd239aba26","change_number":114613,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
