)]}'
{"id":"openstack%2Fopenstack-chef-repo~176366","triplet_id":"openstack%2Fopenstack-chef-repo~master~I6393c815339800fdf0b2ca7c99d4c87caa176398","project":"openstack/openstack-chef-repo","branch":"master","topic":"bug/1447229","hashtags":[],"change_id":"I6393c815339800fdf0b2ca7c99d4c87caa176398","subject":"Conform to update of the cookbook-openstack-client","status":"MERGED","created":"2015-04-22 15:44:05.000000000","updated":"2015-04-23 06:26:23.000000000","submitted":"2015-04-23 06:26:22.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7c0fc446a8b573224ec301aef66b65bdcb06afab","_number":176366,"virtual_id_number":176366,"owner":{"_account_id":14764,"name":"Yuki Nishiwaki","email":"uckey.1067@gmail.com","username":"ukinau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":9488,"name":"chenzhiwei","email":"zhiweik@gmail.com","username":"chenzhiwei"},{"value":0,"_account_id":8112,"name":"Ma Wen Cheng","email":"wenchma@cn.ibm.com","username":"wenchma"},{"value":2,"date":"2015-04-23 06:26:22.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},{"value":0,"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8112,"name":"Ma Wen Cheng","email":"wenchma@cn.ibm.com","username":"wenchma"},"all":[{"value":0,"_account_id":9488,"name":"chenzhiwei","email":"zhiweik@gmail.com","username":"chenzhiwei"},{"value":2,"date":"2015-04-23 01:48:06.000000000","_account_id":8112,"name":"Ma Wen Cheng","email":"wenchma@cn.ibm.com","username":"wenchma"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2015-04-22 19:18:22.000000000","_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},{"value":0,"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"all":[{"value":0,"date":"2015-04-22 21:18:03.000000000","_account_id":9488,"name":"chenzhiwei","email":"zhiweik@gmail.com","username":"chenzhiwei"},{"value":0,"_account_id":8112,"name":"Ma Wen Cheng","email":"wenchma@cn.ibm.com","username":"wenchma"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},{"value":1,"date":"2015-04-23 06:24:21.000000000","_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},{"_account_id":8112,"name":"Ma Wen Cheng","email":"wenchma@cn.ibm.com","username":"wenchma"},{"_account_id":9488,"name":"chenzhiwei","email":"zhiweik@gmail.com","username":"chenzhiwei"},{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-04-22 19:18:22.000000000","updated_by":{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},"reviewer":{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},"state":"REVIEWER"},{"updated":"2015-04-22 21:18:03.000000000","updated_by":{"_account_id":9488,"name":"chenzhiwei","email":"zhiweik@gmail.com","username":"chenzhiwei"},"reviewer":{"_account_id":9488,"name":"chenzhiwei","email":"zhiweik@gmail.com","username":"chenzhiwei"},"state":"REVIEWER"},{"updated":"2015-04-23 01:48:06.000000000","updated_by":{"_account_id":8112,"name":"Ma Wen Cheng","email":"wenchma@cn.ibm.com","username":"wenchma"},"reviewer":{"_account_id":8112,"name":"Ma Wen Cheng","email":"wenchma@cn.ibm.com","username":"wenchma"},"state":"REVIEWER"},{"updated":"2015-04-23 06:24:21.000000000","updated_by":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"reviewer":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"state":"REVIEWER"},{"updated":"2015-04-23 06:26:22.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"9c9cf22a5a67867fa0d8de150a817bd60cc834e5","author":{"_account_id":14764,"name":"Yuki Nishiwaki","email":"uckey.1067@gmail.com","username":"ukinau"},"date":"2015-04-22 15:44:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36dae1cf9b0eade769b44f28deda16d377966e88","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-22 15:57:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-chef-repo http://logs.openstack.org/66/176366/1/check/gate-openstack-chef-repo/52018f5/ : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":1},{"id":"54b0db077f79f458d2102e776c8ec46578a42dc5","author":{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},"date":"2015-04-22 18:19:26.000000000","message":"Patch Set 1: Code-Review+2\n\nfor sake of getting repo whole again, lets merge this.  But I think there\u0027s agreement to try something else with the berks file to allow easier master vs development use cases.","accounts_in_message":[],"_revision_number":1},{"id":"b1751ab91ec6076664b5ac1906d876ceac808c50","author":{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},"date":"2015-04-22 18:28:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nquestion about yum change","accounts_in_message":[],"_revision_number":1},{"id":"3ccadef1afe3791cdad3ecf8a38a59d131097983","author":{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},"date":"2015-04-22 19:18:12.000000000","message":"Patch Set 2: Commit message was updated","accounts_in_message":[],"_revision_number":2},{"id":"897957717a7e99aa59d3e521c2f90e5119285359","author":{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},"date":"2015-04-22 19:18:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ea29d643e319f1635384506dcb177847c8469b73","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-22 19:24:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-chef-repo http://logs.openstack.org/66/176366/2/check/gate-openstack-chef-repo/5b744be/ : SUCCESS in 2m 08s","accounts_in_message":[],"_revision_number":2},{"id":"637e61a209fb152719a639ca369ecebe6f82ffa3","author":{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},"date":"2015-04-22 21:19:03.000000000","message":"Patch Set 2:\n\nrelated common patch for yum: https://review.openstack.org/#/c/176477/","accounts_in_message":[],"_revision_number":2},{"id":"246510579bbc146dfab2f3c5d6eae87c574da475","author":{"_account_id":8112,"name":"Ma Wen Cheng","email":"wenchma@cn.ibm.com","username":"wenchma"},"date":"2015-04-23 01:48:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3cdf3206bb9534746a82434cbe22361c4b8d5794","author":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"date":"2015-04-23 06:24:21.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7eac82ea25c99f90a918e8542e07e73e5b9d1d63","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-23 06:24:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"7be593cce8cc8946b26ffb0225c70386e085d896","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-23 06:26:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-openstack-chef-repo http://logs.openstack.org/66/176366/2/gate/gate-openstack-chef-repo/ed03e71/ : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":2},{"id":"bdef293044fcb87c0370deed640bb4e5dac37271","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-23 06:26:23.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"63c48f49cbdd9ab540d6322a0e9ca6e29c125d1c","revisions":{"209899335215a23e1a248b37babb8028a32f467f":{"kind":"REWORK","_number":1,"created":"2015-04-22 15:44:05.000000000","uploader":{"_account_id":14764,"name":"Yuki Nishiwaki","email":"uckey.1067@gmail.com","username":"ukinau"},"ref":"refs/changes/66/176366/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-chef-repo","ref":"refs/changes/66/176366/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/66/176366/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/66/176366/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/66/176366/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-chef-repo refs/changes/66/176366/1"}}},"commit":{"parents":[{"commit":"2047bb6dbb73897a68867fc8e1fa8395e2cce42d","subject":"Merge \"Remove unused neutron roles from aio nova recipe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/2047bb6dbb73897a68867fc8e1fa8395e2cce42d"}]}],"author":{"name":"Yuki Nishiwaki","email":"uckey.1067@gmail.com","date":"2015-04-22 15:35:47.000000000","tz":540},"committer":{"name":"Yuki Nishiwaki","email":"uckey.1067@gmail.com","date":"2015-04-22 15:44:02.000000000","tz":540},"subject":"Conform to update of the cookbook-openstack-client","message":"Conform to update of the cookbook-openstack-client\n\nBerksfile is updated, because the cookbook-openstack-client is updated.\nThe update to be need to change Berksfile are following.\n1. The name of cookbook was changed from \"openstack-client\" to\n\"openstack_client\"\n2. The dependency version of yum increased than 3.5.2\n\nChange-Id: I6393c815339800fdf0b2ca7c99d4c87caa176398\nCloses-bug: #1447229\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/209899335215a23e1a248b37babb8028a32f467f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/209899335215a23e1a248b37babb8028a32f467f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2047bb6dbb73897a68867fc8e1fa8395e2cce42d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"63c48f49cbdd9ab540d6322a0e9ca6e29c125d1c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2015-04-22 19:18:12.000000000","uploader":{"_account_id":7128,"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","username":"kramvan"},"ref":"refs/changes/66/176366/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-chef-repo","ref":"refs/changes/66/176366/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/66/176366/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/66/176366/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/66/176366/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-chef-repo refs/changes/66/176366/2"}}},"commit":{"parents":[{"commit":"2047bb6dbb73897a68867fc8e1fa8395e2cce42d","subject":"Merge \"Remove unused neutron roles from aio nova recipe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/2047bb6dbb73897a68867fc8e1fa8395e2cce42d"}]}],"author":{"name":"Yuki Nishiwaki","email":"uckey.1067@gmail.com","date":"2015-04-22 15:35:47.000000000","tz":540},"committer":{"name":"Mark Vanderwiel","email":"vanderwl@us.ibm.com","date":"2015-04-22 19:18:12.000000000","tz":0},"subject":"Conform to update of the cookbook-openstack-client","message":"Conform to update of the cookbook-openstack-client\n\nBerksfile is updated, because the cookbook-openstack-client is updated.\nThe update to be need to change Berksfile are following.\n1. The name of cookbook was changed from \"openstack-client\" to\n\"openstack_client\"\n2. The dependency version of yum increased than 3.5.2, client cookbook requires 3.5.4.\n\nChange-Id: I6393c815339800fdf0b2ca7c99d4c87caa176398\nCloses-bug: #1447229","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/63c48f49cbdd9ab540d6322a0e9ca6e29c125d1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/63c48f49cbdd9ab540d6322a0e9ca6e29c125d1c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2047bb6dbb73897a68867fc8e1fa8395e2cce42d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
