)]}'
{"id":"openstack%2Fopenstack-chef-repo~542613","triplet_id":"openstack%2Fopenstack-chef-repo~master~Ief99c2de7bbef8a61d57e7fde59507f791a4519f","project":"openstack/openstack-chef-repo","branch":"master","topic":"bp/modern-chef","hashtags":[],"change_id":"Ief99c2de7bbef8a61d57e7fde59507f791a4519f","subject":"Revert inspec test to use Tempest from git","status":"MERGED","created":"2018-02-09 06:19:55.000000000","updated":"2018-02-14 17:12:27.000000000","submitted":"2018-02-14 17:12:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"542613-1518628347684-b26dfab8","meta_rev_id":"8ffe424c3fbcf7fc7d140d53e2bfa03f15d521ea","_number":542613,"virtual_id_number":542613,"owner":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},{"value":0,"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},{"value":2,"date":"2018-02-14 17:12:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2018-02-09 06:59:08.000000000","_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"all":[{"value":0,"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},{"value":2,"date":"2018-02-14 16:17:50.000000000","_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"all":[{"value":1,"date":"2018-02-14 16:43:39.000000000","_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},{"value":0,"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-02-09 06:59:08.000000000","updated_by":{"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"},"reviewer":{"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"},"state":"REVIEWER"},{"updated":"2018-02-14 16:17:50.000000000","updated_by":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"reviewer":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"state":"REVIEWER"},{"updated":"2018-02-14 16:43:39.000000000","updated_by":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"reviewer":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"state":"REVIEWER"},{"updated":"2018-02-14 17:12:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"695ec1ec4484f03f9e30b17d264e358e2f11c226","author":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"date":"2018-02-09 06:19:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d31fcca6e16fb545913572230d0d1ee66c3c7d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-09 06:38:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-chef-repo-rake http://logs.openstack.org/13/542613/1/check/openstack-chef-repo-rake/ae2d012/ : SUCCESS in 2m 11s\n- openstack-chef-repo-integration http://logs.openstack.org/13/542613/1/check/openstack-chef-repo-integration/af5802e/ : SUCCESS in 18m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9923b6f59f3803a89541392b0cd5b5a07d4a84c4","author":{"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"},"date":"2018-02-09 06:59:08.000000000","message":"Patch Set 1:\n\n* cloudbau-Chef-OpenStack-CI http://ci.cloudbau.de/ci-logs/refs-changes-13-542613-1 : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"1f21ae905ed875e5bc473699d035b83c0a35bcda","author":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"date":"2018-02-14 16:17:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"940a3b6de20f7de6c021977ef5031293899314b3","author":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"date":"2018-02-14 16:43:39.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a8366bd0da65c4ff7e60afcd3f6487d91ff8cbe2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-14 16:49:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2edd5c2c0ed27c3589c7f5537c41d5b7ec6e559f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-14 17:12:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-chef-repo-rake http://logs.openstack.org/13/542613/1/gate/openstack-chef-repo-rake/1e2a291/ : SUCCESS in 2m 11s\n- openstack-chef-repo-integration http://logs.openstack.org/13/542613/1/gate/openstack-chef-repo-integration/e6d4315/ : SUCCESS in 18m 13s","accounts_in_message":[],"_revision_number":1},{"id":"214999989804393bd3cc3fecaaa311040f893f13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-02-14 17:12:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"642b6e7175029fe003003e0679285b176291aa8e","revisions":{"642b6e7175029fe003003e0679285b176291aa8e":{"kind":"REWORK","_number":1,"created":"2018-02-09 06:19:55.000000000","uploader":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"ref":"refs/changes/13/542613/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-chef-repo","ref":"refs/changes/13/542613/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/13/542613/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/13/542613/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-chef-repo refs/changes/13/542613/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-chef-repo refs/changes/13/542613/1"}}},"commit":{"parents":[{"commit":"5c59b15c002503e4489d955eade2f9955be761bf","subject":"Merge \"Zuul: Remove project name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/5c59b15c002503e4489d955eade2f9955be761bf"}]}],"author":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2018-02-09 06:19:06.000000000","tz":-480},"committer":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2018-02-09 06:19:54.000000000","tz":-480},"subject":"Revert inspec test to use Tempest from git","message":"Revert inspec test to use Tempest from git\n\nChange-Id: Ief99c2de7bbef8a61d57e7fde59507f791a4519f\nImplements: blueprint modern-chef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/642b6e7175029fe003003e0679285b176291aa8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-chef-repo/commit/642b6e7175029fe003003e0679285b176291aa8e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5c59b15c002503e4489d955eade2f9955be761bf","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
