)]}'
{"id":"openstack%2Fopenstack-health~232796","triplet_id":"openstack%2Fopenstack-health~master~Ifea383ce23781fbf72ac0d7cc2663f81801800e0","project":"openstack/openstack-health","branch":"master","hashtags":[],"change_id":"Ifea383ce23781fbf72ac0d7cc2663f81801800e0","subject":"Improve gauge chart exhibition","status":"MERGED","created":"2015-10-08 21:49:44.000000000","updated":"2015-10-09 20:02:47.000000000","submitted":"2015-10-09 20:02:45.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7945dcd2e07baaffaa510932678e48e330f00228","_number":232796,"virtual_id_number":232796,"owner":{"_account_id":18001,"name":"Caio Carrara","email":"ccarrara@thoughtworks.com","username":"caiocarrara"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":11682,"name":"Ryan Moats","email":"rmoats@us.ibm.com","username":"regXboi"},{"value":2,"date":"2015-10-09 20:02:45.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5623,"name":"David Lyle","email":"dklyle0@gmail.com","username":"david-lyle"},{"value":0,"_account_id":17001,"name":"Tim Buckley","email":"timothy.jas.buckley@hpe.com","username":"timothy.jas.buckley"},{"value":0,"_account_id":17995,"name":"Glauco Oliveira","email":"gvinici@thoughtworks.com","username":"glauco"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17001,"name":"Tim Buckley","email":"timothy.jas.buckley@hpe.com","username":"timothy.jas.buckley"},"all":[{"value":0,"_account_id":11682,"name":"Ryan Moats","email":"rmoats@us.ibm.com","username":"regXboi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5623,"name":"David Lyle","email":"dklyle0@gmail.com","username":"david-lyle"},{"value":2,"date":"2015-10-09 19:59:52.000000000","_account_id":17001,"name":"Tim Buckley","email":"timothy.jas.buckley@hpe.com","username":"timothy.jas.buckley"},{"value":1,"date":"2015-10-09 18:51:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17995,"name":"Glauco Oliveira","email":"gvinici@thoughtworks.com","username":"glauco"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17001,"name":"Tim Buckley","email":"timothy.jas.buckley@hpe.com","username":"timothy.jas.buckley"},"all":[{"value":0,"date":"2015-10-08 21:50:53.000000000","_account_id":11682,"name":"Ryan Moats","email":"rmoats@us.ibm.com","username":"regXboi"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2015-10-08 21:50:53.000000000","_account_id":5623,"name":"David Lyle","email":"dklyle0@gmail.com","username":"david-lyle"},{"value":1,"date":"2015-10-09 19:59:52.000000000","_account_id":17001,"name":"Tim Buckley","email":"timothy.jas.buckley@hpe.com","username":"timothy.jas.buckley"},{"value":0,"date":"2015-10-08 21:50:44.000000000","_account_id":17995,"name":"Glauco Oliveira","email":"gvinici@thoughtworks.com","username":"glauco"},{"value":0,"date":"2015-10-08 21:50:36.000000000","_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"date":"2015-10-08 21:50:53.000000000","_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5623,"name":"David Lyle","email":"dklyle0@gmail.com","username":"david-lyle"},{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},{"_account_id":11682,"name":"Ryan Moats","email":"rmoats@us.ibm.com","username":"regXboi"},{"_account_id":17001,"name":"Tim Buckley","email":"timothy.jas.buckley@hpe.com","username":"timothy.jas.buckley"},{"_account_id":17995,"name":"Glauco Oliveira","email":"gvinici@thoughtworks.com","username":"glauco"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-10-08 21:50:36.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2015-10-08 21:50:44.000000000","updated_by":{"_account_id":17995,"name":"Glauco Oliveira","email":"gvinici@thoughtworks.com","username":"glauco"},"reviewer":{"_account_id":17995,"name":"Glauco Oliveira","email":"gvinici@thoughtworks.com","username":"glauco"},"state":"REVIEWER"},{"updated":"2015-10-08 21:50:53.000000000","updated_by":{"_account_id":5623,"name":"David Lyle","email":"dklyle0@gmail.com","username":"david-lyle"},"reviewer":{"_account_id":5623,"name":"David Lyle","email":"dklyle0@gmail.com","username":"david-lyle"},"state":"REVIEWER"},{"updated":"2015-10-08 21:50:53.000000000","updated_by":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"reviewer":{"_account_id":7069,"name":"Joshua Hesketh","email":"josh@nitrotech.org","username":"jhesketh"},"state":"REVIEWER"},{"updated":"2015-10-08 21:50:53.000000000","updated_by":{"_account_id":11682,"name":"Ryan Moats","email":"rmoats@us.ibm.com","username":"regXboi"},"reviewer":{"_account_id":11682,"name":"Ryan Moats","email":"rmoats@us.ibm.com","username":"regXboi"},"state":"REVIEWER"},{"updated":"2015-10-09 19:59:52.000000000","updated_by":{"_account_id":17001,"name":"Tim Buckley","email":"timothy.jas.buckley@hpe.com","username":"timothy.jas.buckley"},"reviewer":{"_account_id":17001,"name":"Tim Buckley","email":"timothy.jas.buckley@hpe.com","username":"timothy.jas.buckley"},"state":"REVIEWER"},{"updated":"2015-10-09 20:02:45.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"ead1c850b012353e51f8634ae5d471383dff4be5","author":{"_account_id":18001,"name":"Caio Carrara","email":"ccarrara@thoughtworks.com","username":"caiocarrara"},"date":"2015-10-08 21:49:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36bbea65c37fecd5b36ab89f50aee5072374aba4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-08 22:01:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-openstack-health-pep8 http://logs.openstack.org/96/232796/1/check/gate-openstack-health-pep8/eb73747/ : SUCCESS in 52s\n- gate-openstack-health-docs http://docs-draft.openstack.org/96/232796/1/check/gate-openstack-health-docs/694de56//doc/build/html/ : SUCCESS in 56s\n- gate-openstack-health-python27 http://logs.openstack.org/96/232796/1/check/gate-openstack-health-python27/22623c7/ : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1},{"id":"2616c7d3a4418d3680497937988c12ee36b17296","author":{"_account_id":17995,"name":"Glauco Oliveira","email":"gvinici@thoughtworks.com","username":"glauco"},"date":"2015-10-09 18:51:57.000000000","message":"Patch Set 1: Code-Review+1\n\nNoticed that there is plenty of empty space under each gauge chart. Would it be possible to remove this empty space?\n\nI am fine with the idea of tackling this as part of another patchset.","accounts_in_message":[],"_revision_number":1},{"id":"023f34ad15f5fa45616bf5af58e9e36035b9930f","author":{"_account_id":17001,"name":"Tim Buckley","email":"timothy.jas.buckley@hpe.com","username":"timothy.jas.buckley"},"date":"2015-10-09 19:59:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM.\n\nThe empty space is there because the \"donut\" chart is really just half a full-circle pie chart, and we slice off the bottom. Nvd3 still tries to fit the chart in the svg as if it were a full circle, and as far as I know this behavior can\u0027t be modified.\n\nWe\u0027ll probably want to move to a custom d3 chart for the gauge eventually, but until then this should be fine.","accounts_in_message":[],"_revision_number":1},{"id":"f6f9e0bf88f2e45cf89536c8bde25e47a2f130b5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-09 19:59:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"52198baec427c72d490e1e2fd2a59fa9d7b39335","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-09 20:02:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-openstack-health-docs http://docs-draft.openstack.org/96/232796/1/gate/gate-openstack-health-docs/5a11b67//doc/build/html/ : SUCCESS in 54s\n- gate-openstack-health-pep8 http://logs.openstack.org/96/232796/1/gate/gate-openstack-health-pep8/49e94ec/ : SUCCESS in 47s\n- gate-openstack-health-python27 http://logs.openstack.org/96/232796/1/gate/gate-openstack-health-python27/bf1f354/ : SUCCESS in 1m 35s\n- gate-openstack-health-npm-run-lint http://logs.openstack.org/96/232796/1/gate/gate-openstack-health-npm-run-lint/cc364d3/ : FAILURE in 1m 20s (non-voting)\n- gate-openstack-health-npm-run-test http://logs.openstack.org/96/232796/1/gate/gate-openstack-health-npm-run-test/a81ed8e/ : FAILURE in 1m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7945dcd2e07baaffaa510932678e48e330f00228","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-09 20:02:47.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c84e4e9a3e98e7da50a5872e54d8f6bbeeeb270a","revisions":{"c84e4e9a3e98e7da50a5872e54d8f6bbeeeb270a":{"kind":"REWORK","_number":1,"created":"2015-10-08 21:49:44.000000000","uploader":{"_account_id":18001,"name":"Caio Carrara","email":"ccarrara@thoughtworks.com","username":"caiocarrara"},"ref":"refs/changes/96/232796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-health","ref":"refs/changes/96/232796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/96/232796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/96/232796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-health refs/changes/96/232796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-health refs/changes/96/232796/1"}}},"commit":{"parents":[{"commit":"05d2e630fbc8a76e879bd38bc20124a5b2e23834","subject":"Fix project name encoded in project controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/05d2e630fbc8a76e879bd38bc20124a5b2e23834"}]}],"author":{"name":"Caio Carrara","email":"ccarrara@thoughtworks.com","date":"2015-10-08 21:48:32.000000000","tz":-180},"committer":{"name":"Caio Carrara","email":"ccarrara@thoughtworks.com","date":"2015-10-08 21:48:32.000000000","tz":-180},"subject":"Improve gauge chart exhibition","message":"Improve gauge chart exhibition\n\nThis changes the gauge chart increasing the donut hole at center and\nalso increasing the chart height.\n\nChange-Id: Ifea383ce23781fbf72ac0d7cc2663f81801800e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/c84e4e9a3e98e7da50a5872e54d8f6bbeeeb270a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-health/commit/c84e4e9a3e98e7da50a5872e54d8f6bbeeeb270a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
